)]}'
{"id":"openstack%2Floci~936627","triplet_id":"openstack%2Floci~master~I39b46d3af33edabf2d0da9cc6ac1c7cf5373ad34","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2024-12-05 01:42:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I39b46d3af33edabf2d0da9cc6ac1c7cf5373ad34","subject":"Ignore already installed packages for pip","status":"MERGED","created":"2024-11-28 14:58:04.000000000","updated":"2024-12-05 01:46:00.000000000","submitted":"2024-12-05 01:42:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936627","meta_rev_id":"766b42d5bf10cd72471b4ec33461e6c516758e9c","_number":936627,"virtual_id_number":936627,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-05 01:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2024-12-04 21:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2024-12-04 22:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":1,"date":"2024-12-04 22:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-28 17:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 23:50:50.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2024-12-04 22:07:12.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"f8eac0ad5b74caa03c70852c2d75aa0e49627f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-28 14:58:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"641ac2ee0bf577f228f74232c620924cf3a7ef53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 17:23:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00efa03318334e5abf73904137eb153f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c89b6672cef74864b8ee53ea2b896ff6 : SUCCESS in 2h 19m 48s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4815d4fddd3e476686d36d2412363474 : SUCCESS in 58m 19s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/707acc627a5947f19a0587c92e5de0e3 : SUCCESS in 1h 08m 20s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7090754c948c4b9d9b92df79fa6e36e1 : SUCCESS in 57m 54s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7610d923e60e42228e5608a4a13a1e67 : SUCCESS in 1h 08m 22s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ac6b2b2cdee549808b2285f127b31cc0 : SUCCESS in 58m 52s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/83e3f6dcb7144a0b8eaed7f36ddd4d0d : SUCCESS in 1h 05m 25s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1fecbe0d74664f2ca4ff9e536d72fdbc : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":1},{"id":"05ea507a9e045b13842d4b06b011cfb203c74232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2024-11-28 19:50:18.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"63faa7cf5c4cb1c89c92fbd5c37f3b94ab3bc979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 22:27:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eebb37998af949cbae769cbd002983f5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/e132a14cab2b4090bf52803ff19b9faf : SUCCESS in 2h 32m 28s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9998cffccce74faab725254258790e37 : SUCCESS in 1h 06m 41s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/81df6a9a1b8045fab9ea8082f1f11452 : SUCCESS in 1h 10m 13s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/61e98dc438e44dcabac973ea3374a9fd : SUCCESS in 1h 09m 16s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4647f99824de4a30ac56fe29ca31337a : SUCCESS in 1h 09m 48s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d7a07b1686f4bbf9d32e37e1757ddb1 : SUCCESS in 1h 01m 56s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2acd7218c75944bb8d861d942d5a1698 : SUCCESS in 1h 06m 27s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1e167b408c4c4d228deb2773f357581f : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":2},{"id":"70a73986074f2b2713df33f954df65e194019798","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-12-02 23:50:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6da426bd69c7ac7e1e14349b95d47fd6a6d5a8f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"real_author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-12-04 21:55:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_14525\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"_revision_number":3},{"id":"657c325c1e2428fe530247056fcb7d44c951dba2","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2024-12-04 22:07:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"36e7dbde3e5635883ea576c05e103cad897fc987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 00:44:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e09534b5b3094ab0951e934dc8e4d0e3\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9e43fadde8c6419aa71aa9ab5bf33183 : SUCCESS in 2h 43m 01s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/dd144821fb5448ed92922e2c0fa3c328 : SUCCESS in 1h 07m 24s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a6b1bb10c7b142d3953a73b713719c58 : SUCCESS in 1h 19m 53s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/851ca63b42e74aef8c6422758a164417 : SUCCESS in 1h 10m 27s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a9b7d35017af48359fcc838a2c0e431a : SUCCESS in 1h 12m 05s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3dce314e98c1457a92a377bb46afbab2 : SUCCESS in 1h 08m 07s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0e08c95e93594d088f60cdae24f98e3c : SUCCESS in 1h 18m 37s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2db4dfa7e5e74193bc3673a51439d59f : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8e9cfe37c4503084f95e355741635acf4c535e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 00:44:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"219955bb971fc72bf93bce3e566df9898375366c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 01:42:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a8b8d516824ac88daafc4e71d9c9ce\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0ee14d1baf864c7386588433ada361eb : SUCCESS in 57m 46s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d535314312274ab7ae1008db0bfceb2d : SUCCESS in 53m 48s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9ad4a4016a3f4fb5a246c9ac2c067fe7 : SUCCESS in 52m 09s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/88b744848abb489a861d91461cf3e056 : SUCCESS in 53m 38s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/22d5f98ec7594e1aa4f1c809be3a8647 : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":3},{"id":"fcbc67ee979b9b292c47909928818abad92e0a92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 01:42:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"766b42d5bf10cd72471b4ec33461e6c516758e9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 01:46:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87b596459d734054bddb64ae3c31959e\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d523ad50c414c6aadbb3399ead0dcba : SUCCESS in 2m 48s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c049176b03c341dbab481890ac3be472 : SUCCESS in 2m 28s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/442b9396f3be44188bf85b17ba2623f0 : SUCCESS in 3m 17s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a29a354eb3b846a5a91f234af14e581d : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6224a5a40af43e114e82935caef87d57b95ff87a","revisions":{"162acc4fb191e7d32e808fcb2b2b1099be5a19c8":{"kind":"REWORK","_number":1,"created":"2024-11-28 14:58:04.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/27/936627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/27/936627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/27/936627/1"}}},"commit":{"parents":[{"commit":"dbc29edde0e01391849baa940103237f9868265c","subject":"Allow to build libraries from sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dbc29edde0e01391849baa940103237f9868265c"}]}],"author":{"name":"dbiletskiy","email":"dbiletskiy@mirantis.com","date":"2023-08-08 07:47:25.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-28 14:57:53.000000000","tz":0},"subject":"Ignore already installed packages for pip","message":"Ignore already installed packages for pip\n\nRelated-Prod: PRODX-34482\n\nCo-Authored-By: vsaienko \u003cvsaienko@mirantis.com\u003e\n\nChange-Id: I39b46d3af33edabf2d0da9cc6ac1c7cf5373ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/162acc4fb191e7d32e808fcb2b2b1099be5a19c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/162acc4fb191e7d32e808fcb2b2b1099be5a19c8"}]},"branch":"refs/heads/master"},"c21b90d32d4b756e52e3028e649d343abdaa713e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-28 19:50:18.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/27/936627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/27/936627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/27/936627/2"}}},"commit":{"parents":[{"commit":"dbc29edde0e01391849baa940103237f9868265c","subject":"Allow to build libraries from sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dbc29edde0e01391849baa940103237f9868265c"}]}],"author":{"name":"dbiletskiy","email":"dbiletskiy@mirantis.com","date":"2023-08-08 07:47:25.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-11-28 19:50:18.000000000","tz":0},"subject":"Ignore already installed packages for pip","message":"Ignore already installed packages for pip\n\nCo-Authored-By: vsaienko \u003cvsaienko@mirantis.com\u003e\n\nChange-Id: I39b46d3af33edabf2d0da9cc6ac1c7cf5373ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c21b90d32d4b756e52e3028e649d343abdaa713e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c21b90d32d4b756e52e3028e649d343abdaa713e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6224a5a40af43e114e82935caef87d57b95ff87a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-04 21:55:41.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"real_uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/27/936627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/27/936627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/27/936627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/27/936627/3"}}},"commit":{"parents":[{"commit":"614016bdb7e774556aa56695cfd332b176e917ca","subject":"Allow to build libraries from sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/614016bdb7e774556aa56695cfd332b176e917ca"}]}],"author":{"name":"dbiletskiy","email":"dbiletskiy@mirantis.com","date":"2023-08-08 07:47:25.000000000","tz":120},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-12-04 21:55:41.000000000","tz":0},"subject":"Ignore already installed packages for pip","message":"Ignore already installed packages for pip\n\nCo-Authored-By: vsaienko \u003cvsaienko@mirantis.com\u003e\n\nChange-Id: I39b46d3af33edabf2d0da9cc6ac1c7cf5373ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6224a5a40af43e114e82935caef87d57b95ff87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6224a5a40af43e114e82935caef87d57b95ff87a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
