)]}'
{"id":"openstack%2Floci~938920","triplet_id":"openstack%2Floci~master~Icc4c84a8ec6d13e67a573c5837e595f787b4cf81","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-01-13 19:06:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc4c84a8ec6d13e67a573c5837e595f787b4cf81","subject":"Rename .zuul.d -\u003e zuul.d","status":"MERGED","created":"2025-01-10 15:57:54.000000000","updated":"2025-01-13 19:14:54.000000000","submitted":"2025-01-13 19:06:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938920","meta_rev_id":"cc0092aba250479a00186e0e9009c5b2a0431340","_number":938920,"virtual_id_number":938920,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 19:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-01-13 17:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-01-13 14:41:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-01-13 17:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-10 16:25:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 14:41:04.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-01-13 17:52:51.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"4ba1b08de2739d4b8544da818b55de35e18fefa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-10 15:57:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feab9b8b0b2e597959694b6472d32c80c7d296a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:25:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fc1c30026e84e40b7ba7d04c2fa73b6\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/02a4276be35441168a97e8c55ba6993b : SUCCESS in 23m 31s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c526f20624804feb8d9c72fd8ca464f2 : FAILURE in 4m 25s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/61de82d4a32c431ea1125e6aa9fc321e : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c7f9c84864ff498cac45fee989fdbfdd : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0952cf7ae23b480dad69f69eeea29d20 : FAILURE in 4m 22s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/95e034ec51ff432f92624ecaa4532ad3 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/dba16757712d4f99bd96f08b5fc2c40f : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eeb9f319e64346bba85c1a7ddad9adf3 : FAILURE in 4m 52s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0d9150cdea1d4233b0db6f1b849184bc : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0ccdf0aaa51348ec887ce9c1cd20a2ab : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3bddbb0615474835a2261c7dae18a50c : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e7f6181ce287502e04856eb017fbb72f649ba535","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-10 19:31:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd2dfef8c714522909891ac0338a599852538809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 21:53:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c6241f39b7e40779a864f427a39d5f4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/61d1169eb90c47dca8098b5b28b8c979 : SUCCESS in 2h 21m 06s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cbe54e9ade7947d989c48e29cb913737 : SUCCESS in 53m 38s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9a105d7fe1684f069c0a4443a325b0de : SUCCESS in 1h 13m 50s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ca2b4bf6fbed4dd3b169f40c8f7eb15a : SUCCESS in 54m 17s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2a81fdc7516948aeb2d59375c19514e2 : SUCCESS in 53m 27s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e88d304838cf4e2c9488b16c3dbad009 : SUCCESS in 1h 14m 12s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d70a30d52b694c528376bd15d1fcaaac : SUCCESS in 52m 35s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5f517cbb3e4e4266aee6333b2e81cac1 : SUCCESS in 52m 44s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9adcce817471426e9688d170cc479386 : SUCCESS in 1h 17m 31s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/759eaa8e37494f4384b1d85ff24bf19e : SUCCESS in 52m 48s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fdf2d0509d82465da5205e9bf90182c8 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6a260a8f5719e9760345de67cee3af400ace405b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-10 21:54:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae50d817ec2228d5ddfb71036c050ccd2394471a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-11 00:27:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/424fe113f2c44880a7a8cc84d80a88b0\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/44014a7efba94b90885ef2445630ca67 : SUCCESS in 2h 28m 05s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e524b09a9d7f49d4a863dacbb6350c3e : SUCCESS in 55m 34s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b256fee309aa4fd78c88815b9e310a99 : FAILURE in 1h 05m 54s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/dadd4046625341aba0ce2e1877bbfc1c : SUCCESS in 1h 01m 47s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/12826949a4374ad6b68741a0bb4af904 : SUCCESS in 50m 10s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/70f720974d2f40afaffe2d5b19e37286 : SUCCESS in 1h 11m 33s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90e3302b18e24138bc4c98d7d223ef37 : SUCCESS in 46m 28s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f05b3abe86e442c185ba6e5bf8e89a10 : SUCCESS in 51m 05s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2ee2d1bcee434f899af6227823e20f0d : SUCCESS in 1h 07m 36s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fa3c7e27f8064ff7804e4235a3464c99 : SUCCESS in 47m 00s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89ce730bc4c240a9a8a834a496cec245 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":2},{"id":"1531a9c4396f17a839884ea6ab77f827a41e54d8","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-11 08:39:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c60d76e74ca04820932580f5c759c750c2db1804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-11 11:10:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b953cb6d0c974f9a9c0e035ba99597a2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f9092250faad44648dd8b5a57bb6d289 : SUCCESS in 2h 28m 50s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6ef2e61af3f24a3eabfbd2875ffced46 : SUCCESS in 58m 33s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b3c76749c0454e939d56b83caa7aae86 : SUCCESS in 1h 15m 59s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/47163613cfee40afb73cec8a3c427b13 : SUCCESS in 1h 03m 11s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f1f1411879e243058c6209d4edbabe47 : SUCCESS in 55m 59s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f79a805be4b943da9a140ad9ba9edd95 : SUCCESS in 1h 12m 20s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/15b414aed0b1408088c97c82c0a61749 : SUCCESS in 58m 18s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c2c206410a154d8ab81c1f237d239070 : SUCCESS in 56m 27s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/31524d0917934c5393d308724c7473d7 : SUCCESS in 1h 13m 50s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9877292bc630478ca70fe701b6a9a04e : SUCCESS in 1h 05m 12s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7f63d985c4224044a7448228c8a38921 : FAILURE in 13m 21s","accounts_in_message":[],"_revision_number":2},{"id":"bfa20437423ec11dc185058d4e8c0b6158a956fe","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-12 15:17:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"104f80a1cc2c28133282673f23018a5d318709be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 16:59:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98292c1b82db470eaf4dcc6076348bf5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d2efcfbe19cc4c869a63135f788ea27e : SUCCESS in 1h 41m 16s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0b6eba4e818343c8803bf9aa3209731b : SUCCESS in 40m 21s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d0d2d186b4c240d3bdbfffa29deb0e60 : SUCCESS in 57m 13s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/74ebfb4fcf86424aa73947872e7a5d39 : SUCCESS in 38m 43s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7e48c0469bce4da6afc1b4c8c97d3c35 : SUCCESS in 39m 17s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/98d4a9ddd6dc4635ad61565b69ee5e68 : SUCCESS in 57m 04s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6241b4b91fdb4e73873f850febff297e : SUCCESS in 37m 05s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ced5c8db082b488c81751478e5280c07 : SUCCESS in 39m 32s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f1e3cc81de9145f1b20b6b32b4e4a542 : SUCCESS in 56m 40s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bd8c938334c94865869886fb7cb7d2b2 : SUCCESS in 37m 27s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0611b6f9130743f48205a492b5c00980 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":2},{"id":"01e208575f2e72c6898ca56f3e39ad147d9d8696","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-01-13 14:41:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4e7bb952e2e174c5f67c85bdfd0438cf3e4a0dca","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-01-13 17:52:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d69629c44c6ec5ec51ee1c43ab9c8e5d88d7771","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 17:53:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c7a5a34167eb19216e3fcad4efc2a6bee46c34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 19:06:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e75c06e579204d4d973b8b6068967ee0\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b22927d6765f4555856ddb31d86e5aaf : SUCCESS in 1h 13m 22s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f051f6c5568d4b678575cc8523c4f070 : SUCCESS in 58m 36s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/13853c43cd9e414b901e9e5bd5e1b523 : SUCCESS in 1h 07m 40s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/75f87ba543b7485cbef3e3b43d83d016 : SUCCESS in 56m 06s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/945500fe8a7944f2bd1d4576b8a977a0 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c133d841262bbd486f6a70f29d3e8dad166db586","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 19:06:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc0092aba250479a00186e0e9009c5b2a0431340","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 19:14:54.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5cddf20d0584a05a7b087e85b4b69a9\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/382e9033a20746a99a6f2b8d10e4a726 : FAILURE in 5m 53s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2b29b28975e3491080b1d20fd40de506 : FAILURE in 5m 19s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3b15c081c93f407681516981b84e4fa1 : FAILURE in 5m 19s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0ba67a46f2ce477cae8a0dde4fc72ccf : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ec7a09e9e15cf0c43e4199d66bbc05a96f7a794","revisions":{"a1deec52361339a013ccc843aff21ed0c2e2df0b":{"kind":"REWORK","_number":1,"created":"2025-01-10 15:57:54.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/20/938920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/938920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/938920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/938920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/938920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/938920/1"}}},"commit":{"parents":[{"commit":"915fc739a677b9c725274cf627e283450f56279a","subject":"Use periodic pipeline for nightly builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/915fc739a677b9c725274cf627e283450f56279a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-10 15:54:19.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-10 15:54:19.000000000","tz":-360},"subject":"Rename .zuul.d -\u003e zuul.d","message":"Rename .zuul.d -\u003e zuul.d\n\nThis is to align it with other Openstack-Helm projects.\nWe use same test jobs with multiple git repos and some\njobs have file filters including smth like ^zuul\\.d.*$\nwhich prevents those jobs to start if the directory\nnamed the other way.\n\nChange-Id: Icc4c84a8ec6d13e67a573c5837e595f787b4cf81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a1deec52361339a013ccc843aff21ed0c2e2df0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a1deec52361339a013ccc843aff21ed0c2e2df0b"}]},"branch":"refs/heads/master"},"1ec7a09e9e15cf0c43e4199d66bbc05a96f7a794":{"kind":"REWORK","_number":2,"created":"2025-01-10 21:54:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/20/938920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/938920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/938920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/938920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/938920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/938920/2"}}},"commit":{"parents":[{"commit":"639122fc1163521a8a37ce3d2a2d8571218c542e","subject":"Drop dances around virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/639122fc1163521a8a37ce3d2a2d8571218c542e"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-10 15:54:19.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-10 21:53:07.000000000","tz":-360},"subject":"Rename .zuul.d -\u003e zuul.d","message":"Rename .zuul.d -\u003e zuul.d\n\nThis is to align it with other Openstack-Helm projects.\nWe use same test jobs with multiple git repos and some\njobs have file filters including smth like ^zuul\\.d.*$\nwhich prevents those jobs to start if the directory\nnamed the other way.\n\nChange-Id: Icc4c84a8ec6d13e67a573c5837e595f787b4cf81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1ec7a09e9e15cf0c43e4199d66bbc05a96f7a794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1ec7a09e9e15cf0c43e4199d66bbc05a96f7a794"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
