)]}'
{"id":"openstack%2Floci~939284","triplet_id":"openstack%2Floci~master~I2b43835541faf9c193cd5daff45bf7fca11304f2","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-01-15 19:28:26.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I2b43835541faf9c193cd5daff45bf7fca11304f2","subject":"Temporarily run two promote jobs for docker.io and quay.io","status":"MERGED","created":"2025-01-14 21:56:37.000000000","updated":"2025-02-11 13:08:43.000000000","submitted":"2025-01-15 21:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939284","meta_rev_id":"decf8b8ae7461868b6711b3246c9f1e89821afd5","_number":939284,"virtual_id_number":939284,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-15 21:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2025-01-15 16:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2025-01-15 19:28:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-15 00:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 16:03:51.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"c5b102ebddee6f120c36ac2dff3a0bbb79c50241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-14 21:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baa8e17c9f3dfa0ea46ce88258b8c50d2826a398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-14 21:57:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bd4988f90f34bd2e60c8b1dab049052a074196a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-14 21:58:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"774b3fae8e13e50a1947de091628d927b5b37d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 00:38:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f91de9f692164f05baa5854df04e5a5f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ab8aac15ae08422fbd33e9981261d4a7 : SUCCESS in 2h 38m 45s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/80c0ad6ffc714a3db6b6328950f28681 : SUCCESS in 1h 00m 05s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/68893b12ef584f2c9ed9e00c17148b47 : SUCCESS in 1h 08m 22s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0d800ee5b64e48008e594fa519c03739 : SUCCESS in 57m 42s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/71396dafd4a2409f93b280da462b3947 : SUCCESS in 55m 32s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/55ad0e48e10d483394d453091bc1cc93 : SUCCESS in 1h 12m 42s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/183a6ae241db4a5a9393752a401ca62f : SUCCESS in 57m 21s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d337f30da9640eca983b8653792c787 : SUCCESS in 1h 02m 33s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/dd7ab24d6f4d4b49b3bb1ece6311a4ca : SUCCESS in 59m 22s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89bfecc125f94cb2aed90717f33bc1e1 : SUCCESS in 1h 02m 50s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4a2cfd7c74e0444bb3abdbd1be707893 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":3},{"id":"91e3a988c350eb95ad57b9dd24bdfa5e4a4e80b4","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-01-15 16:03:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a95e4526bcbcd688db16e94739c4658f18c979e2","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-15 19:28:26.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"007193e0c61ef22c7c7425b1933815339aa980c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 19:28:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85c2e3aac33d2518b3137aa67bbaac1899209183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 21:00:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/351e05062f3d47d3b2f3559d1f187591\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c9b9d964c1aa47dca67115ab7b2bcf93 : SUCCESS in 1h 31m 20s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/98295df9359141e9a0677f3175e387c6 : SUCCESS in 1h 11m 51s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/780caf7b761c414286b23f1e6940b928 : SUCCESS in 1h 19m 30s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/11ccb7893d474f5da81b6d2fd0b26544 : SUCCESS in 1h 22m 28s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d838edcbc86c472098b9fe1188087e62 : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1a2d6d7163a6f35abab555b341af81b55745cdd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 21:00:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3f027c2823b634876ac9c1d2386cbe297426eb14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 21:12:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb63ce697eaa436aada30f73c15a39d3\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/be32ace2d2bf44b5bd1640b4d1fc178e : SUCCESS in 5m 24s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/228f36d9250e46bd8916b8e417e884f3 : SUCCESS in 5m 14s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4a3de3583ea742ab96669b36d1add647 : SUCCESS in 5m 00s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/aadec6e735ca476b9296798aa3b1f525 : SUCCESS in 4m 38s\n- loci-promote-2023-2-ubuntu_jammy-quay https://zuul.opendev.org/t/openstack/build/af84c3e943d544629b7a06f92c41601b : SUCCESS in 10m 15s\n- loci-promote-2024-1-ubuntu_jammy-quay https://zuul.opendev.org/t/openstack/build/6f185c07a02a495090835301a237253c : SUCCESS in 9m 15s\n- loci-promote-2024-2-ubuntu_jammy-quay https://zuul.opendev.org/t/openstack/build/f7a5920d49a24361b781126328755c02 : SUCCESS in 9m 55s\n- loci-promote-master-ubuntu_jammy-quay https://zuul.opendev.org/t/openstack/build/c54873d6d0df4ae48fc8c912d0561d02 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":3},{"id":"decf8b8ae7461868b6711b3246c9f1e89821afd5","tag":"autogenerated:gerrit:revert","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-02-11 13:08:43.000000000","message":"Created a revert of this change as Ie50e4473b9f733774200dd3d1fc3e98036bd2b90","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"455c06b78fca3312f048aad75cc4c9a5e7a61968","revisions":{"e85c1c51d879e1f99e6fb230300fb9379e88874a":{"kind":"REWORK","_number":1,"created":"2025-01-14 21:56:37.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/84/939284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/84/939284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/84/939284/1"}}},"commit":{"parents":[{"commit":"37fd93c74f33a9c944060f9e9fa42583346c8984","subject":"Add quay.io api token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/37fd93c74f33a9c944060f9e9fa42583346c8984"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-14 21:32:07.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-14 21:56:25.000000000","tz":-360},"subject":"Temporary run two separate promote jobs for docker.io and quay.io","message":"Temporary run two separate promote jobs for docker.io and quay.io\n\nAfter this https://review.opendev.org/c/zuul/zuul-jobs/+/881524\nis merged we will be able to run a single promotion job that\nloops over two registries docker.io and quay.io\n\nChange-Id: I2b43835541faf9c193cd5daff45bf7fca11304f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e85c1c51d879e1f99e6fb230300fb9379e88874a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e85c1c51d879e1f99e6fb230300fb9379e88874a"}]},"branch":"refs/heads/master"},"d9e1a09247faa7099461ee27a467c9676b90e690":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-14 21:57:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/84/939284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/84/939284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/84/939284/2"}}},"commit":{"parents":[{"commit":"37fd93c74f33a9c944060f9e9fa42583346c8984","subject":"Add quay.io api token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/37fd93c74f33a9c944060f9e9fa42583346c8984"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-14 21:32:07.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-14 21:56:49.000000000","tz":-360},"subject":"Temporarily run two promote jobs for docker.io and quay.io","message":"Temporarily run two promote jobs for docker.io and quay.io\n\nAfter this https://review.opendev.org/c/zuul/zuul-jobs/+/881524\nis merged we will be able to run a single promotion job that\nloops over two registries docker.io and quay.io\n\nChange-Id: I2b43835541faf9c193cd5daff45bf7fca11304f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d9e1a09247faa7099461ee27a467c9676b90e690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d9e1a09247faa7099461ee27a467c9676b90e690"}]},"branch":"refs/heads/master"},"455c06b78fca3312f048aad75cc4c9a5e7a61968":{"kind":"NO_CHANGE","_number":3,"created":"2025-01-14 21:58:41.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/84/939284/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/84/939284/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/84/939284/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/84/939284/3"}}},"commit":{"parents":[{"commit":"ddc70da7d554d705a4b2f6960f1730b1d94f5067","subject":"Add quay.io api token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ddc70da7d554d705a4b2f6960f1730b1d94f5067"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-14 21:32:07.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-14 21:58:41.000000000","tz":0},"subject":"Temporarily run two promote jobs for docker.io and quay.io","message":"Temporarily run two promote jobs for docker.io and quay.io\n\nAfter this https://review.opendev.org/c/zuul/zuul-jobs/+/881524\nis merged we will be able to run a single promotion job that\nloops over two registries docker.io and quay.io\n\nChange-Id: I2b43835541faf9c193cd5daff45bf7fca11304f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/455c06b78fca3312f048aad75cc4c9a5e7a61968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/455c06b78fca3312f048aad75cc4c9a5e7a61968"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
