)]}'
{"id":"openstack%2Floci~945047","triplet_id":"openstack%2Floci~master~I4e9cbbe0a1af1e677ebd3d96d80645d329007502","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-03-27 21:30:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4e9cbbe0a1af1e677ebd3d96d80645d329007502","subject":"Remove openstack-helm-infra from job definitions","status":"ABANDONED","created":"2025-03-19 19:05:28.000000000","updated":"2025-03-27 21:30:07.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f03cf12dbf307cc5baddfc980169504b6268be9f","_number":945047,"virtual_id_number":945047,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-19 22:35:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2025-03-27 16:40:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-19 20:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0f91d833c01525b1d679a3ec45e48a692e1ae9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-19 19:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a34779b444ccceba1573d9124fb546d8183583f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:14:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ef62b90b8a14cdc9d7092698ba1f5dc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c5ebb45059254e8aac71f368692c0bde : SUCCESS in 1h 04m 01s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8f259627824f452fa6db47f6a067c92d : FAILURE in 38m 51s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fc9c7d8b09074b628d6af5b364b27d3e : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4e37af9e059b483dbe5e2c6a7a14ccdc : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/988d4fcfc51745e6b7ad7d3de861f2c2 : FAILURE in 37m 03s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a057ddb13c1b4ad79281a5da5c10f6b8 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/149041aa7dc64d34ad34128b6586fc36 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/680f25e6e77b46bfb1367e38d3c06f43 : FAILURE in 14m 26s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/209b1b89f20441809f81f7c06dc6698b : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7733a160d25749a79b8668a581fa3a5b : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7f836dbc51d44fd099ceb8ce6559efbf : SUCCESS in 56m 40s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d2ce05454dcc4985b8ca122c521b92a6 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7080ab8f36db205c0c606de3cdaa7dc1f16e50f9","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-19 20:38:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5180208d419094fc142f11a1b9116a60016b6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:55:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2e30eaec1f44bf8aa75cb648bb83f0f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/55ab9a5c7cb5429ab700ae909b6fdd8d : SUCCESS in 15m 56s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/38885d44a02f44f49a9a0bec7dca823f : FAILURE in 11m 40s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b90d4102dd89425da937f642ad823d9e : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c3b09437dc0e40ce813f350cc2cb186e : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/429c04e4c3714a49b523f23fa4a5c7e8 : FAILURE in 2m 43s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0609f78cde144531b1c211b0dd44c8ca : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b12bbadfe78b4955928548b5b61c6e21 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bad34378a24d4ab0b905cc9ea1d571e8 : FAILURE in 2m 36s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/16a3a3fdec934d1cba0ffcf747ad8be2 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b7eb3fb4938342e9ba2c1765d1a173f0 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ff1a882eaaab4795824a7d879d19e47d : FAILURE in 2m 54s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a0e4004ef0a849219df8acaeae280294 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fdbca808c668706b713da23ff2b5f81d85fbd9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-19 21:20:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"71185987166795e894d777376d0223b7ae6d5d1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-19 21:27:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2b518622d24b2065f920c7f342e6682f28a54b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 22:35:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbdecf04e62443b583287e2a06618c08\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d64e35f739324b99ab028c6be9b799fc : SUCCESS in 1h 07m 37s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e9da7cd048ab400aaca18441d199a1ff : FAILURE in 15m 38s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c01169d04dcc452dbe5ef952bc2f4d7e : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/53570305841845ae8f53a20b5e5f10dd : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9e7dbf6bb910489ba1c0b1e18f04a6a6 : FAILURE in 17m 01s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/61e58fee185443e5b39879b0c63e2fb9 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ad85929e8c4f41298c50a040db581191 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a1bd1c8de0994917b461d464dbbf748b : FAILURE in 14m 31s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6a87758bcf934d47b4b6d3b529b0e358 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fb153a793a364ba5983bf595bcca39c5 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ec2d861195714b419acfccd1d63cbd1e : SUCCESS in 1h 01m 57s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dbbc92b9d462439ca6afd1abff6bea90 : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":3},{"id":"79162d55fa7a13ff7e3118829bf5cea3c9f66c51","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-20 15:14:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d426254fab599b3d44cf370053027c70c49847c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:43:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dbb6c4be67c406ea60468c713a2aa0b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/96e09564e1c6478caedbf0439504e2ba : SUCCESS in 2h 28m 17s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7e5943c09e7e41999be9b1af7644c04a : SUCCESS in 33m 16s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/32e0ceac74e9465496f8d9c1f2edb22b : SUCCESS in 1h 07m 49s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fefb6d15c7f24f2b81de316b13356d24 : SUCCESS in 37m 46s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d14d209012c44ff6ac042811803a44ff : SUCCESS in 51m 38s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d4966ba6f3bb44dc9bbc18c71e9bff00 : SUCCESS in 1h 10m 56s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ab5765ca676244f081c4b47be1a24f50 : SUCCESS in 38m 21s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/257027fa798542ad899c625f3f81ed58 : FAILURE in 12m 20s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f0b6334cf7294de599f42b68795250cb : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/57e6b53d05e64fb3a8b8dd5baecdd204 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1986b68b6ed048ea9603a8575457a49c : SUCCESS in 55m 38s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8fed6366652644d1a23d4fb18ceb96f7 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5bcee5f175429264f592d456190684a6eb60fb90","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-20 17:54:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bc11a3b8e84f797cf3e34befc95aa7017e3333e","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-20 18:24:20.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"0c34d5b608d233c3c5aee2a25e2dbc0c6695793c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 20:47:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0b5803b5f4b441e9b48c335b4087f26\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/fde79969e7e64779ba4b8e04957f9133 : SUCCESS in 2h 48m 19s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8587eae331ff46b09b2827cb07085d9c : SUCCESS in 1h 02m 55s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/47138c849f604b0c87c1eb13105ed9fe : SUCCESS in 1h 18m 12s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/232511ff07ab40008537523f8d470f84 : SUCCESS in 44m 24s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/edcc39d99e994f139a61254a44e52df6 : SUCCESS in 57m 14s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4d3cb116da60489bab2d02f49951cb9b : SUCCESS in 1h 19m 57s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3c883718a15445fdb2683a1e85bf726f : SUCCESS in 41m 23s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7ee3d5543b0a4cccb263922820e9996d : FAILURE in 15m 37s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3dda5f1e0cc7407ea9c9836f4e8ce1f0 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/25652df6989740e4aaa3ab9f2830f181 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/280b4145523b4ef8ba4dc47a0ec70777 : SUCCESS in 1h 03m 21s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/98c59c04267e4b32b6535976483c9394 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":3},{"id":"cbea25a24c0a4c46cadfc6decb3da9a16a374c76","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-27 16:40:24.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"926c1b7f488879534601df1fbd3ec6e13023177a","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-27 16:40:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e04ca5b35222e22f8fe8363c7e34727ccf6669cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 19:18:30.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83ea24f3fb5f4711bc02ef10b7ebc368\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ed19b3d08b574a2aa2fc9f078b4ee625 : SUCCESS in 2h 17m 43s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9f7b2450004243248372cd6c7cf7e729 : SUCCESS in 53m 48s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5265814de5ce4cd1a1bf52df88665b2e : SUCCESS in 1h 10m 05s\n- loci-cinder-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/62904e344df84bf680d7443c2cd64393 : SUCCESS in 42m 33s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3d3427397800436f9e95e35b3e5488ad : SUCCESS in 56m 23s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/169be8e8d4da44ac8a66cea420ef460f : SUCCESS in 1h 07m 31s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1aabfdad4368408ea719c8d6a72d0b87 : SUCCESS in 40m 01s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/723b33758b9c4280ae284a94b9a5e5a3 : FAILURE in 12m 43s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d3532ff8efab47bda534999dc4bf306b : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5959cc2593a64b97a8f30516d504a1b4 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c5019ed90c81400686755cf95747b54a : SUCCESS in 55m 48s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2de89a0806d644ecaac5219197e81d57 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":3},{"id":"f03cf12dbf307cc5baddfc980169504b6268be9f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-27 21:30:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7694fb108636173ae2a9626308a3d8d02108de47","revisions":{"2bb72cd807e0f0c1412e96ca21d4149f7cb80d21":{"kind":"REWORK","_number":1,"created":"2025-03-19 19:05:28.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/47/945047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/945047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/945047/1"}}},"commit":{"parents":[{"commit":"f9c47d05e07cdb2c36bf453c6a731cd2bac3584a","subject":"Update glance image build variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9c47d05e07cdb2c36bf453c6a731cd2bac3584a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-19 19:02:18.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-19 19:02:18.000000000","tz":-300},"subject":"Remove openstack-helm-infra from job definitions","message":"Remove openstack-helm-infra from job definitions\n\nWe recently merged all the code from the openstack-helm-infra\ninto the openstack-helm. Openstack-helm-infra will be retired\nsoon. We don\u0027t need it in the list of required projects for\ntest image jobs.\n\nChange-Id: I4e9cbbe0a1af1e677ebd3d96d80645d329007502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2bb72cd807e0f0c1412e96ca21d4149f7cb80d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2bb72cd807e0f0c1412e96ca21d4149f7cb80d21"}]},"branch":"refs/heads/master"},"4824f1c498653c264221401c8d920e9b9953920d":{"kind":"REWORK","_number":2,"created":"2025-03-19 21:20:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/47/945047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/945047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/945047/2"}}},"commit":{"parents":[{"commit":"f9c47d05e07cdb2c36bf453c6a731cd2bac3584a","subject":"Update glance image build variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9c47d05e07cdb2c36bf453c6a731cd2bac3584a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-19 19:02:18.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-19 21:19:57.000000000","tz":-300},"subject":"Remove openstack-helm-infra from job definitions","message":"Remove openstack-helm-infra from job definitions\n\nWe recently merged all the code from the openstack-helm-infra\ninto the openstack-helm. Openstack-helm-infra will be retired\nsoon. We don\u0027t need it in the list of required projects for\ntest image jobs.\n\nChange-Id: I4e9cbbe0a1af1e677ebd3d96d80645d329007502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4824f1c498653c264221401c8d920e9b9953920d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4824f1c498653c264221401c8d920e9b9953920d"}]},"branch":"refs/heads/master"},"7694fb108636173ae2a9626308a3d8d02108de47":{"kind":"REWORK","_number":3,"created":"2025-03-19 21:27:11.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/47/945047/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/945047/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/945047/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/945047/3"}}},"commit":{"parents":[{"commit":"f9c47d05e07cdb2c36bf453c6a731cd2bac3584a","subject":"Update glance image build variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f9c47d05e07cdb2c36bf453c6a731cd2bac3584a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-19 19:02:18.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-19 21:27:05.000000000","tz":-300},"subject":"Remove openstack-helm-infra from job definitions","message":"Remove openstack-helm-infra from job definitions\n\nWe recently merged all the code from the openstack-helm-infra\ninto the openstack-helm. Openstack-helm-infra will be retired\nsoon. We don\u0027t need it in the list of required projects for\ntest image jobs.\n\nChange-Id: I4e9cbbe0a1af1e677ebd3d96d80645d329007502\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7694fb108636173ae2a9626308a3d8d02108de47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7694fb108636173ae2a9626308a3d8d02108de47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
