)]}'
{"id":"openstack%2Floci~945994","triplet_id":"openstack%2Floci~master~Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-04-03 16:51:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462","subject":"Build openstack-client image","status":"MERGED","created":"2025-03-31 19:33:56.000000000","updated":"2025-04-03 16:58:44.000000000","submitted":"2025-04-03 16:51:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945994","meta_rev_id":"1a63cc85b5ed9608a4c3f929073bcd9b5877ddb9","_number":945994,"virtual_id_number":945994,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-03 16:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-04-03 14:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-04-03 15:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":1,"date":"2025-04-03 15:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-31 20:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-03 14:41:05.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2025-04-03 15:26:59.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"ac57a0184ba00fd8ae819cf312ca9400a083d201","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 19:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9423698d68b3fdbbf09c159622626c872563b04","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 19:40:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d100b7aae09f38eacc080fc9de1e56f104c042f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 20:01:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b005c54abfe4cf8917158baf664782a\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c812c3f7891843a48835cf4ab91954b7 : SUCCESS in 18m 52s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3131ff696bd64b65977dedcc7cc3da7f : FAILURE in 14m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7019251daf461dc17b28a0894387d65c97f352b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 20:28:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"923c8bbf7f0312d4fb558ced622d415d1b5ecbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 20:58:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5037b8ae29414dbdb11dd20c48bfe8eb\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/496b28b721904396b6b5147439feb858 : SUCCESS in 20m 17s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7d77ef4ff3484a7cada288c09d09771d : FAILURE in 11m 33s","accounts_in_message":[],"_revision_number":3},{"id":"28330711103b0da4b9ee16653eb64857b56e3783","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 21:04:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"770915e88753f4fc31620b45bcc96fd5cad0e3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 21:21:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d224bbb4fc64b588c69f36df9456ef6\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9bc9030705fe4419a083d4b38e6064fd : SUCCESS in 16m 12s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ba6a5532bafd42038d5b81d3c825819b : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":4},{"id":"e26336fb79a186adaf591f5b3d5c407dfd24a48e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 21:41:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b5e2c8224f1c731051bf098181602dcaa5c812b7","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 21:42:18.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"fa1cef11a00e6a3946d74145d5c52ec7c9de75c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 21:45:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a0abc719a6938f79424bd5ad73a55e597f6de226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-03-31 22:15:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"20938ec43c64f2e52e5bb31cf1329acaece9911c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-01 00:01:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3dd9c35a4ced42399d48623af3372df67fad349a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-01 00:41:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4fbda03c0481a6deb51c127f157180d610492b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 00:48:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d36e2f4ce0234f97923f368c274aac03\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f518263d4d294e94aed3130384cb232c : SUCCESS in 5m 28s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a6e32d3e198e4305a093e960fbdb721a : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":9},{"id":"00c5cd8a0bf976ef01cd9ff0d21665c1be704ec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-01 00:56:53.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"3e5ab946b440027af0254017ce4a06314c28fb33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 01:19:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f968bcec945f4d0a935fac5c6f0c04ee\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bce3d2cd26e74e04b077d3966821d806 : SUCCESS in 21m 56s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7c4707de0eb64c32b55f25b13df064dc : FAILURE in 14m 54s","accounts_in_message":[],"_revision_number":10},{"id":"d184aefa29276023e5cb3901ea5b778e2be60234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-02 01:35:59.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"3f9838ef4237a3f0d386b4e502f055b7eee6691a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 04:02:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c32aa35ff5e40ec8bd64fc9818f521d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bce92cf09f3b4591acef74e329f7fd4f : SUCCESS in 2h 25m 47s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bb5dc978a70e4bab878eae50e5ef5098 : SUCCESS in 30m 46s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5d850cbfd0aa4f46a27da660af961b05 : SUCCESS in 47m 32s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/80a4acd6a7b94e59ad30f3662fac2091 : SUCCESS in 31m 20s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/99c7a733e0454c488a2fd913cb44168a : SUCCESS in 47m 56s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/608dac3106664c0c9e462b9239e9a4a1 : SUCCESS in 33m 48s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/57d650bd86aa4dfa93e027a6ccc6ad73 : SUCCESS in 49m 02s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4570282d76aa4a4bae1ec8535dcac89f : SUCCESS in 34m 10s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/468a4b9225f64f489bc8607aef1f3f05 : SUCCESS in 31m 46s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c404de1ba0f549abb2919011b3ba17f6 : SUCCESS in 34m 00s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5b622c1fe65d4e90a7a76e55c4f5b8b9 : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":11},{"id":"afd8cda37cce95999f5fd56dcf8072a58a6e3cc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-02 17:07:45.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"248036141620a351147f5c09227a3e9700458f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 17:22:30.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fa8a7ede6eb4524abe2cdd6a58e7f77\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2777867a5bea43e0882dd24e7c60c74a : SUCCESS in 12m 50s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9d97ea63d241465caa3b41de3abd8e7b : FAILURE in 5m 20s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/983400512de5437e8fc778eca1bc5581 : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6eaa2f3608bc48a6a44eb6f5991d3e74 : FAILURE in 5m 32s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6da4c0c1092341208eb5ef7b631c6404 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bd7b4220daa1451ca08e59ca52f11ccd : FAILURE in 5m 12s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d2968ac32aa04d0cb15a71c41fbf53a3 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a811af4bb99a4a1ebd485ba3e1e30c13 : FAILURE in 5m 04s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/40c7b49b47a34aa7babba7af081f6eb4 : FAILURE in 5m 22s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0222afe71bb24fd5a72f2409b6fce6d7 : FAILURE in 5m 19s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c2dac3b0169b4248afeb135c9752435a : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":12},{"id":"5e14cff39ffd06569bf250434e386312cd5219bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-02 17:28:53.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"975494503a474e3949912a687c2f6466cde43bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 17:45:41.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0347104c6fec4702b9319fff5a8c9af8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7ed702e51e02400999fc76e7cf69e48c : SUCCESS in 11m 23s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8ec1e0ea3c4446a78d5b5f8a55872420 : FAILURE in 3m 39s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4c51625da461442b9f5e5737f85814a9 : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/507b0c85972e4c3e97e1d025efe8ca97 : FAILURE in 3m 59s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/49d21ab5ebce45bbb5734eb0bcb0cfbd : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b910f0623e834644b8bfaa8e1a3647b3 : FAILURE in 3m 53s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e9366c5bd49843f6bfe778a1c4e53ca1 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d629a1096fad41e7b8588911ac7673a3 : FAILURE in 3m 34s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dd893594c9d64a0a95c86033efa90658 : FAILURE in 3m 28s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2980d451408b4fe8b8a479c7696b1513 : FAILURE in 3m 39s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb95f6c8ec184590aaed4c069d8d9162 : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":13},{"id":"97f90cfb78399696e10353a3e44aa09aa20bef35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-02 19:32:46.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"fa5fc929b490ebb18f9cba498346e1e4d78f5180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 19:42:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b68f4527093451c81ec7b2b09d9e078\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b1bb51e3961248c8a819b93872b0a788 : SUCCESS in 8m 13s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3de7b641dbea420b9a2c4c9875f501aa : FAILURE in 3m 43s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/828c795bcb114334b3315ad6ee381eee : SKIPPED Skipped due to failed job loci-build-2023-2-ubuntu_jammy\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/62528db07da1449481608ea60a8709bb : FAILURE in 3m 58s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7bc250c130f845d8bec24f5543a5c007 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8dd21478ae514b8b8e67c53f84d8ec40 : FAILURE in 3m 36s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/46de604f4a8f49999d6a8567abd4504e : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2777a2023e294f7985c820846c7cb889 : FAILURE in 3m 58s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8b084cf3ae214b1bab5fb2da69584fc2 : FAILURE in 3m 36s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5e8097272f3b470981b96f311c6de1f4 : FAILURE in 4m 00s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fa5e8d6f81b545109780b03e9f505342 : FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":14},{"id":"be1b89cbbee3228a86ba4cfef08f1f17ff987b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-02 21:10:13.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"11496ad69748929a346d10a71be1f00fb818a14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 23:37:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d378f5f6fdfb4441997af74429cf83a4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b97d4eeb730140029765308937766c5e : SUCCESS in 2h 22m 30s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0ca5e0c18358422ab4d0036c6549d553 : SUCCESS in 53m 47s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a9eb7cf0cf4c4775b787f0d771d4ce06 : SUCCESS in 1h 12m 11s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8b14f3ec607d4d65b750c7fbc5ae6412 : SUCCESS in 51m 21s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f407a17ad80743cbac1257c026e984c0 : SUCCESS in 1h 10m 56s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/82fff87dbc48413eb085287497ce8ce2 : SUCCESS in 56m 34s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/086798f622b74e75a740b7074a1001e9 : SUCCESS in 1h 12m 44s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5e7d93f75e51485ca47c86bd77611c1f : SUCCESS in 53m 15s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d60b81a2f273457f9d1083e6ee65483e : SUCCESS in 53m 22s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/96736bdbe9e747668f2ca21e1c2d4898 : SUCCESS in 53m 27s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0a9ce8bfbc0845b7b67469deb81589b5 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":15},{"id":"af765ee774fad57e1fdaf6cdfd621b3c0e0bf644","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-03 06:04:41.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"b80d4485d73415b500a5da27d3f33335660b9f1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-03 06:06:24.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"0436ccd655196c0b701271e8cb766f633bf2910e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 08:59:05.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7261c06ebc74de9b6834835ae962cd7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bedd6aff1af84d73a15a560d3b2fde9b : SUCCESS in 2h 39m 03s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e73101a3f8d486da82b5f3014e6caa9 : SUCCESS in 55m 04s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/27ee5d17b4114643affd6d1bf9342e0c : SUCCESS in 1h 17m 15s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9c588c7212d14e57887317220273bce9 : SUCCESS in 53m 53s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/08b8152bf1a141829ebc1a133a61a901 : SUCCESS in 1h 15m 32s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c41c9beadc70420081c34ba5786ecbda : SUCCESS in 54m 04s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b06730b659ad4c749314734273efafe2 : SUCCESS in 1h 17m 42s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/864f0a62f54a4b4292f42096acb6b871 : SUCCESS in 57m 11s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5a1cc50c6ba64b809b45734009bc9dc0 : SUCCESS in 54m 31s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7ad0d01b43644a2a91665a6a9013579d : SUCCESS in 53m 54s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0bc577a805a34b9f8b2996bc817ea8bd : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":17},{"id":"f1aaa86963f5985340eff59c5bd612a22d7c2f88","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-04-03 14:41:05.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"d4df49eaf21151d772059e4ba797e51447b325e4","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-04-03 15:26:59.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"a500fd206546d89fe53286bfda48c9e7a2069406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 15:27:30.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"903098d211183bba31d516bd13b78c262075fcf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 16:51:04.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1c35eace7264ce0b308a348c6362f88\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/498e7adc6da44a3ea5c2da02c7fb4854 : SUCCESS in 1h 19m 20s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/55e75d45211b485a90df212c37b7139b : SUCCESS in 1h 12m 02s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cada5a2b133d46cba013eebdd638e0dc : SUCCESS in 1h 11m 09s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7933d17b98214204b01e9b8a9902061a : SUCCESS in 1h 09m 34s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b0931ab773e34108afb433310688ae07 : SUCCESS in 1h 11m 41s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6e7e5ad650924622bb0ff6d17a268306 : SUCCESS in 1h 04m 59s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/71156a9f38c6487cbab8376f4b8dbd9c : SUCCESS in 1h 05m 04s\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6141e485473545b0bcbaecaef6ec4ac1 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":17},{"id":"363fdb2075b481d80d4864bce1be21f36eca59ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 16:51:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"1a63cc85b5ed9608a4c3f929073bcd9b5877ddb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 16:58:44.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a4f81f6336b499c9f79002263519aaa\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2fb71d963e4a445489444c4d2c7a4062 : SUCCESS in 7m 02s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4a5aed0cabaf4ee2ae11d48eeca274b7 : SUCCESS in 6m 20s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b69ba11e634a490ebe4e22232f8542a0 : SUCCESS in 6m 10s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2467659b66c6450ea9f54bae008d744b : SUCCESS in 6m 07s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6b935912294a4b5dbd7ab44161e40634 : SUCCESS in 6m 08s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/015685b7f59d4b4a94a7abd4d6e7cdd4 : SUCCESS in 5m 59s\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/689b0378ab904183a767c88e6fb8623f : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"2470b7c4ee6290ec6a83f5b23400093fd06a512b","revisions":{"c36fc0fce4d7d1a84f7a608d6af5ea9375bd84d6":{"kind":"REWORK","_number":1,"created":"2025-03-31 19:33:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/1"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c36fc0fce4d7d1a84f7a608d6af5ea9375bd84d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c36fc0fce4d7d1a84f7a608d6af5ea9375bd84d6"}]},"branch":"refs/heads/master"},"07009939c2c7a4226582876c1fea0a41ac29ff1b":{"kind":"REWORK","_number":2,"created":"2025-03-31 19:40:49.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/2"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:40:44.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/07009939c2c7a4226582876c1fea0a41ac29ff1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/07009939c2c7a4226582876c1fea0a41ac29ff1b"}]},"branch":"refs/heads/master"},"cc9139123a6c8eda6d61eb11f669de2a109e6b15":{"kind":"REWORK","_number":3,"created":"2025-03-31 20:28:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/3"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 20:27:55.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cc9139123a6c8eda6d61eb11f669de2a109e6b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cc9139123a6c8eda6d61eb11f669de2a109e6b15"}]},"branch":"refs/heads/master"},"0a21f76f37679ecec27cd9b3a3a80d2d80ffd656":{"kind":"REWORK","_number":4,"created":"2025-03-31 21:04:51.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/4"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 21:04:44.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0a21f76f37679ecec27cd9b3a3a80d2d80ffd656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0a21f76f37679ecec27cd9b3a3a80d2d80ffd656"}]},"branch":"refs/heads/master"},"cfddb288eaf653b6d9d13b3e2657b0c7e475aae4":{"kind":"REWORK","_number":5,"created":"2025-03-31 21:41:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/5"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 21:41:39.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cfddb288eaf653b6d9d13b3e2657b0c7e475aae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cfddb288eaf653b6d9d13b3e2657b0c7e475aae4"}]},"branch":"refs/heads/master"},"11473df870e7ad0cd693b3e09c84a5e8f79b4a60":{"kind":"REWORK","_number":6,"created":"2025-03-31 21:45:38.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/6"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 21:45:33.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/11473df870e7ad0cd693b3e09c84a5e8f79b4a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/11473df870e7ad0cd693b3e09c84a5e8f79b4a60"}]},"branch":"refs/heads/master"},"3af378fb2264066f7b6a019a1012c4acaf865e21":{"kind":"REWORK","_number":7,"created":"2025-03-31 22:15:20.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/7"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 22:15:14.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3af378fb2264066f7b6a019a1012c4acaf865e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3af378fb2264066f7b6a019a1012c4acaf865e21"}]},"branch":"refs/heads/master"},"e77b91bae960692cc2da4d319415b9c2de39a653":{"kind":"REWORK","_number":8,"created":"2025-04-01 00:01:23.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/8"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-01 00:01:18.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e77b91bae960692cc2da4d319415b9c2de39a653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e77b91bae960692cc2da4d319415b9c2de39a653"}]},"branch":"refs/heads/master"},"884b6ddb462eaf98c0775b282bd9f382ad98d558":{"kind":"REWORK","_number":9,"created":"2025-04-01 00:41:55.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/9"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-01 00:41:36.000000000","tz":-300},"subject":"[WIP] Build openstack-client image","message":"[WIP] Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/884b6ddb462eaf98c0775b282bd9f382ad98d558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/884b6ddb462eaf98c0775b282bd9f382ad98d558"}]},"branch":"refs/heads/master"},"674912163e22faeb978c8768e65e465cd882bdbf":{"kind":"REWORK","_number":10,"created":"2025-04-01 00:56:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/10"}}},"commit":{"parents":[{"commit":"39e2b65f17e335ed86ae40ad685d75aa07cc57b9","subject":"Build 2025.1 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/39e2b65f17e335ed86ae40ad685d75aa07cc57b9"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-01 00:56:47.000000000","tz":-300},"subject":"[WIP] Build openstack-client image","message":"[WIP] Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or some unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/674912163e22faeb978c8768e65e465cd882bdbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/674912163e22faeb978c8768e65e465cd882bdbf"}]},"branch":"refs/heads/master"},"2e0f000cf9c955b5d0480d806ab948364901af0e":{"kind":"REWORK","_number":11,"created":"2025-04-02 01:35:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/11"}}},"commit":{"parents":[{"commit":"c60d0716156efda4d5433c59932daf5f046b49cc","subject":"Use head of stable/2025.1 for python-openstackclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c60d0716156efda4d5433c59932daf5f046b49cc"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-02 01:27:35.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e0f000cf9c955b5d0480d806ab948364901af0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e0f000cf9c955b5d0480d806ab948364901af0e"}]},"branch":"refs/heads/master"},"ec7601a65811670a5d2a891aa169f8ee20a381b2":{"kind":"REWORK","_number":12,"created":"2025-04-02 17:07:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/12"}}},"commit":{"parents":[{"commit":"c60d0716156efda4d5433c59932daf5f046b49cc","subject":"Use head of stable/2025.1 for python-openstackclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c60d0716156efda4d5433c59932daf5f046b49cc"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-02 17:07:40.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec7601a65811670a5d2a891aa169f8ee20a381b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec7601a65811670a5d2a891aa169f8ee20a381b2"}]},"branch":"refs/heads/master"},"cef3ba12d4fc3f03b6f89122524b7f71a6af135b":{"kind":"REWORK","_number":13,"created":"2025-04-02 17:28:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/13"}}},"commit":{"parents":[{"commit":"c60d0716156efda4d5433c59932daf5f046b49cc","subject":"Use head of stable/2025.1 for python-openstackclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c60d0716156efda4d5433c59932daf5f046b49cc"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-02 17:28:47.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cef3ba12d4fc3f03b6f89122524b7f71a6af135b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cef3ba12d4fc3f03b6f89122524b7f71a6af135b"}]},"branch":"refs/heads/master"},"c62532ef1499c17b828bb277b55a96b22b3c64e8":{"kind":"REWORK","_number":14,"created":"2025-04-02 19:32:46.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/14"}}},"commit":{"parents":[{"commit":"c60d0716156efda4d5433c59932daf5f046b49cc","subject":"Use head of stable/2025.1 for python-openstackclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c60d0716156efda4d5433c59932daf5f046b49cc"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-02 19:32:31.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c62532ef1499c17b828bb277b55a96b22b3c64e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c62532ef1499c17b828bb277b55a96b22b3c64e8"}]},"branch":"refs/heads/master"},"2b89637d512d50ce3031df65ff2f717da70171de":{"kind":"REWORK","_number":15,"created":"2025-04-02 21:10:13.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/15"}}},"commit":{"parents":[{"commit":"c60d0716156efda4d5433c59932daf5f046b49cc","subject":"Use head of stable/2025.1 for python-openstackclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c60d0716156efda4d5433c59932daf5f046b49cc"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-02 21:10:07.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2b89637d512d50ce3031df65ff2f717da70171de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2b89637d512d50ce3031df65ff2f717da70171de"}]},"branch":"refs/heads/master"},"271ff6ba3204b0c38b7f87e10b4f6801f45efdde":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-04-03 06:04:41.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/16"}}},"commit":{"parents":[{"commit":"9f49737381052e8eaeb262999643f9d49dda3270","subject":"Use python3 venv module instead of virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9f49737381052e8eaeb262999643f9d49dda3270"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-03 06:04:41.000000000","tz":0},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/271ff6ba3204b0c38b7f87e10b4f6801f45efdde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/271ff6ba3204b0c38b7f87e10b4f6801f45efdde"}]},"branch":"refs/heads/master","description":"Rebase"},"2470b7c4ee6290ec6a83f5b23400093fd06a512b":{"kind":"REWORK","_number":17,"created":"2025-04-03 06:06:24.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/945994/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/94/945994/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/94/945994/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/94/945994/17"}}},"commit":{"parents":[{"commit":"9f49737381052e8eaeb262999643f9d49dda3270","subject":"Use python3 venv module instead of virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9f49737381052e8eaeb262999643f9d49dda3270"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-03-31 19:27:31.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-03 06:06:18.000000000","tz":-300},"subject":"Build openstack-client image","message":"Build openstack-client image\n\nThe motivation behind is to have a consistent\nopenstack-client image related to a release or\nwith some changes being tested.\n\nLoci builds so called `requirements` image that\ncontains a consistent set of packages related to\na particular release or from unreleased changes (for testing).\nThen we install these packages from the `requirements`\nimage to all other images to have a consistent set\nof images.\n\nChange-Id: Ie8dac9fed8b09f5b6f2530c22ef9445d5c5cd462\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2470b7c4ee6290ec6a83f5b23400093fd06a512b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2470b7c4ee6290ec6a83f5b23400093fd06a512b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
