)]}'
{"id":"openstack%2Floci~947149","triplet_id":"openstack%2Floci~master~Icbee13fef089362a6e3c0e56c176cc5a0e909ef8","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-04-14 22:45:06.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"Icbee13fef089362a6e3c0e56c176cc5a0e909ef8","subject":"Install libpcre3 on Ubuntu Noble images","status":"MERGED","created":"2025-04-14 19:39:56.000000000","updated":"2025-04-14 23:56:58.000000000","submitted":"2025-04-14 23:51:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947149","meta_rev_id":"cfc028e3573b09932c5a373f18c7ca5953248754","_number":947149,"virtual_id_number":947149,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-14 23:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2025-04-14 21:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2025-04-14 22:45:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 21:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-14 21:37:19.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"6d559d2fd11bfd499625754747af0fda0d4b069f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-14 19:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f97205ab2cf562636c1d8027de1fd8fe5edf7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 21:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f79d8c4c5842bca2d082777f18aa4b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a594d34f08a349f4822ec1d9a414170f : SUCCESS in 1h 55m 49s (non-voting)\n- loci-build-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9de6363e524843bd9ee4eb3e03c1a8cd : SUCCESS in 47m 57s\n- loci-compute-kit-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e74c60ff164843b697374023c11051c1 : SUCCESS in 58m 59s\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/12dd237f711c4e8c9e42a9be38dece54 : SUCCESS in 48m 28s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7659a51f7e38488c935e8ef2a90ead25 : SUCCESS in 55m 30s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b70f4a44250b4b0da0498cdafe3ed672 : SUCCESS in 37m 43s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9548588e85e748d4a1fe115f800f561b : SUCCESS in 1h 00m 08s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6baae0b7b4e64b73a646de322784ad37 : SUCCESS in 48m 14s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/469161e320654d8881c4da13f2e699e8 : SUCCESS in 48m 00s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f6ccc28066ae4e1e9f6e48a42ab47ed1 : SUCCESS in 39m 23s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/05e41f70d0bb48cabe2ed4f2885f7386 : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a5a40e394c66ecdf9612540cb823966873aa6e97","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-04-14 21:37:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cbebbf361bb7f73fe7fe9fa93cab159d9a3e75a3","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-04-14 22:45:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6a436b95f18c9de9c6b723ea599eb8889abb985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 22:45:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7693ac8c4cd12e8cf0dd66a583c618311999801b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 23:51:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80568f97a1d44b16838b032d76279b80\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b88ff9197ea449f8902c17d3a74b2a9e : SUCCESS in 1h 06m 13s (non-voting)\n- loci-upload-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0f3c5f9c4804448cad8ae3415b14eea1 : SUCCESS in 47m 48s\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cca576217acb4e48a81fca2b43e702ae : SUCCESS in 56m 38s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0fa21c3d9169482e874c4a42607a64c3 : SUCCESS in 55m 49s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d6562f6efe354ba1b75876bb78eefdc2 : SUCCESS in 55m 40s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a3a71a7708ad4388b3b3d63073169c10 : SUCCESS in 56m 34s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a505bcf6eb8f4d4c918f339e9790ab14 : SUCCESS in 47m 57s\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/88a13a2688cd4f7eb69693031adf4638 : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7eb8a421d38921bcc8d297413e7811a71f4a7a69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 23:51:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cfc028e3573b09932c5a373f18c7ca5953248754","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 23:56:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0320f2b8eb9b43b7856568544f11a437\n\n- loci-promote-2023-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e4bbb4fd7d3e4f14b5c64475e4799606 : SUCCESS in 5m 00s\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/91df5b6100a84b62a456b151d7803075 : SUCCESS in 4m 46s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7a900975563846c4b8121a6906526932 : SUCCESS in 5m 07s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3bae0a4c08294381ae7344e09e59a03b : SUCCESS in 5m 02s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e7fa8731bf8446998d244e09839e45a4 : SUCCESS in 5m 04s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/312d2cd2a3f443fc815c1c2f64ac89f6 : SUCCESS in 4m 50s\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/31cd5e10f6d64c2dbde360f33f4a1272 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3be9b941395c84fa0e136db36415f77d4671f7b1","revisions":{"3be9b941395c84fa0e136db36415f77d4671f7b1":{"kind":"REWORK","_number":1,"created":"2025-04-14 19:39:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/49/947149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/49/947149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/49/947149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/49/947149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/49/947149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/49/947149/1"}}},"commit":{"parents":[{"commit":"9f49737381052e8eaeb262999643f9d49dda3270","subject":"Use python3 venv module instead of virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9f49737381052e8eaeb262999643f9d49dda3270"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-14 19:38:59.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-04-14 19:38:59.000000000","tz":-300},"subject":"Install libpcre3 on Ubuntu Noble images","message":"Install libpcre3 on Ubuntu Noble images\n\nIt is not installed by default but uwsgi needs it\n\nChange-Id: Icbee13fef089362a6e3c0e56c176cc5a0e909ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3be9b941395c84fa0e136db36415f77d4671f7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3be9b941395c84fa0e136db36415f77d4671f7b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
