)]}'
{"id":"openstack%2Floci~949695","triplet_id":"openstack%2Floci~master~I62028c877736e823f749f0ee46c6bc60caac0dc6","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I62028c877736e823f749f0ee46c6bc60caac0dc6","subject":"[WIP] Mount requirements","status":"ABANDONED","created":"2025-05-14 04:30:32.000000000","updated":"2025-07-03 09:12:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"575e915fb19dc0a94aa4818f0f95cb8a189f0797","_number":949695,"virtual_id_number":949695,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-15 03:34:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-05-14 04:34:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b4b8e572ef29b9b13696190687696ee4405f71f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 04:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2369fc371d87ca564f983ba2bdfde1f0658c85b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 04:34:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13beba6ba7d647fc821e9e8ca375ef9d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/46b031141a8e45fcb15fc8f671266876 : SUCCESS in 3m 49s (non-voting)\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c0774dcb5d6a4e608d7a33c368a5f79b : FAILURE in 1m 07s","accounts_in_message":[],"_revision_number":1},{"id":"53b193d0ec88daa4c15c328fbc17768623e85578","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 15:53:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8535eea2a9cc349e4c655dada80161cf1a3e5cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 16:10:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ab4ee8ca5d74138aedd9cf38bd44950\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4e95ee06afab490ca569a2a7d85fc525 : SUCCESS in 16m 22s (non-voting)\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/880b96d479354b67a576de202b610666 : FAILURE in 13m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ddfda57d20b61fa195f79762d35544805b1abfcf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 17:58:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5bd0c141fe202a2e097417b12e8908a08718c958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 18:20:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7a5dda26554d229c8c3090c80c0776\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/42c88c1df70143968691b9eed58bd299 : SUCCESS in 21m 21s (non-voting)\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0c8627b850ed46c69a7bcf239e24c5cf : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":3},{"id":"495feef8e1f44d58fbebd21300b0034fbd5bcf86","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 18:46:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a7292f1db324217308b28ec9004cee89e7c0a6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 18:58:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/382e3670396e4e8f8427a28ed2850025\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3718a57da688483b95ee5f22315f06fb : SUCCESS in 9m 14s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/edeb5a39bece4653b6dd52d67782f2a3 : FAILURE in 5m 23s","accounts_in_message":[],"_revision_number":4},{"id":"2c4f5650507d71d777322cbb4bdd013ed2b8b694","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 19:03:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"33e3066ef1b82ea58f2367bfe511664f106e48a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:21:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/707d102760294a469ae1660d5c746bb4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/637d4a66745c4a8fb9a9ecd2b8e99b36 : SUCCESS in 17m 44s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cdd5def3051e49429cb47edc14b9af03 : FAILURE in 11m 52s","accounts_in_message":[],"_revision_number":5},{"id":"d91cbcf25bf08b6ba7c7f03b4bcc620caa2b0635","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 19:31:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"84ed3b422190134ed3c31808d5f476655913e0da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:46:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed59bf44c3a14ab497e4178fde6b6a0f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/22eb8db5e2514e84b0ead4b2ccf49752 : SUCCESS in 12m 57s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/549b75b865d3444d8f506e4c9bbb6b63 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":6},{"id":"76f41a73d3c9eae104bd046db82eadce690cb4eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 20:21:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"9aa48720965dc8f71a1c0ffd3fc2d7878928a06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 20:29:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e425acf3a334fc5b51209edad011358\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/48eb925d183341aab0fcaf9fe7c9c817 : SUCCESS in 7m 16s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/42edcf49587e47cbae04eaf3f849c5f2 : FAILURE in 4m 05s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/914e694fe3e84d139c61004cb4de87e8 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy","accounts_in_message":[],"_revision_number":7},{"id":"4a318ceb7b3aa008d331a1bdc9df01726758a921","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-14 23:20:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1eafd6e97ae750226230aeae93f4b1f42910b6df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 00:33:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6311ada18a14aec849309331e3b225c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ce8185b545f94522bfff59bd466fd11d : SUCCESS in 1h 12m 13s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/daa4fd031d42454191e064660f3d029a : SUCCESS in 23m 10s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4fbc19fa80974a618cca273871ca8ced : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":8},{"id":"7b9870d089e812799cae64ec8a59609ab7cf160c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 01:11:23.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"29a443d1ee11338e9b8757dd80a0252e87c8c566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:34:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203f3c7505524addb7e4013d862fcecc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4391e3c4fcf44bb894cb3ae7a25bf3bb : SUCCESS in 2h 18m 25s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1789b19918254cd1a7d0aede9a7d658b : SUCCESS in 41m 21s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/43dd630982484543a5c8933867c2287d : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":9},{"id":"575e915fb19dc0a94aa4818f0f95cb8a189f0797","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2025-07-03 09:12:51.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 49 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"948a631905b83b8fd4fd29a124b8874c4f042913","revisions":{"d8197c98ff577d77b58318c7025dc6ba0e5b648b":{"kind":"REWORK","_number":1,"created":"2025-05-14 04:30:32.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/1"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d8197c98ff577d77b58318c7025dc6ba0e5b648b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d8197c98ff577d77b58318c7025dc6ba0e5b648b"}]},"branch":"refs/heads/master"},"28b81dab355024168d1726bc4cb21c13268f4653":{"kind":"REWORK","_number":2,"created":"2025-05-14 15:53:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/2"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 15:53:25.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/28b81dab355024168d1726bc4cb21c13268f4653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/28b81dab355024168d1726bc4cb21c13268f4653"}]},"branch":"refs/heads/master"},"6737d82a4643047a60c625e267866f518b433c62":{"kind":"REWORK","_number":3,"created":"2025-05-14 17:58:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/3"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 17:58:52.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6737d82a4643047a60c625e267866f518b433c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6737d82a4643047a60c625e267866f518b433c62"}]},"branch":"refs/heads/master"},"eec098294406a3b6a113f7ed4440a0933b1ec922":{"kind":"REWORK","_number":4,"created":"2025-05-14 18:46:49.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/4"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 18:46:43.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eec098294406a3b6a113f7ed4440a0933b1ec922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/eec098294406a3b6a113f7ed4440a0933b1ec922"}]},"branch":"refs/heads/master"},"673abdd804e1ac1750b828d7add79e7fc93bf3ab":{"kind":"REWORK","_number":5,"created":"2025-05-14 19:03:09.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/5"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 19:03:04.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/673abdd804e1ac1750b828d7add79e7fc93bf3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/673abdd804e1ac1750b828d7add79e7fc93bf3ab"}]},"branch":"refs/heads/master"},"891b67cc04779487a0ce800aa1ebd709102840e0":{"kind":"REWORK","_number":6,"created":"2025-05-14 19:31:16.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/6"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 19:31:10.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/891b67cc04779487a0ce800aa1ebd709102840e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/891b67cc04779487a0ce800aa1ebd709102840e0"}]},"branch":"refs/heads/master"},"f18f641f7b9b4eeebe88a5b0f3f5fef055edba5b":{"kind":"REWORK","_number":7,"created":"2025-05-14 20:21:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/7"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 20:21:38.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f18f641f7b9b4eeebe88a5b0f3f5fef055edba5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f18f641f7b9b4eeebe88a5b0f3f5fef055edba5b"}]},"branch":"refs/heads/master"},"ac84b1fed5d3857c77a0b522a0f47a57f019a456":{"kind":"REWORK","_number":8,"created":"2025-05-14 23:20:17.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/8"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 23:20:12.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac84b1fed5d3857c77a0b522a0f47a57f019a456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac84b1fed5d3857c77a0b522a0f47a57f019a456"}]},"branch":"refs/heads/master"},"948a631905b83b8fd4fd29a124b8874c4f042913":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-05-15 01:11:23.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/949695/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/949695/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/949695/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/949695/9"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-14 04:30:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 01:10:54.000000000","tz":-300},"subject":"[WIP] Mount requirements","message":"[WIP] Mount requirements\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: I62028c877736e823f749f0ee46c6bc60caac0dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/948a631905b83b8fd4fd29a124b8874c4f042913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/948a631905b83b8fd4fd29a124b8874c4f042913"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
