)]}'
{"id":"openstack%2Floci~949823","triplet_id":"openstack%2Floci~master~Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-05-23 20:06:36.000000000","reason":"Change was submitted"},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-05-23 20:06:36.000000000","reason":"Change was submitted"},"34520":{"account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"last_update":"2025-05-23 20:06:36.000000000","reason":"Change was submitted"},"14525":{"account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"last_update":"2025-05-23 17:27:46.000000000","reason":"\u003cGERRIT_ACCOUNT_14525\u003e replied on the change","reason_account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}}},"hashtags":[],"change_id":"Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df","subject":"Mount requirements image","status":"MERGED","created":"2025-05-15 02:13:28.000000000","updated":"2025-05-23 20:11:54.000000000","submitted":"2025-05-23 20:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"949823","meta_rev_id":"c5b32a9225d01ceb2e095189cfc2ca3b05ae1aab","_number":949823,"virtual_id_number":949823,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-23 20:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-05-23 18:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-23 17:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-05-23 18:52:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 02:35:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 17:12:05.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-05-16 17:17:13.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-05-19 05:31:52.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2025-05-23 18:52:01.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"c960c56a3298b10d78486cc32891137a62648793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 02:13:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0478410f373deed8f382b6598df36c2e627f079b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 02:35:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ca9f15f5a324b34a37bf406ebed47f5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4c229ae2df0049ed8779c9ad15481f65 : SUCCESS in 12m 12s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c3dd98240e264d2ea6b05c963e39b7bc : FAILURE in 3m 04s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/408cd033850c42b6bf821c0788d76fdc : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e499695d8264b39af0d4cce9612eb55 : FAILURE in 3m 03s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/841a89c26dec406480d2c3f9cd20d0e4 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d22d2073dac412abce4bab8f0249f1b : FAILURE in 3m 34s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a20df596ffbc4d9d9653008a90c30e94 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9f5f6369363545dea05b693b220cedf6 : FAILURE in 3m 27s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/942c02493a98439ea7c51ea36a38a6b1 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a74ac6afc4ee4a1f886f632427b7d971 : FAILURE in 4m 21s\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5a0a387ebc2347318ce0ec8ad7cf746a : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"016e5fd3f93f39c6cb4af99856b21a8a4386da0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 03:21:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f6a82d91f9e93c5c3cf94e1b94c26dabe9c0ee2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:29:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38d59576a39e493da88e11e087b8a351\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/613a379c3b6f430bad1c2579066bd629 : SUCCESS in 6m 34s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/02acdfdb622845b792ceff16587641dc : FAILURE in 3m 22s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6516a3b12bd04709a1901ecfadb2ac57 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy","accounts_in_message":[],"_revision_number":2},{"id":"a343f100e9a17667aef12f632300a107bca390ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 05:14:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c8811cc409244c057f6aa8d7a711060a345663fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 05:23:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6106ffb45ceb429885c9d9d0d8f3cf8c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4272281ee6ae4a1396bd412b19948317 : SUCCESS in 7m 51s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/62b5299cab7b48a5b37998143399472b : FAILURE in 4m 09s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7a5287f7bbb44b3495838bcd5f7da492 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy","accounts_in_message":[],"_revision_number":3},{"id":"e750cd85bdc953a20c7ebaee7dd47799d874e696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 05:32:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dd42a6c1daea22897b228430640bb34579ee11c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 05:42:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9ec573918044541b30c0781309763e7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4a00ff3477a7443ca890013366d88bb0 : SUCCESS in 10m 05s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/26657e85dcc94b12a687faf754698ccc : FAILURE in 7m 49s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3bef36dd5353432c9961fd00afbe0d15 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy","accounts_in_message":[],"_revision_number":4},{"id":"5a5c89cabf1dcf3cf463ea7b2d2674c86e149587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 05:54:03.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1636f006285ac855c0af4283c1f392afa511c765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 06:40:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e1165af8f654c73a7ec09f37639e20b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3b0a56c8b12449258e3ac5c17e58ebb6 : SUCCESS in 34m 45s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b0ac4affc5f747a7af31e8fc232d5909 : FAILURE in 28m 08s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9d21a08d671942d287b1f5d6593b94ba : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy","accounts_in_message":[],"_revision_number":5},{"id":"16949582b45ef9b4906b19ffa15834ff1b81f82c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 16:54:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3a980a31a2b4cfd0196a5bc1011641fab62be78d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 19:11:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5fb2f006f744629cb8a60a3fe7a084\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/fe0f8235c6ce4c7fbc5102a510eb6173 : SUCCESS in 2h 11m 54s (non-voting)\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/85a83e5329424619829beabb9ad9fcf4 : SUCCESS in 51m 19s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/424efc87a1ed4ce89642f895a2f94793 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":6},{"id":"7f447e99ae0d72516a1c3f76703a79d836b429e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-15 20:46:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"4b34a8805bab3399c041e8bef15db335889b78d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 23:40:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f59b8ec0bc94194b14ca49cddbf49c1\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/469d7a8208dc4491b89a3525b85ee333 : SUCCESS in 2h 53m 14s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/903fd56fa31647c899ef7343a1251c21 : SUCCESS in 51m 38s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c36c7d42c7a24c81a9d4a0c427c5a2b5 : SUCCESS in 1h 10m 23s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d2c1f4c4e5ae4928a059bfd53d24fa02 : SUCCESS in 42m 14s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5c9746f178d74bdab52a9dec81df3e0c : SUCCESS in 53m 12s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e06d734d98040af9329359f4c888bcf : SUCCESS in 45m 31s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1ad2f755ca7342a29a52d4d66dabdc7a : SUCCESS in 27m 09s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3a97e80cd4604aeaa0525b71451c02da : SUCCESS in 55m 42s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a62f54a73c3f429e8833b0dfa88b6cac : SUCCESS in 1h 15m 31s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/919daf71f5e94ab396aef1d143f812ef : SUCCESS in 30m 57s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1eecb8bb043747f9a0da04c937308fcf : SUCCESS in 54m 44s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8a5b56313b3e44e8a81eaa0a18ea8bca : SUCCESS in 1h 21m 07s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7e93741c43c9437db6a30d4a437dec40 : SUCCESS in 44m 59s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9be106265ebd40de97d490e29d014537 : FAILURE in 32m 30s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7eec7dd6bcc2457299e0ad1065d8cbea : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2069dbf02894633fb4c7e88b16791d4e56eecebc","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-05-16 17:12:05.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a31f16253fccf684daf7b48665e30e41d7070cbe","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-05-16 17:17:13.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b07d2012d5fcbde82d856e5e754fb6f0766a42d6","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-05-19 05:31:52.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3235bd4361ba1f59f5864da222e0df492b2cef2e","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 16:52:50.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"89cb4f4b1e7208aa9498b86cf2bcde66b832cec9","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-05-22 06:23:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5e47f4c5bc0ea4cc600f8f314ba479edd8970809","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-22 17:37:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2f473faaa11a60f54429aa61bbcfb66a96791f85","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-05-22 17:58:22.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"838032dd15ae736fbdcb8d7eed1769f3fa4daf29","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-05-22 17:58:34.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"09eb4cd493cc1f3bc4cc4d4d932087bb47d58d88","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-22 21:24:35.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"dd771e62c3279e6e7ca290e4c1d7f3496391e630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 21:24:46.000000000","message":"Patch Set 7: Verified-2\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  The key \"voting\" appears more than once; duplicate keys are not permitted.\n\n  name: loci-build-master-ubuntu_jammy\n      voting: false\n      timeout: 10800\n      voting: false\n      dependencies:\n        - name: loci-buildset-registry\n      pre-run:\n        - playbooks/inject-keys.yaml\n        - playbooks/pre-build.yaml\n  ...\n\n  in \"openstack/loci/zuul.d/master-ubuntu_jammy.yaml@master\", line 2","accounts_in_message":[],"_revision_number":7},{"id":"e00443cb700f14c2f9c22ef8754be28726a2eb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-23 03:13:53.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2aa17b884d491f799094f28dbbcacee5774231ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 03:14:04.000000000","message":"Patch Set 8: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  The key \"voting\" appears more than once; duplicate keys are not permitted.\n\n  name: loci-build-master-ubuntu_jammy\n      voting: false\n      timeout: 10800\n      voting: false\n      dependencies:\n        - name: loci-buildset-registry\n      pre-run:\n        - playbooks/inject-keys.yaml\n        - playbooks/pre-build.yaml\n  ...\n\n  in \"openstack/loci/zuul.d/master-ubuntu_jammy.yaml@master\", line 2","accounts_in_message":[],"_revision_number":8},{"id":"4ccc040d2cff3de815ac96fd0573e2fc8f89a97e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-23 03:28:18.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c8ac848c5ebaea303dde6e994b4cea7ee35bf4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 05:49:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/410a9909c1a1430e8f8502d987e9ccab\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/78fb2eedcb94449bb5e185c555b30a66 : SUCCESS in 2h 20m 49s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0e5d0416fcd44a9dafe70e44c8450b01 : SUCCESS in 46m 47s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/14573457e08e450bb7c88d1cc83dc392 : SUCCESS in 1h 11m 25s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d596c1660f104f5089f304308866a6e3 : FAILURE in 10m 18s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a745c673fda54f59bafeceb20f55dd90 : SUCCESS in 51m 05s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/77c495c4e5434a6a8d4fef39e0268e79 : SUCCESS in 1h 12m 56s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/322a64f81e7a44f088bc69da2922dd6d : SUCCESS in 38m 59s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f9d696a3d0bd4b6a8ebcf3a9fd615390 : SUCCESS in 49m 44s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/350d5396d4554310b161585b709d16eb : SUCCESS in 1h 14m 40s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fd4a9106148f471ab318fe95c8af57eb : SUCCESS in 39m 47s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/21a2f83cca944f9685a3c3b3c101f3a9 : SUCCESS in 50m 01s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f22125df802a4df7add358322f13b74e : SUCCESS in 1h 18m 22s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/764274fe98a64a6796575359c430b475 : SUCCESS in 40m 46s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1969db6b1a1a478abc2aef71e8dada39 : FAILURE in 31m 49s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/24a425203ab14fb5b94548a4a9bdeabb : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"21ceb376362e79cf52988e84d952cc318a79d0de","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-23 06:03:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6498d3b086cbb15a3b8e1ab6b3aef7af83606478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:44:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7eaff7d5cd4153b3d2a0e8b8e1e9fb\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/e855d5d7dbc546cdb6b0f5de66d188b3 : SUCCESS in 2h 39m 06s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3fadd9a8dd794f5592fce13ba4a611b4 : SUCCESS in 56m 55s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bfa0d12d3f6446d084c5bca95fc56fb8 : SUCCESS in 1h 13m 28s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0b26a4dc719845519bee99efd2ade834 : SUCCESS in 42m 40s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c02c451b58b6485c8444685351a063cc : SUCCESS in 53m 01s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4aa15bd0fa104c20a8c7e16bc95fbf4a : SUCCESS in 1h 16m 55s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/92c3dfcf8c2142c29c2447d12c53ba1d : SUCCESS in 42m 41s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f0d112e06bc44a228d588ccb3b616674 : SUCCESS in 57m 25s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/306072c47ea24bb184e7e87704a48f5b : SUCCESS in 1h 18m 11s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a02d81fd367c42668a2a08390da00ee6 : SUCCESS in 43m 55s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8e512568220340db84161492f075125b : SUCCESS in 57m 42s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3afc6fa02e58481a92af474654f00b3e : SUCCESS in 1h 23m 44s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9a01841f8f0b4d9f9d7c4ae982af1e34 : SUCCESS in 44m 26s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d70326b4b8d74d089870374fc26b30c7 : FAILURE in 33m 59s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3d339c762e324e8387bd686f241c3ead : FAILURE in 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"048f2ea56c7839332814b9938e4e4028272bb323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-23 14:12:11.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"042ac4df4961704663b43f958d52e7b3d7320eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 16:09:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee0d0c7ae4f4040811cf52652be3ed9\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/cf01e5c4f06b4d2f9a11ff8e2c8d7cb7 : SUCCESS in 1h 56m 33s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1eee64f6297949b28d0dda13c8135c3e : SUCCESS in 42m 13s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e255064cbba54ee882bb5610f6f5f0b9 : SUCCESS in 1h 01m 17s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/75219dfc66fc4b2ebfe111a0a776d19d : SUCCESS in 43m 36s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a26d03cff87e42ee91e871f820836625 : SUCCESS in 1h 06m 14s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/08be4f033ca44361a9d386142c5126da : SUCCESS in 43m 11s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/dbb6c905bb7f479daa94064a51533f0f : SUCCESS in 1h 04m 25s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a22f8bfec25f4cb588a821e6a75e89d9 : SUCCESS in 43m 16s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/297e7bb216014087b4d3306fe3f8f105 : SUCCESS in 1h 09m 32s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8140e8d342cd4e98853a1c0424349326 : FAILURE in 12m 29s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/da21d0e6255d4ba8aadac42684190338 : FAILURE in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d1f24f6606116c1d7798db31659b1e53852ca983","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-05-23 17:27:46.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"26554f56cecf4c824e8a4d9b37680a380fc649e3","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-05-23 18:52:01.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"76f397eed68d2cf2eb0b7256a37eec96e24a056e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 18:52:09.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"85d0b2fa83e436923651ed31d2010fcd2d2e8822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:06:36.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79cd69b6e3624cffa9cc4b96462d9b27\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/98627c82d4094d13ade9f5183aafb471 : SUCCESS in 1h 14m 23s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1cf44e5898744e2bb29f5424166254b0 : SUCCESS in 58m 55s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/22e41645f5a64da7bd8853c66bb4520c : SUCCESS in 1h 04m 07s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fe14d90838fe496885a81d6eb6fd1f6e : SUCCESS in 51m 00s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/867a809d22ee47c08bc565a75156d82b : SUCCESS in 1h 05m 55s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d722166b97e426da84bc9c15bd5948a : POST_FAILURE in 40m 24s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/39aaf1f603cf4e8da05d426439fbf402 : POST_FAILURE in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8a693de866d3597b7ecff4a93034cf11c7502ea5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:06:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"c5b32a9225d01ceb2e095189cfc2ca3b05ae1aab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:11:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e908f86c6ce42daa8a9f717d6c3aaea\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/203bf7ea7f9f45e0a7aba03ee2df9e32 : SUCCESS in 4m 25s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f7b29ad07a554338a0b5594bec908f44 : SUCCESS in 5m 09s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1d250ed183f84fa1bb916a80b691af3a : SUCCESS in 4m 12s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7a19f707588c4ac0b72bf2b255848791 : SUCCESS in 4m 04s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d1170c90c054d19b77367b37bd0cbdd : FAILURE in 4m 08s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/afa0c60296a04a4aa5fc6abb2a936a72 : FAILURE in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"dd303889906a20c43bcfd5f23a53791d56c38eb0","revisions":{"a2337f780b29279da2dc0f24aa014984ecebd006":{"kind":"REWORK","_number":1,"created":"2025-05-15 02:13:28.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/1"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:13:00.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2337f780b29279da2dc0f24aa014984ecebd006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a2337f780b29279da2dc0f24aa014984ecebd006"}]},"branch":"refs/heads/master"},"93f56c93038dc06abe63f96855a55242ed119635":{"kind":"REWORK","_number":2,"created":"2025-05-15 03:21:37.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/2"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 03:21:32.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/93f56c93038dc06abe63f96855a55242ed119635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/93f56c93038dc06abe63f96855a55242ed119635"}]},"branch":"refs/heads/master"},"be40d9762545c24b14bf61eac0da9ab07f8a9090":{"kind":"REWORK","_number":3,"created":"2025-05-15 05:14:37.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/3"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 05:14:31.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/be40d9762545c24b14bf61eac0da9ab07f8a9090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/be40d9762545c24b14bf61eac0da9ab07f8a9090"}]},"branch":"refs/heads/master"},"cc3253a219eaee4b0af1ed1a8aa6d005e2088ecf":{"kind":"REWORK","_number":4,"created":"2025-05-15 05:32:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/4"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 05:31:54.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cc3253a219eaee4b0af1ed1a8aa6d005e2088ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cc3253a219eaee4b0af1ed1a8aa6d005e2088ecf"}]},"branch":"refs/heads/master"},"7e19152b8ff4bbe9e5959a68dfd4a73d37d08f9b":{"kind":"REWORK","_number":5,"created":"2025-05-15 05:54:03.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/5"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 05:53:58.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7e19152b8ff4bbe9e5959a68dfd4a73d37d08f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7e19152b8ff4bbe9e5959a68dfd4a73d37d08f9b"}]},"branch":"refs/heads/master"},"24bbc0e3e432362e23469bf1f0896911583f660b":{"kind":"REWORK","_number":6,"created":"2025-05-15 16:54:50.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/6"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 16:54:33.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/24bbc0e3e432362e23469bf1f0896911583f660b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/24bbc0e3e432362e23469bf1f0896911583f660b"}]},"branch":"refs/heads/master"},"e13b7fcdd9744dc10e7672163e2202f8b22d55fb":{"kind":"REWORK","_number":7,"created":"2025-05-15 20:46:49.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/7"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 20:46:43.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e13b7fcdd9744dc10e7672163e2202f8b22d55fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e13b7fcdd9744dc10e7672163e2202f8b22d55fb"}]},"branch":"refs/heads/master"},"9a64c5a09b3bff0c1339037b42cb97023206da63":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-05-23 03:13:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/8"}}},"commit":{"parents":[{"commit":"26c6e749eb02e6cfe9328f5d1a79a1beab992c76","subject":"Remove pylxd workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26c6e749eb02e6cfe9328f5d1a79a1beab992c76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-23 03:13:53.000000000","tz":0},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9a64c5a09b3bff0c1339037b42cb97023206da63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9a64c5a09b3bff0c1339037b42cb97023206da63"}]},"branch":"refs/heads/master","description":"Rebase"},"4846d8ff891d2f2075983c8b50f2c11349e92ff5":{"kind":"REWORK","_number":9,"created":"2025-05-23 03:28:18.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/9"}}},"commit":{"parents":[{"commit":"26c6e749eb02e6cfe9328f5d1a79a1beab992c76","subject":"Remove pylxd workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26c6e749eb02e6cfe9328f5d1a79a1beab992c76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-23 03:28:10.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4846d8ff891d2f2075983c8b50f2c11349e92ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4846d8ff891d2f2075983c8b50f2c11349e92ff5"}]},"branch":"refs/heads/master"},"dd303889906a20c43bcfd5f23a53791d56c38eb0":{"kind":"REWORK","_number":10,"created":"2025-05-23 14:12:11.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/949823/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/949823/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/949823/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/949823/10"}}},"commit":{"parents":[{"commit":"26c6e749eb02e6cfe9328f5d1a79a1beab992c76","subject":"Remove pylxd workaround","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26c6e749eb02e6cfe9328f5d1a79a1beab992c76"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-15 02:09:44.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-23 14:12:01.000000000","tz":-300},"subject":"Mount requirements image","message":"Mount requirements image\n\nDocker build allows you to mount images during\nbuild to get access to their contents.\n\nLet\u0027s use this functionality and mount requirements\nimage instead of fetching wheels directly from a registry\nor via http.\n\nDepends-On: Id2b78dba5c99f38599345f16707683f3bde533e4\nChange-Id: Ia9eaf8598d2ef05f2bdf54f51c42d1ec4e4383df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dd303889906a20c43bcfd5f23a53791d56c38eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dd303889906a20c43bcfd5f23a53791d56c38eb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
