)]}'
{"id":"openstack%2Floci~950344","triplet_id":"openstack%2Floci~master~I855143382bc16c211a1ef40cfa2ef1cb501f5901","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-05-21 21:30:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I855143382bc16c211a1ef40cfa2ef1cb501f5901","subject":"Get rid of libnss patch","status":"MERGED","created":"2025-05-19 17:52:32.000000000","updated":"2025-05-21 21:35:04.000000000","submitted":"2025-05-21 21:30:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950344","meta_rev_id":"c465f0355b52d1190117da6d37f6c5fbee76fdc1","_number":950344,"virtual_id_number":950344,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-21 21:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-05-21 20:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-05-21 17:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-05-21 20:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 21:28:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 17:05:19.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-05-21 20:24:11.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"3ff5961b4e5406e10c50526411281402c93194ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 17:52:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d810bd319a422a52dcbb739ce15e96766a38e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 19:00:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ce4a01745a9788341416aa181855630ed8d2028a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 21:28:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6673f96a2f694dbbb77d8a4798120869\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/efd0fb7eea13473f843a10accb55830c : SUCCESS in 2h 15m 52s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/335343805da84849a917835cd41aaeb0 : SUCCESS in 42m 25s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2bb8438df1a54b19825a8b7d9e260447 : SUCCESS in 1h 18m 12s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/06340c3baa53482e8891c0f72ba52aa9 : SUCCESS in 50m 23s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ce77a7aac1eb4455a6a465d41a985319 : SUCCESS in 1h 12m 48s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/30f096a5890d4740813d4a391177d86f : SUCCESS in 50m 41s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5fa65cdef5764af09dcc729d192d4120 : SUCCESS in 1h 14m 07s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d7190e673b8a482eb836952c8fb3ecba : SUCCESS in 55m 19s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/657b675d87f648bfbed45bc2b804e061 : FAILURE in 12m 37s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/242d1d5211d74ac2973c21190492899a : FAILURE in 37m 46s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/22b447879973481a83fb96fc953df460 : FAILURE in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e67d1109f748430c6091a7f0ae3e6f596d01081c","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 22:23:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95d491065414d1e9778044fd26ecd1d5f321835d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:58:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca61a2ff01e64f39847d3f883127083b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/28a9e6b54e95443ea5bd949f098557c1 : SUCCESS in 2h 32m 48s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1a194e12e9fb496cb58d1bf24ec7d759 : SUCCESS in 55m 50s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/403c53cae07f40228dd02e414a00c73c : SUCCESS in 1h 09m 55s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/da30b83b3ba14da5b82752e527f0a125 : SUCCESS in 53m 10s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4c693105135c40f38b85b4f453cb4cfc : SUCCESS in 1h 14m 27s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/301c2d50d5364bf79e4a536a4de5859f : SUCCESS in 56m 00s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/721b9f3efc4f40b1a80dc3594efd80b2 : SUCCESS in 1h 14m 10s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/181d28f67ff04fd9ae8ec2ce34c3b62d : SUCCESS in 54m 28s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/20cecb2d9741496fa373667a3026ba3b : SUCCESS in 1h 18m 08s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/392a508789f84e7f8839e485817cf493 : FAILURE in 37m 10s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/227021a09a9149a1afd1e820e7bd4de1 : FAILURE in 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8d63e4c5960b56068935756a52c6bc7818d31f4","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-05-21 17:05:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"291933709b37961537270abaf2b09093e89f6e87","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-05-21 20:24:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0474b3cfd4b48491e593ef25e8b48b64722c5d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:24:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b566e2c309f0de3f5649ce027ea7c1330c9d893c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:30:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2078a6ab245342db878158035b888ca2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d4458d3813df4130baee3d377f769e4c : SUCCESS in 1h 05m 40s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c5005da52aa84cfea0c12c46095f8319 : SUCCESS in 1h 02m 47s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0b273de2696e475fa9a7569baf3a2d10 : SUCCESS in 58m 59s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/aa57f3b6122844919c900d0d976fed9f : SUCCESS in 1h 02m 32s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/05f3a6e628f94c0691ab7dc9136186be : SUCCESS in 58m 28s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/55ab508d6a984d60b425c82747dbbaa6 : POST_FAILURE in 41m 11s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ad9593fd57ed424289624f864db245e6 : POST_FAILURE in 40m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf32ac1fa4cdc2cdf3531a7e08288c0a4615c507","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:30:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c465f0355b52d1190117da6d37f6c5fbee76fdc1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:35:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a150f2dc05414dbe95acfafe0ee4d3\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7797d8ececc645aaa90359ed92bd5cff : SUCCESS in 4m 18s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/44a5e0d2626c409cbd9305540e835fc1 : SUCCESS in 4m 30s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7e193f77657445a79e0ea19aea1154a9 : SUCCESS in 4m 22s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/42904aaf374e4d5aa91472e012fc41b9 : SUCCESS in 4m 27s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2c5682bb691a4314862aec2bdbbd6297 : FAILURE in 4m 11s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1a2b2b42f0e14e548a79d86ab5cb3495 : FAILURE in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3de072728bb1efd7737f6942fb5f0847defdf64a","revisions":{"94ae67d4105037f604c656f6371f98c6cfc15e38":{"kind":"REWORK","_number":1,"created":"2025-05-19 17:52:32.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/44/950344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/44/950344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/44/950344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/44/950344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/44/950344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/44/950344/1"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 17:50:05.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 17:50:05.000000000","tz":-300},"subject":"Get rid of libnss patch","message":"Get rid of libnss patch\n\nIt was necessary for older releases where\npython-nss\u003d\u003d\u003d1.0.1 was in the list of upper\nconstraints but was incompatible with libnss3-dev\npackage in Ubuntu Focal and Jammy.\n\nChange-Id: I855143382bc16c211a1ef40cfa2ef1cb501f5901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94ae67d4105037f604c656f6371f98c6cfc15e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94ae67d4105037f604c656f6371f98c6cfc15e38"}]},"branch":"refs/heads/master"},"3de072728bb1efd7737f6942fb5f0847defdf64a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-19 19:00:04.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/44/950344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/44/950344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/44/950344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/44/950344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/44/950344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/44/950344/2"}}},"commit":{"parents":[{"commit":"96e9f7cad00089ed66f744bcc6a8db4e40c67342","subject":"Make master branch build jobs non voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/96e9f7cad00089ed66f744bcc6a8db4e40c67342"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 17:50:05.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 19:00:04.000000000","tz":0},"subject":"Get rid of libnss patch","message":"Get rid of libnss patch\n\nIt was necessary for older releases where\npython-nss\u003d\u003d\u003d1.0.1 was in the list of upper\nconstraints but was incompatible with libnss3-dev\npackage in Ubuntu Focal and Jammy.\n\nChange-Id: I855143382bc16c211a1ef40cfa2ef1cb501f5901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3de072728bb1efd7737f6942fb5f0847defdf64a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3de072728bb1efd7737f6942fb5f0847defdf64a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
