)]}'
{"id":"openstack%2Floci~950351","triplet_id":"openstack%2Floci~master~I26e9758a5a8d3301f2e3060b49191f689fec9aab","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-05-21 21:33:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26e9758a5a8d3301f2e3060b49191f689fec9aab","subject":"Remove python-qpid-proton workarounds","status":"MERGED","created":"2025-05-19 18:40:10.000000000","updated":"2025-05-21 21:39:43.000000000","submitted":"2025-05-21 21:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950351","meta_rev_id":"ccd1504deb5ff7d3f5685bfb8e90d48398e978f0","_number":950351,"virtual_id_number":950351,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-21 21:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-05-21 20:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-05-21 17:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-05-21 20:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 21:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 17:05:23.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-05-21 20:24:35.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"dfadc63797b5a67a0069678faa2667420000a087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 18:40:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe922fe88e18600dc079134bd72c76863a1b4d11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 19:00:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b8f81166dffe9519d085bf9a714344c7c73beb51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 21:52:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1865d1d42d5940c3a91f59e560bd4a8d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/51cc7c8a8cb744068afe3f1c0151184f : SUCCESS in 2h 31m 16s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b5a3f0e1e6794607aa793dba0398988d : SUCCESS in 52m 28s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e562f89f73bc4ed8990f97ba48bb7216 : SUCCESS in 1h 20m 29s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/19de0566852745358b2d092e6d3e7a68 : SUCCESS in 51m 30s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9b1e056a1ccf4335ba948a89cd689586 : SUCCESS in 1h 18m 26s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/11e5363ae63f4da49e64a264b5d00ac7 : SUCCESS in 58m 24s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/01fd9a911d69418b873e6f589a0a7c20 : SUCCESS in 1h 12m 58s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b6b18c8aef724af1a4f6938aed157e9c : SUCCESS in 58m 28s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/defd75695a714c64958ea53315736f2e : SUCCESS in 1h 17m 07s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9d45f4beecf94c059c5b1020f9ad671a : FAILURE in 35m 51s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/939ba0b7bf3d48bcb51a5bc652f868a8 : FAILURE in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"080df1f45a775b6be18b83ea0e1000dbed71b900","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-05-21 17:05:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f081941931210520ac752231c55c026a7024931","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-05-21 20:24:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c3a60184bd718c61cca5977eb8dd3758f56e7766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:24:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e5f9eeb3b40d6a2923958426f747e0d149f09d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:33:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b50fc01ec8374187ad2124dc6b5cb27b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/14962007b6834bc8b9fe4acb8d3da309 : SUCCESS in 1h 08m 37s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c522ba49a88b47d485151b0b8602ca5e : SUCCESS in 1h 03m 13s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fc5bdedda79645b0a176820629c73042 : SUCCESS in 1h 02m 41s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b69b1a7c65ca42f28ece615325a312d5 : SUCCESS in 59m 53s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/327bab732bd249fa851974de1e039b16 : SUCCESS in 1h 05m 10s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e933bb8190c4a8697af67477407d155 : POST_FAILURE in 45m 08s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c1159bc3bce840c2800eef86af586a06 : POST_FAILURE in 39m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b201f68e5698735f61134f20fc0ebffd99323a51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:33:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ccd1504deb5ff7d3f5685bfb8e90d48398e978f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:39:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f3b05fef744fca9a5ddbeb5172744e\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bbda213bd1924368988d4367718fb26e : SUCCESS in 4m 29s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eb435d17b0a54340ba680e42814acd27 : SUCCESS in 4m 14s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eb8b0fb5c37f414ab1babe88f7feb82a : SUCCESS in 4m 28s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/36d5542c9d754accbff34bc97a3dd9a8 : SUCCESS in 4m 19s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ff5ad369365e41f4af65ab824ea61918 : FAILURE in 4m 07s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e857709df7bf48b6aa4d4fb8ce4948ff : FAILURE in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8831191b86b7dba9d2c688caa30e685b578e930c","revisions":{"48dd5888dc796066e3d9817d47ff3e8426aa8fef":{"kind":"REWORK","_number":1,"created":"2025-05-19 18:40:10.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/51/950351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/950351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/950351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/950351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/950351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/950351/1"}}},"commit":{"parents":[{"commit":"94ae67d4105037f604c656f6371f98c6cfc15e38","subject":"Get rid of libnss patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94ae67d4105037f604c656f6371f98c6cfc15e38"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:36:51.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:36:51.000000000","tz":-300},"subject":"Remove python-qpid-proton workarounds","message":"Remove python-qpid-proton workarounds\n\nThey are not actual any more. Originally\nthey were added to address compatibility issues\nwith openssl 1.1.0\n\nSee for details https://issues.apache.org/jira/browse/PROTON-1381\n\nChange-Id: I26e9758a5a8d3301f2e3060b49191f689fec9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/48dd5888dc796066e3d9817d47ff3e8426aa8fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/48dd5888dc796066e3d9817d47ff3e8426aa8fef"}]},"branch":"refs/heads/master"},"8831191b86b7dba9d2c688caa30e685b578e930c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-19 19:00:22.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/51/950351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/51/950351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/950351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/950351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/51/950351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/51/950351/2"}}},"commit":{"parents":[{"commit":"3de072728bb1efd7737f6942fb5f0847defdf64a","subject":"Get rid of libnss patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3de072728bb1efd7737f6942fb5f0847defdf64a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:36:51.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 19:00:22.000000000","tz":0},"subject":"Remove python-qpid-proton workarounds","message":"Remove python-qpid-proton workarounds\n\nThey are not actual any more. Originally\nthey were added to address compatibility issues\nwith openssl 1.1.0\n\nSee for details https://issues.apache.org/jira/browse/PROTON-1381\n\nChange-Id: I26e9758a5a8d3301f2e3060b49191f689fec9aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8831191b86b7dba9d2c688caa30e685b578e930c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8831191b86b7dba9d2c688caa30e685b578e930c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
