)]}'
{"id":"openstack%2Floci~950353","triplet_id":"openstack%2Floci~master~I00e37751021ad79c3d5fe401dd5b4292af427542","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-05-21 21:44:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00e37751021ad79c3d5fe401dd5b4292af427542","subject":"Remove futures workaround","status":"MERGED","created":"2025-05-19 18:54:14.000000000","updated":"2025-05-21 21:48:50.000000000","submitted":"2025-05-21 21:44:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950353","meta_rev_id":"8a19503b5624db6a6615bd06b4f07100dfcdc38a","_number":950353,"virtual_id_number":950353,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-21 21:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-05-21 20:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-05-21 17:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-05-21 20:24:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 21:27:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 17:05:28.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-05-21 20:24:53.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"12a5a324f986366eaa2d19df6e0c0d46ed3088bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 18:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf23f64f86e16d9225cb00f20728b3862a76f0c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 19:00:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1f1f41c8d998794d65fc04246a516065832bf3ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 21:27:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ffe2ef1334448bbeb2d064ab796dc5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/5f285121241b4504ac21639aa5bd4ca9 : SUCCESS in 2h 26m 01s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/57817b1511724e47a41e4528d103016f : SUCCESS in 55m 09s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/16df049de7a945218a0d57dd8673e7f0 : SUCCESS in 1h 10m 16s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1ee7d8afe72642ee93b4ad864829cb26 : SUCCESS in 53m 04s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2aae75a14a064d2eb090315e60b5bfa1 : SUCCESS in 1h 13m 45s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/428b87801ab54157a5e4c6e9199470b8 : SUCCESS in 57m 13s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f240c894ef7c4f00ab2769a374d5185b : SUCCESS in 1h 11m 10s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eabd3745a1a048f9a1a11fe90c3d63f6 : SUCCESS in 56m 35s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4ad6ecc2846c49efa035fc34636619d1 : SUCCESS in 1h 16m 42s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90a5f81332c24dff92a0ad02036cabac : FAILURE in 34m 47s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba10562eaeec4fa797837b1156e700b3 : FAILURE in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e293439076a1685516ac430fd1beba174a23d8d","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-05-21 17:05:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"70abacb62b1b90e6f2787474cbea7bd6fb7edf84","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-05-21 20:24:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e13f83d34d7db1f2bd241ed0c5d64e5dd237400c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:25:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"54cd1de336f9b98625a2b96d03a9124f2fb7e724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:44:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b0be525fca4590a2a4c06f75d65a13\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/956572f00b014cc0b3d5bf955da96d83 : SUCCESS in 1h 19m 14s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b75a7f3360ae46ceaf0fe08323130630 : SUCCESS in 1h 11m 58s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b9e118903e134dd39ec9333e3d4f2c08 : SUCCESS in 1h 10m 52s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6c47308b29994f9ea5ed3e340b5e2230 : SUCCESS in 1h 13m 34s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/89e9e557e4a14c079df67e5d9f60454d : SUCCESS in 1h 11m 38s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5d1563a9a3534e74aa1df7a8f3fda328 : POST_FAILURE in 48m 24s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/450d4a6f55cc4ebf826be45cbcd67667 : POST_FAILURE in 48m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0b3c79618cf0d07ba9069bd3f4fdd6c598fc985","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:44:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a19503b5624db6a6615bd06b4f07100dfcdc38a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 21:48:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/761b3ad6b8f141bc8a142ec2f798007d\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/def104f4ab134217acfe641c7e0fa138 : SUCCESS in 4m 16s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/91a5fd8aec3d4a36ae0a587aab171e4c : SUCCESS in 4m 10s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/de33ade0bac44a46a2101a4cc5f75992 : SUCCESS in 3m 58s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/093c2117dc344bc493b250771936e615 : SUCCESS in 4m 09s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e25cb9e9976641a198735cdd8f962c1d : FAILURE in 4m 21s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/95d000d8412c4fc5a61b7da533da9975 : FAILURE in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97928720d3fe6def97be3bcc204dc1028da7fa16","revisions":{"131143f28bef75977a7c0c6d6e36750d035e761a":{"kind":"REWORK","_number":1,"created":"2025-05-19 18:54:14.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/950353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/950353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/950353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/950353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/950353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/950353/1"}}},"commit":{"parents":[{"commit":"48dd5888dc796066e3d9817d47ff3e8426aa8fef","subject":"Remove python-qpid-proton workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/48dd5888dc796066e3d9817d47ff3e8426aa8fef"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:53:06.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:53:06.000000000","tz":-300},"subject":"Remove futures workaround","message":"Remove futures workaround\n\nThe futures python package is not in the upper\nconstrains any more. The workaround is not needed.\n\nChange-Id: I00e37751021ad79c3d5fe401dd5b4292af427542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/131143f28bef75977a7c0c6d6e36750d035e761a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/131143f28bef75977a7c0c6d6e36750d035e761a"}]},"branch":"refs/heads/master"},"97928720d3fe6def97be3bcc204dc1028da7fa16":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-19 19:00:27.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/950353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/950353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/950353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/950353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/950353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/950353/2"}}},"commit":{"parents":[{"commit":"8831191b86b7dba9d2c688caa30e685b578e930c","subject":"Remove python-qpid-proton workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8831191b86b7dba9d2c688caa30e685b578e930c"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:53:06.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 19:00:27.000000000","tz":0},"subject":"Remove futures workaround","message":"Remove futures workaround\n\nThe futures python package is not in the upper\nconstrains any more. The workaround is not needed.\n\nChange-Id: I00e37751021ad79c3d5fe401dd5b4292af427542\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97928720d3fe6def97be3bcc204dc1028da7fa16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97928720d3fe6def97be3bcc204dc1028da7fa16"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
