)]}'
{"id":"openstack%2Floci~950354","triplet_id":"openstack%2Floci~master~Ia51fb96358bf833af18a91814ecdc2ee7ec84642","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-05-21 20:00:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia51fb96358bf833af18a91814ecdc2ee7ec84642","subject":"Make master branch build jobs non voting","status":"MERGED","created":"2025-05-19 18:59:26.000000000","updated":"2025-05-21 20:05:39.000000000","submitted":"2025-05-21 20:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950354","meta_rev_id":"c1e80fbbce304feed0a9c41cba95c2792a97144f","_number":950354,"virtual_id_number":950354,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-21 20:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-05-21 18:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-05-21 17:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-05-21 18:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 21:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-21 17:05:09.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-05-21 18:42:12.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"efab8d687b6213450ac1bf3516c5531d8d92aa32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-19 18:59:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3da0d056530957ef7b3c87ebb4f72a135f2cd60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 21:34:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/112c2f7b81674ed68b8481b670372e1f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2f9a3d862c5e4428a629d01fc9198c0e : SUCCESS in 2h 34m 39s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/62931e3ea921401494dc92ebed8809c5 : SUCCESS in 49m 08s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ed35f86aa3ce4438aff2a046edae3780 : SUCCESS in 1h 03m 05s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/039878b952b74c829d62e2d9e731d93a : SUCCESS in 39m 49s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/823310bee59746e887bcd6e0483e1190 : SUCCESS in 58m 16s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/64a60f826f5f4646a1e2dd43a8e4ca7e : SUCCESS in 1h 03m 10s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7402bbd1da564b55a4b1223a79b586b0 : SUCCESS in 36m 21s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/63817d3a5e42495887ef56e4e06fcbb2 : SUCCESS in 52m 31s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2f700cd6cc8645b0878de562b11b6190 : SUCCESS in 1h 15m 24s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9fd9f34b870f4d17a6ab81a96eb50156 : SUCCESS in 37m 04s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5b1664403ba842818ca117a470650f1e : SUCCESS in 59m 39s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/adf859efe9b043868cc7e2d7b3e14622 : SUCCESS in 1h 18m 06s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ab2a1a9faa6b47ba8fb51910b5613a96 : SUCCESS in 43m 05s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a97ff084f54f410b9a3992e0a4bb9642 : FAILURE in 35m 27s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/08add5581ec6428db610804cdf23510f : FAILURE in 29m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1166c97dbf9dab18c3fa792dc5fd613dd9148644","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-05-21 17:05:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"825805b3b33edaa03ff0dff60273edcc4fbf1fc4","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-05-21 18:42:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e4229fcca22a0d77266bd6a55deab21cf818cad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 18:42:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6809e7f08ae20e32dc9606e4a8304949d307336b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:00:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c3ed7d455c467e90259b4359704dc1\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/71c56ffb6dc24b128e4e46ae09903b75 : SUCCESS in 1h 13m 26s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4134adb9c4144ea0a12474c127239baa : SUCCESS in 1h 04m 10s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/98c07588d32e43c38950f9847e3ed4e3 : SUCCESS in 59m 20s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8f4c055f00da412c9a8b9db6d718b836 : SUCCESS in 1h 03m 07s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7ed0221190a34374bb030f444ba4b284 : SUCCESS in 1h 03m 47s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5190844d03a64b42a09325bb75f9e9cd : POST_FAILURE in 42m 04s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6156b2e9053b4c21987c42bf368b4e49 : POST_FAILURE in 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56b6c66a97a3d7af75eaa6ca777e277c61c82bf3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1e80fbbce304feed0a9c41cba95c2792a97144f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 20:05:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edca92192af44e2dbfea13dd629626c6\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5c2d50835159403fa7b51f2bb294105c : SUCCESS in 4m 23s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0e9c7c2ae3184e1c82a48910a2bacddc : SUCCESS in 4m 46s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b09ddb20531c4daf869f3f43d615afab : SUCCESS in 4m 31s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/15a3a163e04c4a36b18240e99da05d30 : SUCCESS in 4m 15s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d8e479c94fef4514975431084548b75f : FAILURE in 4m 47s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/55abcf2585e74b6f99ae17653651917b : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96e9f7cad00089ed66f744bcc6a8db4e40c67342","revisions":{"96e9f7cad00089ed66f744bcc6a8db4e40c67342":{"kind":"REWORK","_number":1,"created":"2025-05-19 18:59:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/54/950354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/950354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/950354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/950354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/950354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/950354/1"}}},"commit":{"parents":[{"commit":"db285247c2da3149d8139ab6047ba91b5d2b1fa4","subject":"Align with the release cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/db285247c2da3149d8139ab6047ba91b5d2b1fa4"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:56:35.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-19 18:56:35.000000000","tz":-300},"subject":"Make master branch build jobs non voting","message":"Make master branch build jobs non voting\n\nBuild jobs for the master branch can be\nbroken in the mid of release cycle which\nis normal. This must not block Loci development.\n\nChange-Id: Ia51fb96358bf833af18a91814ecdc2ee7ec84642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/96e9f7cad00089ed66f744bcc6a8db4e40c67342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/96e9f7cad00089ed66f744bcc6a8db4e40c67342"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
