)]}'
{"id":"openstack%2Floci~951196","triplet_id":"openstack%2Floci~master~Id1c6bb5c9c486c62f28feb7ba429f7af84e88db0","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-06-03 16:25:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1c6bb5c9c486c62f28feb7ba429f7af84e88db0","subject":"Remove Centos related code","status":"MERGED","created":"2025-05-28 19:21:56.000000000","updated":"2025-06-03 16:38:24.000000000","submitted":"2025-06-03 16:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951196","meta_rev_id":"9be62d380d207a419045dec6e04f443c3d90c950","_number":951196,"virtual_id_number":951196,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 16:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-06-03 14:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-06-03 13:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-06-03 14:39:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 21:18:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 13:29:30.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-06-03 14:39:33.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"898bde6142e29025062d4a99cb6d82cdca856535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-28 19:21:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c84ec4586aac04e4b02e2174b1f7308f350dac73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 21:18:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56eef4adb7a2489bb5b4619e461656f3\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/8b47ff4f98b6441f80388cd23f855208 : SUCCESS in 1h 18m 23s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f716f7fffdfb43adb11e5a1f0c80b53f : SUCCESS in 58m 59s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3f3b3d135a114a02bfc50f51d4ec6561 : FAILURE in 5m 54s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9355cefc23fe450a9fd843e75bed3577 : SUCCESS in 57m 06s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c1a582eb985b42a4b7848da1e3cad870 : FAILURE in 7m 26s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/efe13c4e25b34fbbbf203b7643425bca : SUCCESS in 58m 50s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2bdf3374c7a44ce7968b4b797d206c68 : FAILURE in 6m 57s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8c5ab32cd65542d2abc0591b691fb0e9 : SUCCESS in 54m 40s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2e339be8aceb4966a465118b03179661 : FAILURE in 8m 49s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a520afee81d54cd5abe37ae590eba76e : FAILURE in 34m 40s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a0382274ee0d448ebb3602e3fa64dad4 : FAILURE in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fb26dcf0bce6ee9be31776c4d64ac319ab01907","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-28 22:18:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"071d0f55e52ece213ba18f16efba32d7b94469cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 01:18:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/959d907a554e4bf689eb72d44b74a5cc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c3695d40aa6f409880d929929156d8dc : SUCCESS in 2h 59m 23s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/12ec9ed6c3284450ab573160db4755e1 : SUCCESS in 1h 35m 30s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eebd4397bdc1444583abaa614f7161bc : SUCCESS in 1h 11m 16s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/51c2fd6818494c179c36c684b7218008 : FAILURE in 23m 39s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/697c2706e90e42e89f37dd59944d9b43 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3185b4e9e2134e44888e485a7ecabc7e : FAILURE in 9m 37s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/33f460cdbbeb48619fc3099b25420b6b : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/132ac97796c84eae90131c1ae78db533 : FAILURE in 23m 41s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d6741b70bef248f584321641209b9ad7 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a3dd6a4ae4b94a6e9f3743becf89c7df : FAILURE in 25m 54s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6646dd937fe044758a85daf8a68dcdfb : FAILURE in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"993ca434c42cb10a56a712e754e2946a4c61c696","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-29 04:57:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99cc193f672c2abae922645dc5758f5da0e282d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 05:39:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd3cf216ca444528bdeb238bc288f58d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/378c33569dae4b8aa96ce9ab89739972 : SUCCESS in 39m 38s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0759f93b769643cb848356ac68861dfd : FAILURE in 7m 18s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/706fd12cbc594e58bbba3d075dd2dbf8 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/93ce45d1714f43b7a9774c85d129e42a : FAILURE in 10m 32s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6dddf9f1419641579c02ed4d802554ef : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5ac3dae245514a02b25cf51c91a69962 : FAILURE in 13m 46s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/945a470acc5a484ab0576f4549da998b : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6e9d9c2aa19e4d43b849b2205f8ab41d : FAILURE in 35m 24s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5c9b1d034c204c7cb6135eb873017d5a : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e78e3f81670e41119fda300c0022a3cb : FAILURE in 10m 17s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5bd503bfc91a4bd287ccab29a2354438 : FAILURE in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9de6e4fdf253018ce11dde53bdf4ea631696878","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-29 15:23:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"703542be2a9a68b1d46405ff6d72e90af29318f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-29 17:28:20.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f8899848d62a5803e581b5d87086cc2628fb981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 20:58:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64b31915cb0642efba528b6ff666b522\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/19da83cb21e14b7b9990ebf871b3b288 : SUCCESS in 3h 02m 56s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7f8118e9b9b64501bfe1737029e131f8 : FAILURE in 33m 35s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fe10739a005546c68720ae5f25d8e1f6 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d173a31367b44d2ca651c46ed1c7685d : FAILURE in 31m 41s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7c5b52fbca10440eb1a3475d01492b16 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/20ebc96b8bb9456e82d52083c124aac3 : SUCCESS in 55m 01s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/312dfb9c39854b74bb99b8583b20aa41 : SUCCESS in 1h 06m 50s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2e71d72851834af2b8dfd015f3172a8f : SUCCESS in 46m 40s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4c45ddf7a1bd4e62865e7692e9e3b671 : SUCCESS in 1h 07m 47s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5260536393444080842d04af8b076c26 : FAILURE in 33m 59s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b7cf0ec3de13443880d36bf05ae6e10c : FAILURE in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50aeecb7c5fa1b94131b74281aed02ffd7e3523a","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 02:10:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34973c6eaf6c12cfafefd78b6dd92ee61f44582e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 04:44:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a627dde33a774cc185abb33fe6daf2bf\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4599235afb88435099f8f0051847320b : SUCCESS in 1h 58m 02s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bc782d1dbede409db0d9d85186194ae1 : FAILURE in 25m 35s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/862fbe409b5146cda2e0546529f527f3 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d321435fc0874887841fd13d21a3655e : FAILURE in 32m 52s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e93599be82634f259a39bf3bfb6b65df : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a3cb389e736e42d6bf5c93780a86e9b3 : SUCCESS in 50m 02s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f6bec3a19c9149799049d7e464527925 : SUCCESS in 1h 01m 52s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5e979caf2f054059b4526703a4379975 : SUCCESS in 46m 51s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1eea328454bc4e6daa231f5ce0e94773 : SUCCESS in 1h 04m 18s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4581699a2cd240c8a68170fbf163b592 : FAILURE in 26m 02s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7fbe03987dbf4bd6a587720c39fe6280 : FAILURE in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c6dc470240c4eed20391a87f3a76cbfd06d2354","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-31 02:16:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"658fb01a907b553a729f6c7d41032986dfb45d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 02:44:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63a7ef47a74a47d5b75605e37ecc2ca2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/53567912ce3243eaa50f67dfcfbc4ad4 : SUCCESS in 20m 03s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/687b3eb678454c15a1a60d01833df5c1 : FAILURE in 2m 55s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5b3ed7e215524fa69379ca11e9fd0fed : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c5c8cd20eff345c6a8e33e718ec1b08d : FAILURE in 3m 09s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f9a096f278c148c9b55e7b74bde531db : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e6642d89b2fe45788302b50a6ef54f16 : FAILURE in 3m 17s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2eacda8a79c547f88a3936c039282c20 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/21265d66b13142ceadb17ad65d0e30e2 : FAILURE in 2m 57s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/60873e4a530a4671a183dc5cbe23b4b6 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ed79184a99214a5e90b6288af4713b75 : FAILURE in 3m 04s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ce0daad1634449eeaa5a99259ad42cbb : FAILURE in 3m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"808cb392359eb80cdc6fbfd201bb40fe9fe3b71d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-31 04:00:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"300c7fbef31f45ec8315927f3b5cbf36543a9131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 05:54:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22666816fbe647debda47ac532c80223\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/e1024a58d8254efba122425ad0674b31 : SUCCESS in 1h 53m 00s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cf89dbb695684152b4bb4b2063e26ae5 : SUCCESS in 41m 03s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7f83d1fc4f9c444fafe7d22c6d481731 : SUCCESS in 1h 00m 54s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/718c8be2b99a44b49162b8095cf37816 : SUCCESS in 36m 57s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4f30a5d7851840abb6b84ab297f99d6a : SUCCESS in 1h 02m 05s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/422f6bf3b6374194a671a351476490aa : SUCCESS in 43m 28s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ce7febd4aaaf4ddd8e18fc35832015ec : SUCCESS in 1h 04m 00s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1531f658539240a9937321a701b98c0a : SUCCESS in 38m 55s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cbff26a527b447258412f9763710ff98 : SUCCESS in 1h 06m 22s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/56c8fb7a399b4b4ab5ec673a37c00d5d : FAILURE in 29m 58s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9e9a8a6d5c354b3e961679243d78470a : FAILURE in 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e87fa9a6e929b5acaaeafaa806a030e50509d4aa","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-06-03 13:29:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"763b4f97daa697c1fd38c4716a32a5a27a603666","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-06-03 14:39:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"72e87da203aaadb09e80b96c06f944c16207d2bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:39:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6244e10907a565a23d3a4f804bec143ae4cd0d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:25:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ec4545863c840f0a647dca2568837db\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b56bf2179b0b47039139791baa9e8223 : SUCCESS in 1h 04m 24s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/15f2ac926d6a4b1884e99914751c9fa6 : SUCCESS in 54m 24s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b4c2e50a23d34377a5373d13ab82bf60 : SUCCESS in 49m 50s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e2bf642b8d0441f2be1eefee84d1f1ec : SUCCESS in 54m 07s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a0f619c9e3f84f7f8c12e497a79f71b7 : SUCCESS in 59m 15s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b9a21d68da1f4cfca88f8d769dcd11aa : POST_FAILURE in 35m 42s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/57a14d60d66a44639bb83c3c453fb26e : POST_FAILURE in 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e136a3ddafd04714a8a9f655c048e41023fa0e3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:25:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9be62d380d207a419045dec6e04f443c3d90c950","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:38:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63e68e1b929b478a956baabc10014476\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1ca372b65e0c4aaa8cf51f4784dc7943 : SUCCESS in 6m 38s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cd192317c79d4b1dbd40e886ab84dfec : SUCCESS in 5m 05s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89c1cbdb47f4481c8d1a0220179301d8 : SUCCESS in 4m 41s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b1ef68886abd45c49f896be301994ce8 : SUCCESS in 5m 06s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c2f481878534490fafc7d9c520ceee33 : FAILURE in 4m 23s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9ec8c849ed0f415fab29b2eb3b05487d : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"304d131aa637cf37f320b6c3b32a9dbb643adc6f","revisions":{"7f13830d70d7732c44033d760bf8ae2aa962cceb":{"kind":"REWORK","_number":1,"created":"2025-05-28 19:21:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/96/951196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/951196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/951196/1"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-28 19:15:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-28 19:15:04.000000000","tz":-300},"subject":"Remove Centos related code","message":"Remove Centos related code\n\nWe have not been testing this code for quite a some time and\nthere has been little interest in maintaining it.\n\nTo reduce the maintenance burden, let\u0027s remove it.\n\nSee the related discussion [1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/GWRYBFSYE3ZLW64WK2ECIOGN3QZYYLTD/\n\nChange-Id: Id1c6bb5c9c486c62f28feb7ba429f7af84e88db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7f13830d70d7732c44033d760bf8ae2aa962cceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7f13830d70d7732c44033d760bf8ae2aa962cceb"}]},"branch":"refs/heads/master"},"ad227a7f2701b8732dff30929df5120bf3df4e6b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-29 17:28:20.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/96/951196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/951196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/951196/2"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-28 19:15:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-29 17:27:36.000000000","tz":-300},"subject":"Remove Centos related code","message":"Remove Centos related code\n\nWe have not been testing this code for quite a some time and\nthere has been little interest in maintaining it.\n\nThe change removes all Centos related code to reduce\nthe maintenance burden.\n\nSee the related discussion [1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/GWRYBFSYE3ZLW64WK2ECIOGN3QZYYLTD/\n\nChange-Id: Id1c6bb5c9c486c62f28feb7ba429f7af84e88db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ad227a7f2701b8732dff30929df5120bf3df4e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ad227a7f2701b8732dff30929df5120bf3df4e6b"}]},"branch":"refs/heads/master"},"e167bdd4b875b45b7d26d3d3f501665182d1590e":{"kind":"REWORK","_number":3,"created":"2025-05-31 02:16:48.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/96/951196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/951196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/951196/3"}}},"commit":{"parents":[{"commit":"bd8bcf001547f579a52657591066cbc467e7a2bd","subject":"Use infra apt mirrors during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bd8bcf001547f579a52657591066cbc467e7a2bd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-28 19:15:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-31 02:16:35.000000000","tz":-300},"subject":"Remove Centos related code","message":"Remove Centos related code\n\nWe have not been testing this code for quite a some time and\nthere has been little interest in maintaining it.\n\nThe change removes all Centos related code to reduce\nthe maintenance burden.\n\nSee the related discussion [1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/GWRYBFSYE3ZLW64WK2ECIOGN3QZYYLTD/\n\nChange-Id: Id1c6bb5c9c486c62f28feb7ba429f7af84e88db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e167bdd4b875b45b7d26d3d3f501665182d1590e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e167bdd4b875b45b7d26d3d3f501665182d1590e"}]},"branch":"refs/heads/master"},"304d131aa637cf37f320b6c3b32a9dbb643adc6f":{"kind":"REWORK","_number":4,"created":"2025-05-31 04:00:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/96/951196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/951196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/951196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/951196/4"}}},"commit":{"parents":[{"commit":"bd8bcf001547f579a52657591066cbc467e7a2bd","subject":"Use infra apt mirrors during image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bd8bcf001547f579a52657591066cbc467e7a2bd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-28 19:15:04.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-31 04:00:42.000000000","tz":-300},"subject":"Remove Centos related code","message":"Remove Centos related code\n\nWe have not been testing this code for quite a some time and\nthere has been little interest in maintaining it.\n\nThe change removes all Centos related code to reduce\nthe maintenance burden.\n\nSee the related discussion [1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/GWRYBFSYE3ZLW64WK2ECIOGN3QZYYLTD/\n\nChange-Id: Id1c6bb5c9c486c62f28feb7ba429f7af84e88db0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/304d131aa637cf37f320b6c3b32a9dbb643adc6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/304d131aa637cf37f320b6c3b32a9dbb643adc6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
