)]}'
{"id":"openstack%2Floci~951431","triplet_id":"openstack%2Floci~master~I5590507f09351709c00daafb47ecebdc9fdd536a","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-06-03 16:09:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5590507f09351709c00daafb47ecebdc9fdd536a","subject":"Use infra apt mirrors during image build","status":"MERGED","created":"2025-05-30 18:08:31.000000000","updated":"2025-06-03 16:14:57.000000000","submitted":"2025-06-03 16:09:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951431","meta_rev_id":"36fd025cffabd18323eb4635e33ca3a920bdec94","_number":951431,"virtual_id_number":951431,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 16:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-06-03 14:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-06-03 13:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-06-03 14:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 18:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 13:28:52.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-06-03 14:37:54.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"710cf1645f8171b4bdb9d2a3f2f8a4a522660bc1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 18:08:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"928a3ae1f70c1289cb34c2b4697c7cd5bdeea3c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:08:43.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Job loci-build-2024-1-ubuntu_jammy depends on loci-buildset-registry which was not run.","accounts_in_message":[],"_revision_number":1},{"id":"d86df212bce5b9d729ec00d4ec895f770b880024","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 18:10:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8ff75ad526fa8ba126e2f27c7295545d69406f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:13:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b9d26317f304799893dcf192fcc9449\n\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/741297bb3de74e12bb1f85438dc12a5b : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"813f7f49752bd1d9efe0117ecb4658b42ddfad08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 20:02:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5e2394b56ec8e82d3d8374e31523e3c1893a19fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 20:37:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad3d1fcf9e344142ba18bbb4bcd69c6c\n\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/10bb265f1bf844cfabb47ed30fcd5569 : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ad1e4b044c1cb50738feb31fe2e011ff2692230b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 20:45:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9319ebf07dc0a23079f5858d8ddb197902e7395a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 20:50:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60deafaf405a4a39b1f96845d05f6dce\n\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/13e817568c17491f8d10d7d9007398f6 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":4},{"id":"57ceffd4b35f34af935f1008409f09c229d8aa7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 21:01:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4524bf451b5de1aa5f3ac6466e395ef2245a2273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:36:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9341270de66f46529f80f8e585b4e89e\n\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/48fe6b820caf49959bdba88f5f9a7714 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":5},{"id":"78f35cb7e0e6a00792a43c64a58ae6fa16394a31","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 22:08:31.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a2c85168402482af63109db4b6863e5d1d579575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:23:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62a7f6044b9f4171b176e2e59faaffeb\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/601bd6a80ae044a6857f0f17e41bbf83 : SUCCESS in 10m 15s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4947712e88bc4a86b5495600c34a577a : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":6},{"id":"6d84dadfb01450f5cf08b6691b5eb74c5bb0eff7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 22:39:45.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5722765c61ec0448bf70ef6ec14481a40f826085","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 23:33:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9e66b041961fb9c9a0b573c07ab4a92014db2c88","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-30 23:37:04.000000000","message":"Patch Set 8:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":8},{"id":"38f260b1446b23170ba89e23d34bf45c1410dde7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-31 00:03:56.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"4186fee688bbe77410cd439726f19854bc6106e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-05-31 00:23:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a5b3bc25fdcfa25b62dff7081465a53711e235f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 02:04:35.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1765f390683411b8e8a7bb4b59be8da\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b28c87dc36a243839ca0ed464ae3784c : SUCCESS in 1h 40m 51s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e7c8215137134abbbc3e2f37560b4e86 : SUCCESS in 35m 47s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/03a997ec2c204bfe88e8d824c8f28174 : SUCCESS in 57m 23s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f12140c93b8648a5a8e1a87591cc279a : SUCCESS in 34m 32s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/87ab5b1536904cdaa6235d5fa8ad08f6 : SUCCESS in 58m 24s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a3f1575c27904c73abfca8ce6fd954d7 : SUCCESS in 36m 27s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d40d5ba71c1444308d82a78bc33254e8 : SUCCESS in 57m 58s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/97e64f935c264934bf5436bd567d4ac9 : SUCCESS in 38m 00s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba3394365a9e4d3a8278db56b7427071 : SUCCESS in 59m 19s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/91f084a1de1547e9bd5ca1700901f48a : FAILURE in 21m 57s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d40a767a5bb84e0b84b66de7ba9ec396 : FAILURE in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dbb257ce990c3b94e011f593f2de29e49d7997a4","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-06-03 13:28:52.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"c7540f4b097758db9d8a0001b1b470c8d10655b4","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-06-03 14:37:54.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"679258b3e2c0ce9a3a140e86b4eafed03ba027c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:38:08.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8d7ccdec793b62441bc20d127c037c95c0045010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:09:44.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89b193fdea8a45fb84d0f856ba1d6403\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/70c4917132b641d2adbaef66669f06a5 : SUCCESS in 1h 25m 04s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/91d8faa5df7743df9901ba0b2972aeec : SUCCESS in 1h 10m 29s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eefbc0c54a3c4165a8ab82a057e39362 : SUCCESS in 1h 03m 58s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a9dee8e63c4a4a96b7da4897981ee558 : SUCCESS in 54m 25s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fddd27380da54b02ad24dfc0c9df9109 : SUCCESS in 1h 07m 53s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a5ae6ce73cb64b25b8b7ba855465f1cb : POST_FAILURE in 40m 45s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bf86ab54689441d89fda4c8d7232545f : POST_FAILURE in 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4722fe4f47187c6c99fa45e29c1b00538d4a66e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:09:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"36fd025cffabd18323eb4635e33ca3a920bdec94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:14:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55baf79cba634d0a887987ad4e57e398\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4105fbe90823430caa7874ac8e45fdbc : SUCCESS in 5m 00s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/926ad6e7cf6246cd88a5935980301d58 : SUCCESS in 4m 41s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d096aa499649486bbda97f155ebec364 : SUCCESS in 4m 32s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/714d34c4c7374a4aaca203956b617cf9 : SUCCESS in 4m 49s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4b5adeda814e4ccda9f465a50eb34365 : FAILURE in 4m 50s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7158ad2e8a5b44708766cb2d43748d35 : FAILURE in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"bd8bcf001547f579a52657591066cbc467e7a2bd","revisions":{"b8e04f9fdde8b4f05197cf675996b61d36ad2f80":{"kind":"REWORK","_number":1,"created":"2025-05-30 18:08:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/1"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b8e04f9fdde8b4f05197cf675996b61d36ad2f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b8e04f9fdde8b4f05197cf675996b61d36ad2f80"}]},"branch":"refs/heads/master"},"c05b9da5bb13d95f33bacf4347887f56968553c7":{"kind":"REWORK","_number":2,"created":"2025-05-30 18:10:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/2"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:10:24.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c05b9da5bb13d95f33bacf4347887f56968553c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c05b9da5bb13d95f33bacf4347887f56968553c7"}]},"branch":"refs/heads/master"},"c231b613f1b90396ac5f35b71a4b9caa0b04b061":{"kind":"REWORK","_number":3,"created":"2025-05-30 20:02:22.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/3"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 20:02:12.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c231b613f1b90396ac5f35b71a4b9caa0b04b061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c231b613f1b90396ac5f35b71a4b9caa0b04b061"}]},"branch":"refs/heads/master"},"d592321af52b5f6680c5e150f23f198009ad1184":{"kind":"REWORK","_number":4,"created":"2025-05-30 20:45:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/4"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 20:45:25.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d592321af52b5f6680c5e150f23f198009ad1184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d592321af52b5f6680c5e150f23f198009ad1184"}]},"branch":"refs/heads/master"},"04714c7c30736aa9de9f15e078924c07bda26c22":{"kind":"REWORK","_number":5,"created":"2025-05-30 21:01:41.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/5"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 21:01:33.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/04714c7c30736aa9de9f15e078924c07bda26c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/04714c7c30736aa9de9f15e078924c07bda26c22"}]},"branch":"refs/heads/master"},"c9dcc6c808981400aec9291635181398a6cb2613":{"kind":"REWORK","_number":6,"created":"2025-05-30 22:08:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/6"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 22:08:23.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9dcc6c808981400aec9291635181398a6cb2613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c9dcc6c808981400aec9291635181398a6cb2613"}]},"branch":"refs/heads/master"},"0dfee73df38fdbe94e45270fe02fce87bbb75e04":{"kind":"REWORK","_number":7,"created":"2025-05-30 22:39:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/7"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 22:39:39.000000000","tz":-300},"subject":"[WIP] Use local mirrors","message":"[WIP] Use local mirrors\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0dfee73df38fdbe94e45270fe02fce87bbb75e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0dfee73df38fdbe94e45270fe02fce87bbb75e04"}]},"branch":"refs/heads/master"},"305a9a07a29c85ea2a58becdec9683f6cfe563e9":{"kind":"REWORK","_number":8,"created":"2025-05-30 23:33:40.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/8"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 23:33:29.000000000","tz":-300},"subject":"Use infra apt mirrors during image build","message":"Use infra apt mirrors during image build\n\nImage build often fail due to upstream\napt mirrors issues (unavailable due to maintenance,\ntraffic limits, etc.). The change enables using\ninfra apt mirrors.\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/305a9a07a29c85ea2a58becdec9683f6cfe563e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/305a9a07a29c85ea2a58becdec9683f6cfe563e9"}]},"branch":"refs/heads/master"},"12337392027f9a3fbf7987a348dac3ee44588946":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-05-31 00:03:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/9"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-31 00:03:34.000000000","tz":-300},"subject":"Use infra apt mirrors during image build","message":"Use infra apt mirrors during image build\n\nImage build often fail due to upstream\napt mirrors issues (unavailable due to maintenance,\ntraffic limits, etc). The change enables using\ninfra apt mirrors.\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/12337392027f9a3fbf7987a348dac3ee44588946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/12337392027f9a3fbf7987a348dac3ee44588946"}]},"branch":"refs/heads/master"},"bd8bcf001547f579a52657591066cbc467e7a2bd":{"kind":"REWORK","_number":10,"created":"2025-05-31 00:23:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/951431/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/31/951431/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/31/951431/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/31/951431/10"}}},"commit":{"parents":[{"commit":"6269127b52a1259bcf38c618648666caaf6580e6","subject":"Remove support of infra project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6269127b52a1259bcf38c618648666caaf6580e6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-30 18:08:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-05-31 00:22:59.000000000","tz":-300},"subject":"Use infra apt mirrors during image build","message":"Use infra apt mirrors during image build\n\nImage build often fail due to upstream\napt mirrors issues (unavailable due to maintenance,\ntraffic limits, etc). The change enables using\ninfra apt mirrors.\n\nChange-Id: I5590507f09351709c00daafb47ecebdc9fdd536a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bd8bcf001547f579a52657591066cbc467e7a2bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bd8bcf001547f579a52657591066cbc467e7a2bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
