)]}'
{"id":"openstack%2Floci~958495","triplet_id":"openstack%2Floci~master~I56a8da49277e54eb1dee78324a301d37a8eb013e","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-08-26 02:56:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56a8da49277e54eb1dee78324a301d37a8eb013e","subject":"Fix cinder image","status":"MERGED","created":"2025-08-25 22:39:21.000000000","updated":"2025-08-26 03:03:15.000000000","submitted":"2025-08-26 02:56:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958495","meta_rev_id":"c0cf1fbb00dbab8e87e9ec823cee2aa317eba905","_number":958495,"virtual_id_number":958495,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-26 02:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-08-25 23:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-08-25 22:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-08-25 23:26:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 22:41:24.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-08-25 23:26:44.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2025-08-26 01:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3e7492a04539cb4824a6cdf9ec1f63c723b930b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-08-25 22:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf650034759b06b6e7b5585bb771f63c6eef25cb","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-08-25 22:41:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75c19c7a2e32a20a4d212653399eb7bfcc4bb7b5","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-08-25 23:26:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d0ab76e57a323469bf0fe4516fb54c8c629aaf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 01:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/490d34eede5d4cebbc63de0dcc8707de\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6edeac63a08f4930b43b1a5115486940 : SUCCESS in 2h 36m 21s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/178193eac72d464da6a85f5243bfbb37 : SUCCESS in 57m 26s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a60bdd8d29264628a67ab0cb0baa24dc : SUCCESS in 1h 01m 04s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5f92db6fa4ea4001a7e5387f25d3b808 : SUCCESS in 32m 16s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/970378c7d0804076abbb497d8e6c1f18 : SUCCESS in 55m 44s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/11fed2764bf34663b4532364f41ad35a : SUCCESS in 1h 02m 35s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2d06df3aff8e45d4b3de0984cacf7a75 : SUCCESS in 35m 38s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6a6aaccada324ee2a609a72aac8da1ff : SUCCESS in 1h 06m 15s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6f84eec5732c42df80003048d6313bed : SUCCESS in 1h 18m 51s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f800e1472fee44e9a832f4256aa87287 : SUCCESS in 47m 26s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/038f9b0052484ac99a560eb0e77af138 : SUCCESS in 1h 16m 58s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7b3e2f78494942b4b0d913ba7ec48d9d : SUCCESS in 1h 05m 19s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4e195640fddc4ad997cad5c26fdf744e : SUCCESS in 43m 57s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c2c036b683e04d32a28b442a6dd3aa69 : FAILURE in 38m 17s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/99c14e06c484475ea4bb8dec353e3033 : FAILURE in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"166b8ffd410d0f9addc3e8f3dd781677eb6dc02c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 01:16:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12767db05e202ccbaa0d3c0cb7a69c4c8874f10d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 02:56:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79420f5c18fe471da6c5e80e8db81bf2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d1363cada0b047dd9198ae3c07eff685 : SUCCESS in 1h 38m 55s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c58dda13f0ee4e369bd1d1d3b14f3b92 : SUCCESS in 1h 16m 17s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/59521cf3802d44a08e5ae2445aac1726 : SUCCESS in 45m 36s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/908b365b32ac4376a794719069d3ed43 : SUCCESS in 1h 29m 24s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2fae5d7b883c4f02821665f45bc53679 : SUCCESS in 1h 35m 56s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b2f1dc29dfdd4d5ca3763c96ee3bce67 : POST_FAILURE in 28m 58s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ab34f458234c42169adc8438854d65f7 : POST_FAILURE in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22d4c3daff70278d49ff99d87fa9dcbc9052a66d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 02:56:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0cf1fbb00dbab8e87e9ec823cee2aa317eba905","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-26 03:03:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/321385f41cdd4175b0453b69ad17f9c5\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d478118a997243edb34db9390296988c : SUCCESS in 4m 54s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9c1bcf9cffb24daf9bac168d48dd8227 : SUCCESS in 4m 42s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eebd7203c4a64a56bb06348e7a558eae : SUCCESS in 5m 40s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/40c6be7886c94052b24a3d4ab988f95a : SUCCESS in 5m 50s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3965cc0aadb14cea866287855089815a : FAILURE in 4m 25s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5ade607912ea428986334656bfbd811a : FAILURE in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf4bdc33cd491dda254090d9c10f811cc874b20b","revisions":{"cf4bdc33cd491dda254090d9c10f811cc874b20b":{"kind":"REWORK","_number":1,"created":"2025-08-25 22:39:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/95/958495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/95/958495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/958495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/958495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/95/958495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/95/958495/1"}}},"commit":{"parents":[{"commit":"bb90c9c1a5950abcf84b6bf9ec78ba0b55f16c9a","subject":"Add CloudKitty image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bb90c9c1a5950abcf84b6bf9ec78ba0b55f16c9a"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-08-25 22:37:35.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-08-25 22:39:10.000000000","tz":-300},"subject":"Fix cinder image","message":"Fix cinder image\n\nAlso always run cinder test job. Last commit\nsucceeded because cinder job was skipped\ndue to files filter in the parent cinder job.\n\nChange-Id: I56a8da49277e54eb1dee78324a301d37a8eb013e\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cf4bdc33cd491dda254090d9c10f811cc874b20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cf4bdc33cd491dda254090d9c10f811cc874b20b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
