)]}'
{"id":"openstack%2Floci~961288","triplet_id":"openstack%2Floci~master~Ic4a67fb6d74440b0801049638600e6b6fd41e74f","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-09-16 19:53:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4a67fb6d74440b0801049638600e6b6fd41e74f","subject":"Fix image build","status":"MERGED","created":"2025-09-15 19:47:26.000000000","updated":"2025-09-16 19:59:04.000000000","submitted":"2025-09-16 19:53:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961288","meta_rev_id":"d6233c41c6eb508c4c75756e0404c23cf52b6c3d","_number":961288,"virtual_id_number":961288,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-16 19:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-09-16 17:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-09-16 17:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":1,"date":"2025-09-16 17:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 20:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-16 17:10:14.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2025-09-16 17:30:42.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"00b4e62e6cac1623966e028d27e4c9a9127f07d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-15 19:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccc6ed2643fc2458e62d64f64824e491b9c9d414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-15 20:14:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49191e503df35243f6296f322326b3603ed51225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 20:28:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b21e8c657c2f4c8aa05e416544249e6e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6ea885efa5a34c699488e09428eb5d8e : SUCCESS in 12m 22s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bd7f7ddecb454ec8b221b0d2e3ca9f81 : FAILURE in 2m 41s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1d4f362a7eaf46d6b2553b316cb4b60c : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e808bab79cad45b2ba41b283fd5bfcf9 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2e2a3ad6b1714f14a8787eb0779f1ac7 : FAILURE in 2m 40s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d8b9789aed644011b7955cbe5c680cc4 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0b1422d7e9124044809cf116afde1536 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fc8c65f24bb04c4b993a63dc56b02141 : FAILURE in 4m 07s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3944114a3009420c85f3aadba94bd9ab : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/314153a3a52842558129ef174c7c5ff9 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/03e9d5a9e8654f798016fc1b4e0cf658 : FAILURE in 2m 37s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/442866543c8047c5b5b27b326dc5a709 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7603579f75444c98916f3af3f9d1b4d2 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e17b7532cb5c415282facae3096665c8 : FAILURE in 2m 33s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bfa7e2741053496dbb20e0e07ca6703b : FAILURE in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0cbc72d51c6becc97d832a9eaff32b805fa9d8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-15 21:54:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b55acce31c1799d2f850222d40f9308e724cb9d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 22:33:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48dd754d8fb6497990563af75fcf3d84\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7bedc8849d9a4d75b1aed46f69ebb566 : SUCCESS in 37m 28s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e69701e4d124a878d434f2986e20fb6 : FAILURE in 8m 29s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/070d9274237a440aaa513b6918158774 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b13ee7675aeb4ae9b9d9a4d3893b89e1 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/954e07e93e9246fab54874c31082e462 : FAILURE in 10m 33s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a5dd4264d5f2424cbb47b0d786df3627 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3a77cbbf8cb947e6bf6c350d93731ce2 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/52fcc7808ce84842ace302955d0f3c74 : FAILURE in 10m 34s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3786ebc0240442aa8e0ac12bd54546d7 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/20f2d2e076444d5993c7ebf6b9722f22 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b6d4cfac926445f09e657bd550185221 : FAILURE in 10m 29s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/070fddadafec43d589b723fd9ea44aa9 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/24970d4506b542c5ac871b8c51c7444a : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6fc79e1eb27a4589b597c99e59daec73 : FAILURE in 8m 14s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b6691d1089e74ed1bd6ee9772ddbceb5 : FAILURE in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"204c6492b11effd041c636ce805ca6b388491e1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-15 22:53:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"15ba77f20903b0d0679b45c3e10fb78c447cb670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 01:28:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a39f10859d74d419c513d06377496a0\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0910d86ecfec4938951e93319e56c151 : SUCCESS in 2h 34m 39s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a38146bafd3b4d3fbd8987af40736099 : SUCCESS in 1h 00m 38s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7d2d58e21adf41909ffff012a9481115 : SUCCESS in 1h 00m 58s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/476f051263be4f659b4bdd23d6640297 : SUCCESS in 41m 54s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7acec5c9adbb41bd875f546d98ffe87d : SUCCESS in 1h 03m 42s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7d7a132d8272431188fc246a32ad9ae9 : SUCCESS in 1h 15m 35s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b1503b2a8b5048eeb735d670295e53e1 : SUCCESS in 36m 29s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cb37ae9650ba4ff99f38b57ecf926d01 : SUCCESS in 1h 09m 44s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/41df6ef8e9f541c697508972d5d4751d : SUCCESS in 1h 16m 22s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2edb9d4f1ae849789e639b0870fab8d3 : SUCCESS in 36m 27s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/12d4ff58328b4b8db9bf48c28bb49d01 : SUCCESS in 59m 45s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ea32fed30c4a4377aa0a754890110023 : SUCCESS in 1h 18m 23s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e56f85e201544a8ba530f4e218c770cd : SUCCESS in 36m 56s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89f0e0f95a66457ea40a4be24193c634 : FAILURE in 35m 37s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d981899dc18a485d8eaa178859766e20 : FAILURE in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"53eb0775c84c282841a6ecd860e11bb70c8c457d","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-09-16 17:10:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"86d194a834436aa605b6b01cde62aafab241ea37","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-09-16 17:30:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"edb638994895b49a79bda406bbf88c9332570fee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 17:30:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c8762b62bb0a8477d952f9032d4b974adb95156a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 19:53:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baac4aca86b2457c86f78bda1f49000b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/27c669d14c3547c2ae60ef96f9ae6418 : SUCCESS in 2h 02m 33s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/34cc20a628ac4ddf9bb0b2ed1638092d : SUCCESS in 1h 31m 03s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d8736d0fef3d4c8fa86d8fdc39c2b060 : SUCCESS in 1h 14m 53s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f30ea265688a4e9391ecb19269805cda : SUCCESS in 1h 30m 08s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b13af2e9302f48238440c9b118612851 : SUCCESS in 1h 43m 26s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1884242aa74f4198b8e88d7b15edae95 : POST_FAILURE in 1h 02m 33s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4b8379a0614d4c7ea12431f5a4959599 : POST_FAILURE in 1h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08834e7cc2966c8eccac95446d5abeb2fcfed081","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 19:53:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d6233c41c6eb508c4c75756e0404c23cf52b6c3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 19:59:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/893428fe2d2c4484bf2158068b37c75e\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0015c567886a4850ab6d5ce9fd8bd072 : SUCCESS in 4m 26s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/04a4e4d0350c48d78818d4b7ff05a33d : SUCCESS in 4m 37s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2596f339c17048d7a3e7b8de6a7fd4c5 : SUCCESS in 5m 23s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0061c79130f449ebb4a42f95cd06a244 : SUCCESS in 5m 30s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a18fc0ceba464511a5bf38a56fc8d115 : FAILURE in 4m 50s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/99d497d55805411fa48a7b28fa731821 : FAILURE in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"10539ce036b359cb6ac729915ca2cd46f9c9d254","revisions":{"8a743c9f0421f1ea674574db208f120b2fded09a":{"kind":"REWORK","_number":1,"created":"2025-09-15 19:47:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/88/961288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/88/961288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/88/961288/1"}}},"commit":{"parents":[{"commit":"cf4bdc33cd491dda254090d9c10f811cc874b20b","subject":"Fix cinder image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cf4bdc33cd491dda254090d9c10f811cc874b20b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 19:45:06.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 19:47:12.000000000","tz":-300},"subject":"Fix image build","message":"Fix image build\n\nRecently the build-container-image role was updated to\nuse buildx by default. One of changes is that the role\nnow checks if image repository is a full name and\ncontains \u0027/\u0027.\n\nChange-Id: Ic4a67fb6d74440b0801049638600e6b6fd41e74f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8a743c9f0421f1ea674574db208f120b2fded09a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8a743c9f0421f1ea674574db208f120b2fded09a"}]},"branch":"refs/heads/master"},"85698b08c2c59d4f0e2d9ebb7970720603895334":{"kind":"REWORK","_number":2,"created":"2025-09-15 20:14:11.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/88/961288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/88/961288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/88/961288/2"}}},"commit":{"parents":[{"commit":"cf4bdc33cd491dda254090d9c10f811cc874b20b","subject":"Fix cinder image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cf4bdc33cd491dda254090d9c10f811cc874b20b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 19:45:06.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 20:13:59.000000000","tz":-300},"subject":"Fix image build","message":"Fix image build\n\nRecently the build-container-image role was updated to\nuse buildx by default. One of changes is that the role\nnow checks if image repository is a full name and\ncontains \u0027/\u0027.\n\nChange-Id: Ic4a67fb6d74440b0801049638600e6b6fd41e74f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/85698b08c2c59d4f0e2d9ebb7970720603895334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/85698b08c2c59d4f0e2d9ebb7970720603895334"}]},"branch":"refs/heads/master"},"1478cfdd7a1499e5ce5426734f10d76efb97db2d":{"kind":"REWORK","_number":3,"created":"2025-09-15 21:54:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/88/961288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/88/961288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/88/961288/3"}}},"commit":{"parents":[{"commit":"cf4bdc33cd491dda254090d9c10f811cc874b20b","subject":"Fix cinder image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cf4bdc33cd491dda254090d9c10f811cc874b20b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 19:45:06.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 21:54:34.000000000","tz":-300},"subject":"Fix image build","message":"Fix image build\n\nRecently the build-container-image role was updated to\nuse buildx by default. One of changes is that the role\nnow checks if image repository is a full name and\ncontains \u0027/\u0027.\n\nChange-Id: Ic4a67fb6d74440b0801049638600e6b6fd41e74f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1478cfdd7a1499e5ce5426734f10d76efb97db2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1478cfdd7a1499e5ce5426734f10d76efb97db2d"}]},"branch":"refs/heads/master"},"10539ce036b359cb6ac729915ca2cd46f9c9d254":{"kind":"REWORK","_number":4,"created":"2025-09-15 22:53:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/88/961288/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/88/961288/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/88/961288/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/88/961288/4"}}},"commit":{"parents":[{"commit":"cf4bdc33cd491dda254090d9c10f811cc874b20b","subject":"Fix cinder image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cf4bdc33cd491dda254090d9c10f811cc874b20b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 19:45:06.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-15 22:53:17.000000000","tz":-300},"subject":"Fix image build","message":"Fix image build\n\nRecently the build-container-image role was updated to\nuse buildx by default. One of changes is that the role\nnow checks if image repository is a full name and\ncontains \u0027/\u0027.\n\nChange-Id: Ic4a67fb6d74440b0801049638600e6b6fd41e74f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/10539ce036b359cb6ac729915ca2cd46f9c9d254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/10539ce036b359cb6ac729915ca2cd46f9c9d254"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
