)]}'
{"id":"openstack%2Floci~962314","triplet_id":"openstack%2Floci~master~Ifb245aece9306bad169e25188fd9d7e7103b105a","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-09-27 08:37:45.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"Ifb245aece9306bad169e25188fd9d7e7103b105a","subject":"Add Mistral image","status":"MERGED","created":"2025-09-25 21:18:15.000000000","updated":"2025-09-27 10:29:14.000000000","submitted":"2025-09-27 10:23:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962314","meta_rev_id":"3ddf7b51be33ef8b4adb2420c22fce03355b12e9","_number":962314,"virtual_id_number":962314,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-27 10:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2025-09-26 18:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2025-09-27 08:37:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-25 23:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-26 18:59:38.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"1e5fd75a2951049c9054c0ad04a8802272f9a33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-25 21:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa822c3eda33be83796330d75dd21493486147ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 23:46:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75602867a9b340b391ed189906897bd5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a458aa78a8f24a7a962fcf4a1ed65b1d : SUCCESS in 2h 27m 49s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/36b306c036c1408198cad6528b4afc44 : SUCCESS in 48m 54s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3b8a169e008f479a9f13ef28259ef592 : SUCCESS in 46m 19s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5b8ac304432046fe97efe4a696933cda : SUCCESS in 41m 22s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3d24d76684bb40a7b588426d22bbfab5 : SUCCESS in 46m 13s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/da7a46215aea4c40994255492a7b69c7 : SUCCESS in 1h 14m 52s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1002940ad65b4bfaa7afbe8f335ca0db : SUCCESS in 40m 32s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9e67f406ce464e6c95e6a45d55215359 : SUCCESS in 55m 49s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/054b5b80122f460e800f1b4114326319 : SUCCESS in 1h 10m 42s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/628366b8e7c6486cbe2f4c9281ccf074 : SUCCESS in 31m 54s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a6f8b02128d04d55b3872b7dd561b4b6 : SUCCESS in 1h 04m 19s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/85da6e3a7dc749568d47f9ecf5383293 : SUCCESS in 1h 15m 10s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1a164484ab144d91a427ab14fa13e149 : SUCCESS in 43m 02s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6958e435606c4b44a2d683710fefc7dc : FAILURE in 36m 52s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f07d1aa8d24845ccb9d2a1ca696d0b25 : FAILURE in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f71d984fb60171b507ad59dbcfbe41e70044e70b","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-09-26 18:59:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e29752a14243a48756d28cc30a2dc68c2fffeaa","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-27 08:37:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"283998681fd9b858cb07e4a4287a00df1272212f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 08:37:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b5ba68bc60f2df535d59b921dfd21bbc290b08b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 10:23:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ec728eba984985b5b33aa55e1056c3\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9bbe37de2f074deca864239eb595e7ab : SUCCESS in 1h 44m 32s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/51ec7ed3b2344451ad4ccb247e6a4185 : SUCCESS in 44m 04s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2284732de60f415aac296cfa0ea25f78 : SUCCESS in 1h 19m 34s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d5182d43a5624520aa34bff4a13a233b : SUCCESS in 1h 29m 46s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ac2176954afc4cd19ec9fdfc35b2d108 : SUCCESS in 1h 38m 18s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/390b1ef1f36941aab750c7a68498b75c : POST_FAILURE in 33m 58s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bf880c29cbdf4125b4c880de6c359ec3 : POST_FAILURE in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c449fdae4ac0fb4dce191c8bcc5af8a979ce293c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 10:23:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3ddf7b51be33ef8b4adb2420c22fce03355b12e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 10:29:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c41161ee76a84b36ac08a3a125efc619\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ef3ff645ec0448aca57daacead5c1676 : SUCCESS in 4m 30s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5cb86608a02e48b9a40432c31ab68fd3 : SUCCESS in 4m 37s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/be8ba6a7ac72449d923aed80c3b6faa8 : SUCCESS in 5m 22s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bd9f914dcdaa4aac88bc3dedd54b98c6 : SUCCESS in 6m 02s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a602c32d437146f2b80249dcb6d965ac : FAILURE in 4m 30s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e6858a8217bc42e48b46db5342484510 : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f051211d29abe73969643560f2ea3c8d07a489c6","revisions":{"f051211d29abe73969643560f2ea3c8d07a489c6":{"kind":"REWORK","_number":1,"created":"2025-09-25 21:18:15.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/14/962314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/14/962314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/962314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/962314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/14/962314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/14/962314/1"}}},"commit":{"parents":[{"commit":"8a142eaae239adc09ac1895b29aefff424e13192","subject":"Add Redis module to Zaqar image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8a142eaae239adc09ac1895b29aefff424e13192"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-25 21:17:38.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-25 21:18:05.000000000","tz":-300},"subject":"Add Mistral image","message":"Add Mistral image\n\nChange-Id: Ifb245aece9306bad169e25188fd9d7e7103b105a\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f051211d29abe73969643560f2ea3c8d07a489c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f051211d29abe73969643560f2ea3c8d07a489c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
