)]}'
{"id":"openstack%2Floci~962420","triplet_id":"openstack%2Floci~master~I7ba7a6e51ec38e6b62dca1d7adeb8ec984718b11","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-09-29 19:30:57.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I7ba7a6e51ec38e6b62dca1d7adeb8ec984718b11","subject":"Add ceilometer image","status":"MERGED","created":"2025-09-27 11:00:26.000000000","updated":"2025-09-29 21:13:19.000000000","submitted":"2025-09-29 21:07:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962420","meta_rev_id":"02aafdc5f2d43ffef86b8662d60483e5aabaf57e","_number":962420,"virtual_id_number":962420,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-29 21:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2025-09-29 18:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2025-09-29 19:30:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-27 14:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 18:42:03.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"7c8b46e19419dc142d5105ee4cb096b3c3d7989d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-27 11:00:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e20d796218f345dec3bb5c17626f298c529713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 14:02:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff11b43e408426aa695ca5ac2ad0e68\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4f3ce9bb98904f4da0ccd74a9921a77d : SUCCESS in 3h 01m 02s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f4302483376448e4b0c8beb397a07d30 : SUCCESS in 32m 39s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9069ce1d6f3b4ac8a8eeed602b0316e0 : SUCCESS in 1h 12m 35s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/56a39eaa0439407c9f1dc8cd3b9e2d0a : SUCCESS in 27m 56s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/770c76f75a494a82967169e6c0517d7d : SUCCESS in 32m 34s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f259c3ffc0ec43be91bcdd3ce67fadd6 : SUCCESS in 1h 03m 51s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/86018399529146efa1f2c552a3939ebe : SUCCESS in 30m 18s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b81c6f87cf80477faba0725aa6dd3dfd : SUCCESS in 55m 20s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8d9d764bd8df44cfa8c703e3844af741 : SUCCESS in 59m 57s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cb778aecc7694f8c92591c2950c0e69e : SUCCESS in 34m 01s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/71a037acfe564befbd48fb089a17242a : SUCCESS in 1h 29m 27s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2b5b07af415543bbac69f3d0e6f4c4f1 : SUCCESS in 1h 18m 42s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb3e120433f0446bb9df6dd06e3fcd37 : SUCCESS in 40m 55s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eaff9ce7333e498885e6a83a2d38eec7 : FAILURE in 48m 49s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8ff5bb414e7a4da98b41b69b07887add : FAILURE in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e8d75512d1eac1274a0e8d8c5b9b699103600c5","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-09-29 18:42:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c88291ae86f85d90195c15a02558cfdfea4efed3","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-09-29 19:30:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8099cba12c4eefe061787b8363dcb619cd9a58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 19:31:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b26fc6a37ee825b456eb6224904b735ab6bb540","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 21:07:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c6fcce15dfe4972b8a963e898f88f79\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/180fdb2df14849e3a4f34a8dfbdacaad : SUCCESS in 1h 30m 28s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2f9ce9c9693947beb2302025bfb814e9 : SUCCESS in 1h 20m 59s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f8e7b0fab63a4634ad238ff450c7ba41 : SUCCESS in 1h 23m 15s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c163a8f0fe794a7aa4f3cc62614e8975 : SUCCESS in 54m 25s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/04e512876cc54feda3606b4024578f94 : SUCCESS in 1h 04m 49s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/598e01e7fa1f4d679e0f8bf69d3b35ad : POST_FAILURE in 55m 51s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8001b107a64344ada8dea7d2e68a7ae4 : POST_FAILURE in 1h 00m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"311e78c4c79074f72a85967fd7b5c2ac7a4ac4f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 21:07:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02aafdc5f2d43ffef86b8662d60483e5aabaf57e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 21:13:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6dab561562242a8a2d0a99a5f4a13d9\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/30d56526f9644894befe3c93968568e3 : SUCCESS in 5m 20s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2767016fc0024a2f83f7208b07f08e14 : SUCCESS in 5m 07s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/68915ab292494e06a2d5d7b10ee57202 : SUCCESS in 5m 57s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a1a41bdc14a14ea0ac97a298e307220f : SUCCESS in 5m 50s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b41885ea203a4cecb592ebcada088c3d : FAILURE in 4m 56s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/71523a93f0c94c77b82afad613394357 : FAILURE in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71c33d4d441036e54486baaee2397f3211876827","revisions":{"71c33d4d441036e54486baaee2397f3211876827":{"kind":"REWORK","_number":1,"created":"2025-09-27 11:00:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/20/962420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/962420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/962420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/962420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/962420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/962420/1"}}},"commit":{"parents":[{"commit":"f051211d29abe73969643560f2ea3c8d07a489c6","subject":"Add Mistral image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f051211d29abe73969643560f2ea3c8d07a489c6"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-27 09:38:54.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-09-27 11:00:17.000000000","tz":-300},"subject":"Add ceilometer image","message":"Add ceilometer image\n\nChange-Id: I7ba7a6e51ec38e6b62dca1d7adeb8ec984718b11\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/71c33d4d441036e54486baaee2397f3211876827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/71c33d4d441036e54486baaee2397f3211876827"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
