)]}'
{"id":"openstack%2Floci~964332","triplet_id":"openstack%2Floci~master~I37393416e40c66b292bdee22dc9e541490d428aa","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-10-23 15:02:43.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I37393416e40c66b292bdee22dc9e541490d428aa","subject":"Use customized zuul-jobs roles","status":"MERGED","created":"2025-10-17 19:49:21.000000000","updated":"2025-10-23 20:15:43.000000000","submitted":"2025-10-23 20:07:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964332","meta_rev_id":"c632762188319e4804b48f8f2f1d3f551300a54b","_number":964332,"virtual_id_number":964332,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-23 20:07:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-10-20 20:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-10-20 18:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-10-20 20:40:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-17 20:08:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 18:30:00.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-10-20 20:40:31.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"bb231b6af9bac1e172aa14d677b07234d5a58835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-17 19:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e63a0174a410128a3d7baa7e623f845754d98d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 20:08:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15882c14a309467a9024b24654f6a463\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c8bab25845fd403bbfc4fa319ca5686b : SUCCESS in 13m 29s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c6193d2086064861834d2126a0ff077c : FAILURE in 1m 41s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7a06cc7deccc4e5287d17434a7b1b6be : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/37edf1707e50473c825f895f25625b06 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f163884050e54ac8b1d7ededb31b3040 : FAILURE in 1m 43s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/48a9a245e4564c1fbb1026e94457ada9 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b33cd8742ee0448a8f87682eb6041776 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e4e60ef5a2f741d48f056a240de8f711 : FAILURE in 2m 50s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/dee2f8ceef4c48d6a5f14657e594e6af : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bd15be5fda964adbb00b11fc27a942af : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d761a1c1580d4012abe2751c41fd24f7 : FAILURE in 4m 04s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5c3e0cfdb81142b48779965b80122e48 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1308f8679a574e6eb7d614a5b25acc6c : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5eb204ed905842febcfa8ddf5820a2f4 : FAILURE in 2m 46s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebee890ce0904643a10b191d0759e36f : FAILURE in 2m 28s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5d4d9ca886484d7f8e99d6bab2cccaff : FAILURE in 2m 45s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1500eabe9d5f4178a4026a71dc9eca36 : FAILURE in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23cefe44495c7065bf2b85a0a2af9d0d35c98ada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-17 21:09:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6009f7267a00ce094d4b604a8b44f86e9a333475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 00:12:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51fe1b2ac30b4639803a5da91016f188\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c826214570d24ad7ab020684eba14cca : SUCCESS in 2h 58m 23s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e9e1a3fcb0774c3b88d4254f7e77915f : SUCCESS in 1h 12m 20s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e5606094eb0c4a689cdc973abb311cbf : SUCCESS in 43m 36s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/215791cfa4d84634b5e803acb11cd0e4 : SUCCESS in 39m 46s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fb2988574a3546b880ba91b839e831f7 : SUCCESS in 1h 09m 09s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bd9f5496328f4f81a7125e65f5236a6a : SUCCESS in 1h 17m 27s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/65e77ba1aea84a75a67221156c94b011 : SUCCESS in 40m 56s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e44db0648b7d4e57b98fca06a73fa9c5 : SUCCESS in 54m 39s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/433e1902520a462c86f5e6d57b489e55 : SUCCESS in 46m 21s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bcc524dec6fd40ea956d08f51c24e6f4 : SUCCESS in 34m 54s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/aad114143f3c434c9e44ede2dba19fe8 : SUCCESS in 1h 29m 21s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/65b48ee0bf744d43ad43e523d036ba92 : SUCCESS in 1h 16m 37s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/26b6d87f283e4b92b5086c8776c67c68 : SUCCESS in 45m 28s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c267afcbcc65492aaab749eb78a8a0e1 : SUCCESS in 1h 21m 37s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ff891463805f4b57853b30b1fb1229e0 : SUCCESS in 55m 04s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8365c205a03644d9aac258c55c6e573d : FAILURE in 51m 08s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/97835de3cc9949aaa7cc2a8fbd8954de : FAILURE in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e13b8a7f0de610446e7670d0f498300d11818a1","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-10-20 18:30:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dca3f752afae91fedb1656ce10e72d9414ef6c50","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-10-20 20:40:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e8b308f284f4db2804d7c44d74da36f4318addb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 20:40:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4be6b5797febfc496de6604160402ec1bb694a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-20 22:19:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6359bb4c9fd43d29da2aaa26733d9cd\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4c3bb7d27ba041669bf7468938564b37 : SUCCESS in 1h 36m 59s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d67c8b800f374784849e8165a3dc1a36 : POST_FAILURE in 52m 27s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cdf44b1025aa4dceaa0101fccbe3d46b : POST_FAILURE in 1h 11m 35s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90c4bdf4e5804f1aa06e7436973d454c : POST_FAILURE in 1h 21m 21s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/054ed02e0ffc405885edd479cae0f6fa : POST_FAILURE in 1h 23m 24s\n- loci-upload-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bfc034e27c0647a2939f3c8220ae88c4 : POST_FAILURE in 1h 18m 45s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f9d2cd5199cb438998f2500523883cae : POST_FAILURE in 1h 22m 51s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ac2fbda0a21f469bbbd1419d071c5e8e : POST_FAILURE in 56m 15s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/27e914b65d234aff9b6ec4db861b9284 : POST_FAILURE in 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c05aeb02c1924b9ba4be637e67700c4d37d44365","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-21 19:12:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2ccf5afb040ae12d08d74d779f3f8039afb6b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 22:05:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7811523bdddd4bac807ea4bfe0110c7b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/8aaad124be50419587902d16bb81154f : SUCCESS in 2h 48m 22s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cb2f88f4df7c40fe9e9b35de6b0c1567 : SUCCESS in 44m 23s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7e86a697feae414e972c15644796a996 : SUCCESS in 57m 53s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3ea1e4a0768443e490c3c938fe2cca98 : SUCCESS in 26m 58s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b8c0c684fa12410f8bb016135b22c0ce : SUCCESS in 1h 17m 23s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2bde4e3fe3854a1393ed978bd47450b1 : SUCCESS in 1h 17m 21s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/df59374d11e048978b0a14d5f18de092 : SUCCESS in 35m 55s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4994a8ddc80a4e2488f1253120546570 : SUCCESS in 1h 18m 39s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e2ca2bcde4d34015936c6ba48a81ac29 : SUCCESS in 1h 10m 58s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cab81c2c172642c2a81e1499c7336d22 : SUCCESS in 32m 33s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a1c521fc42e74024b9382515ddd848c4 : SUCCESS in 1h 22m 59s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/61c1a3a4d8644560bad5132b8d904d44 : SUCCESS in 58m 21s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7bd9e4a907d142598cc8d1726f631096 : SUCCESS in 45m 28s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/434b6930c361449b978774e2cfeed183 : SUCCESS in 40m 54s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a0f50f84593a4b3eae252d53220b8224 : SUCCESS in 41m 15s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/13a5ed800af048018d98a5c9c48990a6 : FAILURE in 51m 31s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6d57a46058cf47f9aba997fc85b2ce01 : FAILURE in 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9af79d01f648b213507c17c4abf503f2cd67e89a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 22:06:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32f91499ffb0f19bfd6c2927a01fd5fd926d5fe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 00:02:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d71884a8f6e4bc0bc5c436bae44fa2e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0735b810487640e8b8e5c1f7bcb76498 : SUCCESS in 1h 50m 38s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e39329bf6f444029917b42da7338a782 : SUCCESS in 1h 27m 31s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f928de8be36c4f638ae1deda36d322e1 : SUCCESS in 1h 05m 20s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eff3148d86a64db8b4dd03f891208ee0 : SUCCESS in 57m 37s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9b4c4f2c6082447a8aa89f1213aded69 : SUCCESS in 1h 22m 36s\n- loci-upload-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2a0dc21b9f974c8c87f833a1a85c0bbe : SUCCESS in 1h 43m 12s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5bfe281476e24e3c9d03b23aaefff308 : POST_FAILURE in 1h 02m 30s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0215d7dc8f304fa48a9e4564961d06db : POST_FAILURE in 48m 19s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/748b3f9ba342477ca0827b1025642737 : POST_FAILURE in 38m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09bbe0c4f25a6f9919d98abc77a309ee13a5dee5","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-22 01:47:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ee115e8a203aa5f4c8ae67af5972aaea6a5eaf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 04:30:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b7b8109dcdf4e04a0280bbdaf0553ad\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4ab468ca643f4780997e5a343c273eb2 : SUCCESS in 2h 39m 03s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ec45b0679bde453391578a3dc5987c1b : SUCCESS in 1h 13m 03s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/93a70bb436ba46b7a0f65e25a962ab54 : SUCCESS in 1h 13m 36s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5bc7dbb16ed949dc8df5ef2325e14152 : SUCCESS in 43m 09s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e12d0d5ac95c47a9a9aabcc0a4a11e20 : FAILURE in 43m 37s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eccfe2bdbe22489ab5889c33abafa16c : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b7382a2a415b4ad591d6937d2ce76f6c : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/de180240fa3547e2b01256472e8708a8 : FAILURE in 43m 50s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4e513783814b4c7dacde123ebbfc0599 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ed0ddb9b02334111a4f62ab39c78a71a : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fae59e1846f946d08b8059d96d5225f7 : SUCCESS in 1h 29m 56s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/84989606f4134aaf96f8324a37c2fe5e : SUCCESS in 44m 47s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8b6c7ef331884d26ad466ae5cff18c0e : SUCCESS in 30m 36s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/54008a85914c432f938bac78a0c4dd71 : FAILURE in 44m 09s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/180a8aba36654d3e8a134dbab33fca35 : SUCCESS in 1h 32m 51s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0908401919a6448bb9ab8aa3427219eb : FAILURE in 51m 07s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3f62ad5afabf4574bd466f8a16135838 : FAILURE in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6a33b2033c962e6f7999cb10aeff6383577b3cd","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-22 14:09:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b264d22b6bbe8a73976af9d63efe35f124c77ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 17:10:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7160aeb2fe347fa90815356819d5e0e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bcb66ef47061475ab7234588ddc8eca4 : SUCCESS in 2h 56m 20s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/71c9e3c769974ad6a5f28f67be22d8b1 : SUCCESS in 35m 39s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/da9cc8c94b6c45739ececd77d6130ba0 : SUCCESS in 46m 36s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b452588b6670401dbc2ba60f38faad44 : SUCCESS in 36m 53s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/37d5c332dda04ed99ac3450e007655fc : SUCCESS in 56m 06s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a63f6a8bddd345548529ba82a61af892 : SUCCESS in 1h 14m 07s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b35594200ca34e7e9193ca3947fd8935 : SUCCESS in 41m 52s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d32e265ad35340ca96638b193baa6a92 : SUCCESS in 41m 55s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b42eb21c4cd2412aa09ea57ef9cf5041 : SUCCESS in 1h 06m 50s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3f774fb271084dae92c75da21c89c3e5 : SUCCESS in 49m 20s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e81f3ac092e6452ebcee4496449b41d8 : SUCCESS in 1h 26m 24s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c5964e2143b247dc93a86f6befdebe62 : SUCCESS in 1h 17m 52s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/53ff068253f74e3892381462f37e4536 : SUCCESS in 35m 05s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9ffc12202c734815a9f0348da23e2d57 : SUCCESS in 1h 17m 05s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba85f61cb5934bbb87043040eaa9c54d : SUCCESS in 1h 22m 49s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6d27fd25dd4e466fa65185d0d6387b22 : FAILURE in 39m 52s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6ab981ea2f744cd9bfa15bd873f2da79 : FAILURE in 57m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93b086cff9926f0d6e8e9de14c13385286d97c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 17:11:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bb669457e7552d4799e84c6aeb6e51bc61e05c30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 19:10:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efa39de55caa48b88e89b3a28f3bee0a\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/07ba39c8ff65437692b118140190b232 : SUCCESS in 1h 58m 41s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/105b7aaa3b1a4436b7e603633bacbbe8 : POST_FAILURE in 23m 19s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4de21f45930d4784b77b8a5f2344d45e : POST_FAILURE in 23m 21s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/95c398b043d7433baa74883ce81afdbc : POST_FAILURE in 14m 47s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fa2289a8746d4573b86c41a2f8d5d85e : POST_FAILURE in 1h 53m 02s\n- loci-upload-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4880638c4eee45b083ecc00873d1316e : SUCCESS in 1h 00m 38s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/aeefdfe5526b4d6b8dd42a78b3e23a83 : POST_FAILURE in 21m 52s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cf2c4d4b3e7e4c9b894350c3f6f2d92d : POST_FAILURE in 1h 13m 22s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/beb7fe251f89478ca39a4f612f644d5b : POST_FAILURE in 55m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"921fa1a5414ffdf50e717936237dc8b92153f8d1","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-22 21:52:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"897477001468c21cc12c00364d4c10f4b67fbca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 00:43:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31a5b65546754434873a13b498188ac7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f7534ef1d5134850bbedcf5cc7fb3865 : SUCCESS in 2h 50m 08s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7a807c768d6c40c4ac1f7990977097f5 : SUCCESS in 1h 10m 35s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ba08ae30f1ab4dbeb007e60aa8e5668d : SUCCESS in 1h 11m 07s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/28bc39f13ef8477299b9a00d84d8aa5d : SUCCESS in 42m 23s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/08a460bd486347399d12785c630f11bb : SUCCESS in 37m 22s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1adf3f71df5d4f8e9381860a4f545be8 : SUCCESS in 49m 45s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/62f2bcf890714d258f9ae6233e7aa226 : SUCCESS in 45m 01s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ca6e3212e6c240fdadae7bac1941ce54 : SUCCESS in 1h 25m 46s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5f2a0dcb2d0d49e48b76406244237069 : SUCCESS in 1h 13m 00s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2ced868a6251419aaa15c3d358090b8f : SUCCESS in 41m 51s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/26e8e85b0d694ea8a64fb00180efa1a6 : FAILURE in 30m 51s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8861d4ed48214b0f84921fa72cb1b2de : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/00f1136b2a58472890b3fe0bfbc80e19 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6fddd975baec4ae59649e45d45c6920b : SUCCESS in 39m 54s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/87fff941f9604e4e9e67215e2d8c38d9 : FAILURE in 17m 24s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d4dc3e55fb1c451b94f667722b834727 : FAILURE in 25m 19s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2b912b184dd5493fba2b02aa20a52a31 : FAILURE in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a726e523ccd2e9858cb902eebd7d8567e9759c7","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 01:04:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d4a3d9c2890eeed78d7ab45a98b5e67d009e4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 04:05:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/380e92a1cc204dc89536fc2baefde00c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1b04ab2be4f64742be04635050bb5a69 : SUCCESS in 2h 58m 02s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b56aff2de6804fe0ae31188d15401bc4 : FAILURE in 32m 18s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/926317845dd4481fac78861e870fed86 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f5485bdeec6346a39d994549986e2562 : SKIPPED Skipped due to failed job loci-build-2024-1-ubuntu_jammy\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/656b0e93b696412a9ba8ba74325ae994 : FAILURE in 35m 37s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/285efcaa95cd4c97bc6afe6488136356 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2c196f8ee85c442882025c6fd4787115 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1ddaa02a39d54f5b87cf3f850ff65177 : SUCCESS in 1h 29m 15s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/df19b29aa34148419dfc1ccc85132fc1 : SUCCESS in 1h 13m 47s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/12890eaf8e7144da95808315a5c45edb : SUCCESS in 41m 20s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d2c4ea774c1a49908e7b60507de8803c : FAILURE in 35m 23s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7f969f51a123486b8817bac6cd9a133e : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b904c17cdac04b4dad70868ff983aa65 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a50b9d451b0c4d4e935be83a8273c599 : SUCCESS in 55m 06s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dfb84d369dd147d89e8c16bb1f1c37a4 : SUCCESS in 44m 41s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a834a391d5bd4f1bb57db298cbefcb12 : FAILURE in 24m 39s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dd03aeaf2d1c4ecea1b2d549a828f962 : FAILURE in 26m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"650533ff1cc847013e6f90d809e157368acbc62f","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 15:02:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"405b64499c1e503118cb9006b53dfcb84dd77318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 17:39:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b2a31c21e1a4a25a586a9533303cc5c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3bc857a2c5094421b64afb2c503e7440 : SUCCESS in 2h 35m 18s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f01df1034fad4f2ea8cb4cfcc4b18110 : SUCCESS in 54m 41s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fa32642e4c4b4c768f4e71e881c0fe0d : SUCCESS in 1h 17m 05s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6229dff55b4f4a4081648357c577b453 : SUCCESS in 36m 36s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/22f4d4f2207e44ff999f5a1e443a767f : SUCCESS in 1h 09m 37s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e05cefd9c09349bbacb2b246194507ed : SUCCESS in 58m 56s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/72e6dadd3ebe4e5682f6f9b566816099 : SUCCESS in 44m 08s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a179a7454c4e4912ab4ddfd65690b60c : SUCCESS in 1h 05m 03s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/35b57350ae7b448b9b3b0255bbe5a52b : SUCCESS in 1h 13m 12s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c752ba9c0e594868b4d2e37e3524dd0c : SUCCESS in 36m 18s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/51efe3a15e4047ee9510f9f4fa92266e : SUCCESS in 1h 28m 58s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0a4a447052eb4ea488fbe42e54342845 : SUCCESS in 53m 25s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d0aefce5f48047e5a36a4a4496e4ea1e : SUCCESS in 30m 30s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3309ac20a26948448a58c85da64ecfeb : SUCCESS in 37m 00s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e596b91d77704d71bdf70716d5187be6 : SUCCESS in 1h 05m 39s\n- loci-build-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/082b8667adaf4584a6c409360301144d : FAILURE in 42m 02s (non-voting)\n- loci-build-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c88ca7d5ad3e4e23b472a13092e505ea : FAILURE in 52m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"185121f3aee3c1bcd9c577af4122e0f9737da491","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 17:39:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"23f0a5db711f966eef575e3bc7905ffc4779848c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:07:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43cf08b9627499da5eebcb382411fda\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/67b1b9977a7948ed9cb72620d9404e44 : SUCCESS in 2h 00m 50s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d8f974f6dc314e18830321df782e86ad : SUCCESS in 1h 29m 35s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/636baa037bdf495ea04b920cf1cb1f4d : SUCCESS in 50m 00s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/15a5800592d745fa90bd11e46714353d : SUCCESS in 1h 23m 36s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/be9bd68e157043f0b79f31dd98b66ac5 : SUCCESS in 1h 53m 20s\n- loci-upload-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b5273f082d434bc786b6ab93fd2e548f : SUCCESS in 1h 22m 28s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dd25173d2c6f4bd8b085e80c2090ec06 : SUCCESS in 1h 46m 41s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/87f752c311ec4363b1749c70d3790a10 : POST_FAILURE in 1h 03m 47s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5cab4f766fca4e9bb884d28b42f4ce85 : POST_FAILURE in 1h 07m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce30f906aa2b027866cc8ac5ce3cb8c1558cc818","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:07:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c632762188319e4804b48f8f2f1d3f551300a54b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:15:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae968c29cb134ee5bf0c8556301f5a9d\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/72d71127e04a45389e239dbd17da97a1 : SUCCESS in 5m 28s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c68489df6d47421780865b3aef59a201 : SUCCESS in 5m 55s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f201de5e558c46bbbc37986a3ee5592d : SUCCESS in 7m 00s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6f3e32fcf43949e5a26a9806e69bdd7a : SUCCESS in 6m 08s\n- loci-promote-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ea3df8a6cee444f696503eaaadc6b699 : SUCCESS in 5m 49s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/81c993a4f2c64d3892a9bf19e5b821b0 : SUCCESS in 6m 08s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/679c8ca5e29049879922fe22a84c0d52 : FAILURE in 5m 04s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/68dd355513754fdc917d8c756797c05e : FAILURE in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a02b9a49d1ee45a7878e54eecb732ac9f4360e26","revisions":{"933269eb7ecc62a519400421b6d478bbde4d85a3":{"kind":"REWORK","_number":1,"created":"2025-10-17 19:49:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/32/964332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/964332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/964332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/964332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/964332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/964332/1"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-17 19:34:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-17 19:49:14.000000000","tz":-300},"subject":"Use customized zuul-jobs roles","message":"Use customized zuul-jobs roles\n\nThis is a temporary change until this is merged\nhttps://review.opendev.org/c/zuul/zuul-jobs/+/964255\n\nChange-Id: I37393416e40c66b292bdee22dc9e541490d428aa\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/933269eb7ecc62a519400421b6d478bbde4d85a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/933269eb7ecc62a519400421b6d478bbde4d85a3"}]},"branch":"refs/heads/master"},"a02b9a49d1ee45a7878e54eecb732ac9f4360e26":{"kind":"REWORK","_number":2,"created":"2025-10-17 21:09:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/32/964332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/964332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/964332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/964332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/964332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/964332/2"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-17 19:34:08.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-17 21:08:59.000000000","tz":-300},"subject":"Use customized zuul-jobs roles","message":"Use customized zuul-jobs roles\n\nThis is a temporary change until this is merged\nhttps://review.opendev.org/c/zuul/zuul-jobs/+/964255\n\nChange-Id: I37393416e40c66b292bdee22dc9e541490d428aa\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a02b9a49d1ee45a7878e54eecb732ac9f4360e26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a02b9a49d1ee45a7878e54eecb732ac9f4360e26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
