)]}'
{"id":"openstack%2Floci~964668","triplet_id":"openstack%2Floci~master~Icea2cfa7d3b058e581aa730062bb0adee9a955d0","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-10-23 18:09:18.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"Icea2cfa7d3b058e581aa730062bb0adee9a955d0","subject":"Build skyline image","status":"MERGED","created":"2025-10-22 22:19:14.000000000","updated":"2025-10-23 23:04:48.000000000","submitted":"2025-10-23 22:57:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964668","meta_rev_id":"0490d974e536a280584832c210ef57b4e71e9b03","_number":964668,"virtual_id_number":964668,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-23 22:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-10-23 16:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-10-23 15:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-10-23 16:02:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-22 23:57:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-23 15:58:06.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-10-23 16:02:07.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"ecbc5f4660212cacdccaae32d4148d222470b053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-22 22:19:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99fbd57957bbbd50a2e4d284b15ca8f3b85ab2da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-22 23:26:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4b355d308c01708416293c6925b2a8e360ad45a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 23:57:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29f5a5b355a34999949b1bc3585b7a2f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6e91dbd21d2845178006eb49e6b49122 : SUCCESS in 27m 02s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/de6e42e3f5954863896d55b1de0fcfef : FAILURE in 18m 25s","accounts_in_message":[],"_revision_number":2},{"id":"0ad4aae658eceacc7bbb34c19bbf20a632d035ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 01:06:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"422cec6798cedff287fa3edbc42b4794a46bcd31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 01:07:00.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"39a41528fd773c025c8a828b3ee7063e08ef83d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 01:23:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/495420be77884e7882005caf3711dee9\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/176589ac2cc04c50ac9c9ed0e2cca420 : SUCCESS in 16m 03s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e9d367804bfa48d6b6eea50282fd9ed6 : FAILURE in 12m 59s","accounts_in_message":[],"_revision_number":4},{"id":"14a2cc710582f6687b8f40845ae261ab39576c0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 02:05:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a48cf01fc52ddda2afae1b9a7657b4bff2e7f9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 02:51:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ee252cf45a4fbd925f47fa182e2245\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2cc1118c50ad4f2a9431ca9cca30f93a : SUCCESS in 29m 17s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b902ad28ea0c4970bdf532d203f0cf57 : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":5},{"id":"cd1e8737ef399d7aa2187e09024344b1e31ea1b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 03:06:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"6234056f86cfd3c53524d9205cd5c332a3679742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 05:20:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4766f013751b4889a83403ec2f11a7dd\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a52ccfcaad1e413eaf67608b48860cc1 : SUCCESS in 2h 12m 35s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9f364326cb324ddca4a48da1f4b183df : SUCCESS in 33m 54s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/958a8588294349bd89d3022462e3ba92 : SUCCESS in 50m 39s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b8cfafc420b643adaccbea03368582e9 : SUCCESS in 35m 20s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/77a5cab99a694d89980da3079006e75c : SUCCESS in 34m 19s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c0b0d680463d4ed69b77e0b3d53f2c24 : SUCCESS in 59m 58s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/085c46280d574ccdba37479461faaaf3 : SUCCESS in 35m 51s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/aa6478f719b44e018aee1e48d1caa0e0 : SUCCESS in 38m 57s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/686f3745b727451880fb9736524c8066 : SUCCESS in 49m 38s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e550bd889516487ca3e4c12be7f388ba : SUCCESS in 30m 34s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/53cd5b070cf3455d902980b23b339f54 : SUCCESS in 41m 37s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0601fc99072447e89fb7b1d207ea9879 : SUCCESS in 1h 16m 00s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7fd00da05f394e9488286f42e0fb4659 : SUCCESS in 42m 59s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/552c9510658840b49dffea029f1d9240 : SUCCESS in 36m 59s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f2d793597b61476f9ad0b9a2539e570e : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":6},{"id":"86f806efdac194cc41f368a6dcb728662be95890","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-10-23 15:58:06.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a4e01ec194e11f058174243fcac6ff581edd5b62","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-10-23 16:02:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5ac2925e215fc899225f6fd1c0deb03cc5dc9a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 16:02:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"333f8ac39a5c69bbe2f8d56d1f5a74c2d864c110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 18:02:32.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad67de6d6b654d119b8f6f65a5481ed4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7810c26840c946db9bae14625de4041b : SUCCESS in 1h 55m 29s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f3f1229800c24a7c8a07f5c18577e772 : SUCCESS in 55m 57s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/726a77a66f4e43069cfd2a622047ff6c : SUCCESS in 1h 07m 48s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/95c492f14a144650b784241334881bb7 : SUCCESS in 57m 20s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/da0992d35b9947f18a0c66b5d25fbd02 : SUCCESS in 1h 07m 54s\n- loci-upload-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d9767044e7d04dfd928a00030e98a7e5 : SUCCESS in 1h 14m 28s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e2569b0005c4499fbded1abebb67f78b : POST_FAILURE in 1h 51m 16s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/31a92c5e69384a099879a851a3180fea : POST_FAILURE in 1h 04m 31s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4cbfd80147104da3adec92ef7d5260b7 : POST_FAILURE in 51m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"12ee9647726835d6f6e0078121be0a8aca5385c9","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-23 18:09:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"683339aa82783dbd796d6c3818fed93b5b62b63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:59:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6505a34d56b847e0b8ce2a021f0665aa\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b76690df4df8444fa6c3472df1a45240 : SUCCESS in 2h 48m 49s (non-voting)\n- loci-build-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/287180ffe6f144cea6f1f79aeb0bdb17 : SUCCESS in 35m 22s\n- loci-compute-kit-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4ce5b8f273474388a6d0c5a24f69f409 : SUCCESS in 1h 16m 18s\n- loci-cinder-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1fcf5c3c542c4b48ac5959a46b70b327 : SUCCESS in 36m 12s\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eee857786b5f4e89b0b454fcbf6fdd11 : SUCCESS in 1h 14m 51s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/54b07d0afceb49dd85ae7b7c4775fe77 : SUCCESS in 1h 13m 56s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c3355af2d72542d6bc531c3e723e20a6 : SUCCESS in 40m 56s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/696a57c81d3f41649bd0664ef3905ed2 : SUCCESS in 1h 16m 45s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f0b6c737a19f49ea9d5c2b7f62393491 : SUCCESS in 1h 18m 31s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/667dfac072894bee832376d7e837936f : SUCCESS in 32m 53s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1bcfff10d4174ddda019c33d4dd3f127 : SUCCESS in 51m 33s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/428d3da99ae345299100c5b5d97c68e3 : SUCCESS in 1h 16m 40s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebe6b53fb6474d138d3c135bea7910d1 : SUCCESS in 44m 52s\n- loci-build-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5d1d29eff84d406883fd5bb00587cc88 : SUCCESS in 1h 15m 03s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6f5776494f644e69a48045b6f6f6d013 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":6},{"id":"2d7489ab297e1a39eee81caacf54bbedf5a9d2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 20:59:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"12f06d86712addee7eb0724f013c612d6538c174","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 22:57:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8fa9a5f4d324a4f905f678ca9a5a7bf\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3cfe4d92c60249c0af288b86653129cb : SUCCESS in 1h 57m 57s (non-voting)\n- loci-upload-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c93e27bcb8f74d2688c20df146618ada : SUCCESS in 1h 34m 20s\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9a1f650d9baf4f9da0a35617b613b741 : SUCCESS in 51m 12s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c6f30ef07da04c039c5f00aff3bc389b : SUCCESS in 1h 47m 17s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ea47a434c2424f2f8550b83587894333 : SUCCESS in 1h 50m 13s\n- loci-upload-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c2695ad0ce2445ec954ff02c2a928a79 : SUCCESS in 56m 37s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b0d62509a01b44a9a7290214e4c43664 : SUCCESS in 1h 28m 23s\n- loci-upload-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8e50252c62c54f2098af759911d516b9 : POST_FAILURE in 38m 20s (non-voting)\n- loci-upload-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1cab4e181fcd4529807d13c8adf1ced7 : POST_FAILURE in 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9c5841063eb51d0817aa78f57eb716172ec5cbb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 22:57:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"0490d974e536a280584832c210ef57b4e71e9b03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 23:04:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb751e7ec694ab6a766a2ccf7878163\n\n- loci-promote-2024-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e60caa53d6ca4fd78b5cdd21f3b2a220 : SUCCESS in 5m 39s\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c814cc6d389040a69162cba12ea6202e : SUCCESS in 5m 40s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/66cb70fc5a914b528a3f819a586e424d : SUCCESS in 6m 11s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7806dc15752d461d9c298aeb8898d058 : SUCCESS in 6m 21s\n- loci-promote-2025-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/633ca0660d1f4d8693641711563bad2a : SUCCESS in 5m 54s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/71f23784b6e24ed29bbd7e8fc01fca89 : SUCCESS in 6m 19s\n- loci-promote-master-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/114da4038d844236bd49201631c51e85 : FAILURE in 4m 54s (non-voting)\n- loci-promote-master-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1df6409360e44cc4afc7c4d6535f1603 : FAILURE in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"650657e7fcb5a762be70c4b51f1c3aa686f5bb60","revisions":{"470d98fbe426234c0027453172e2aba1ad3c06a9":{"kind":"REWORK","_number":1,"created":"2025-10-22 22:19:14.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/68/964668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/68/964668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/68/964668/1"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:18:47.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:19:09.000000000","tz":-300},"subject":"Build skyline image","message":"Build skyline image\n\nChange-Id: Icea2cfa7d3b058e581aa730062bb0adee9a955d0\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/470d98fbe426234c0027453172e2aba1ad3c06a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/470d98fbe426234c0027453172e2aba1ad3c06a9"}]},"branch":"refs/heads/master"},"6099fa04ae3a857f7ffe71b25ba03cc2c1031ef2":{"kind":"REWORK","_number":2,"created":"2025-10-22 23:26:11.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/68/964668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/68/964668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/68/964668/2"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:18:47.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 23:26:05.000000000","tz":-300},"subject":"Build skyline image","message":"Build skyline image\n\nChange-Id: Icea2cfa7d3b058e581aa730062bb0adee9a955d0\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6099fa04ae3a857f7ffe71b25ba03cc2c1031ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6099fa04ae3a857f7ffe71b25ba03cc2c1031ef2"}]},"branch":"refs/heads/master"},"41bc0ae460ce76133b4790d1dde3934238c849d8":{"kind":"REWORK","_number":3,"created":"2025-10-23 01:06:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/68/964668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/68/964668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/68/964668/3"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:18:47.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-23 01:06:36.000000000","tz":-300},"subject":"Build skyline image","message":"Build skyline image\n\nChange-Id: Icea2cfa7d3b058e581aa730062bb0adee9a955d0\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/41bc0ae460ce76133b4790d1dde3934238c849d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/41bc0ae460ce76133b4790d1dde3934238c849d8"}]},"branch":"refs/heads/master"},"8e09ef154ec91b5281af5b96d105f036570f91e8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-10-23 01:07:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/68/964668/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/68/964668/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/68/964668/4"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:18:47.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-23 01:06:44.000000000","tz":-300},"subject":"[WIP] Build skyline image","message":"[WIP] Build skyline image\n\nChange-Id: Icea2cfa7d3b058e581aa730062bb0adee9a955d0\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e09ef154ec91b5281af5b96d105f036570f91e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e09ef154ec91b5281af5b96d105f036570f91e8"}]},"branch":"refs/heads/master"},"3fbbaf1d9e16b3aeacf13c5366859d47885f4a31":{"kind":"REWORK","_number":5,"created":"2025-10-23 02:05:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/68/964668/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/68/964668/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/68/964668/5"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:18:47.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-23 02:05:50.000000000","tz":-300},"subject":"[WIP] Build skyline image","message":"[WIP] Build skyline image\n\nChange-Id: Icea2cfa7d3b058e581aa730062bb0adee9a955d0\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3fbbaf1d9e16b3aeacf13c5366859d47885f4a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3fbbaf1d9e16b3aeacf13c5366859d47885f4a31"}]},"branch":"refs/heads/master"},"650657e7fcb5a762be70c4b51f1c3aa686f5bb60":{"kind":"REWORK","_number":6,"created":"2025-10-23 03:06:58.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/68/964668/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/68/964668/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/68/964668/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/68/964668/6"}}},"commit":{"parents":[{"commit":"1dcdd534abdb4dd122e91e758c46b0181e6ac652","subject":"Build magnum images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1dcdd534abdb4dd122e91e758c46b0181e6ac652"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-22 22:18:47.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-23 03:06:49.000000000","tz":-300},"subject":"Build skyline image","message":"Build skyline image\n\nChange-Id: Icea2cfa7d3b058e581aa730062bb0adee9a955d0\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/650657e7fcb5a762be70c4b51f1c3aa686f5bb60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/650657e7fcb5a762be70c4b51f1c3aa686f5bb60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
