)]}'
{"id":"openstack%2Floci~965823","triplet_id":"openstack%2Floci~master~I1a82bb1cc70c33d49c79ff7c724ab6842467aa11","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-11-04 05:05:33.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I1a82bb1cc70c33d49c79ff7c724ab6842467aa11","subject":"Disable unnecessary jobs","status":"MERGED","created":"2025-10-31 16:47:29.000000000","updated":"2025-11-04 20:29:59.000000000","submitted":"2025-11-04 20:22:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965823","meta_rev_id":"ae96943296ea3b421e1789de108bc760489c9850","_number":965823,"virtual_id_number":965823,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-04 20:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2025-11-03 20:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2025-11-03 20:37:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 21:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-03 20:10:57.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"b93a2737d74be7beac1e0fd48a47f472924c9cc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-31 16:47:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2401d20148e71c4efd59066808529dc89f725b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:07:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5630a8d960c47b6aa3c82efef5d15c8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/33c6a3177445465c84278388b2e87700 : SUCCESS in 4h 08m 17s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7fe0066fe38c431c80dc8ecb9d1a3bac : SUCCESS in 1h 15m 33s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ea01f7905592430b97adbbfea5ed4344 : SUCCESS in 1h 16m 03s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b1e9f46c48d840f4ad05b059199cc6a7 : SUCCESS in 33m 01s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b0ff0acdcc1a471a82b2f14f95bc9b61 : SUCCESS in 1h 39m 19s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/812a1a095eaf47308263c664a16ffeda : FAILURE in 2h 07m 41s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2ccd8d866edc47d4b288e72af49b18fa : SUCCESS in 30m 02s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/229b04e8e2fe473ebab2cc7090183a61 : SUCCESS in 1h 36m 07s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e8e7648c54ca44daad22c01fe3beaf3a : SUCCESS in 1h 18m 20s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9fdc8ead57b444c4827e4649bbb9f53f : SUCCESS in 45m 43s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a9ccfb815f184b59aaeee5dd5d171592 : SUCCESS in 1h 25m 45s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d5ee5915c6de48c0a2374f7dce6802ff : SUCCESS in 49m 05s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d977b08a3645485ebed46554163f113e : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6edc1988158aca537222bcf22b2180243d89167a","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-10-31 21:40:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd20568af22fe8406c2325f92d7592883af22f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 01:00:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552308bffc724b319bb315301fff27c7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/688635475c604718b65694d34a496498 : SUCCESS in 3h 02m 00s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ced5cba211d94a5fa283b9c32017656a : SUCCESS in 56m 19s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5fb13797073c4d5bb5cdb52ac48afc9e : SUCCESS in 1h 32m 16s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d189a91b81bf46419992f0764163e868 : SUCCESS in 37m 38s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0d5ec5fb52c84b66b6713d8530c29013 : SUCCESS in 1h 02m 05s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e30467d9ef274b3eae1db206d0a2c787 : SUCCESS in 46m 24s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fb9252c0f95041ca81cbd27d798d979c : SUCCESS in 43m 18s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c169d5746d8e4e6581245bdbb24d5580 : SUCCESS in 1h 13m 35s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eff34fc19b104cbea17183eada008258 : SUCCESS in 1h 39m 06s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ddb3563029864ebaadefedbb3e883f3b : SUCCESS in 35m 07s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9bc76c96b9c843d181a2c153b1f88c89 : SUCCESS in 40m 37s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2a251bd7664c4ee296a7ec7b5dce351e : SUCCESS in 1h 17m 49s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/31547a6651c74685b34f3b59ec3150e4 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c67c194b9302f022e23090cccc85b3420902d7c7","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-11-03 20:10:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90854f56f311a0e154c534a4ce3e522cc47d0984","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-03 20:37:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4525f332f96343dd99b00abd5a9ffe663e244a96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 20:38:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"142a158c25b2f02e745c672f939a4d99bf2ba7e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 22:10:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f48557ccab1f4a79937ea38b4b27502d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3696c2f3689647adba6a767b8cebf269 : SUCCESS in 1h 31m 07s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1c38f14ae9864e589293408e41e87a7c : SUCCESS in 1h 12m 30s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a567984eac384f74b9690a5f7f993948 : POST_FAILURE in 1h 20m 01s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dec193d8aa124425bba7f4b8ac921644 : SUCCESS in 1h 27m 09s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7ea81c03971c4188971614dad574a770 : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8226934cec1f708ae8f8f87977c7aa9f634be5ba","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-03 22:14:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0eeceba599cde3441fd645d554bd47453891f9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 01:26:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d86e577cef724ae39c657e9624bb55dc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/aa42ed7bf87846b79e62a08e97bd41da : SUCCESS in 3h 07m 05s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/aa0ee1a3c2b34caa9d5cd6b1770c0bc7 : SUCCESS in 1h 10m 18s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/04b5a719be2c4231aacd2a13de967ebe : SUCCESS in 1h 10m 26s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8980f5afd5dd4e86b0ccd3442433de8d : SUCCESS in 43m 33s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b906cfbf70b2414b862f188748881313 : SUCCESS in 1h 26m 19s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a0469906c4f8404aae0306cfb923426b : SUCCESS in 1h 13m 08s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2355854260344e14a68308cb66d6fae0 : SUCCESS in 43m 53s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3b8550a87b2e4676aaea8809adf75fde : SUCCESS in 1h 02m 57s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a882e7332947405688470513ece5a6ef : SUCCESS in 1h 03m 25s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9e699d4899f3407f9bce277201b03f21 : SUCCESS in 36m 05s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/be10f263ac5243bc94ce0e08daff8517 : SUCCESS in 1h 25m 37s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4dad117b262f4f0b8f5ea102159e45d9 : FAILURE in 1h 33m 08s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a9ef1085ba3148688a3ac67d17ee3ee7 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6ab19e861f123f09b8c296266b3e7f1707a5172e","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-04 05:05:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ed19894159f392def7b6d34c4401cabb4ea5512","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-04 15:17:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72c2d3dcfcd4ee896e37ca908ec44f08ce9a6d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 18:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a9bd79dbbf40f7aa352285b2514f4c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/18684c346b4c4c6eb45e8003c5ae17b2 : SUCCESS in 2h 59m 13s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c528189dd3a34e9aa56806bd8cd4546c : SUCCESS in 1h 08m 09s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d2ea54b3f0a549d287676aa7d6e3a7dc : SUCCESS in 1h 11m 22s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9ef71455c5474ce99a299127936aadfd : SUCCESS in 41m 47s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3ce962eb05d242b39f6ff053325fd4cc : SUCCESS in 1h 17m 12s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5f9b52c310554513a8597db047ee51ec : SUCCESS in 1h 13m 44s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c11a378b18bc4d368f3723dcb95b9dbe : SUCCESS in 33m 04s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7ff08477ca1645878db1d0d1582cb8f4 : SUCCESS in 1h 24m 29s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4361ead40efe445c84e092d63737ad02 : SUCCESS in 1h 19m 32s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/77317bbb2a6c40e39f9217e38c7d5756 : SUCCESS in 32m 10s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4b6dffd73f654543917526c7eaa412d9 : SUCCESS in 1h 10m 33s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/54343809bbee4402a4c7d27a91ae1a36 : SUCCESS in 1h 20m 10s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d7b8e15d4b154cb8817270ea0409934c : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1ec9a062d583a6d312b94613ec4133c95ac805c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 18:19:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3f8df164bb9e6298d4cf06f2c08907309bdb495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:22:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ffc94d623ed420d86bebafbbc8553ef\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ca8ebbd2c6c94341938262ccc5273bf7 : SUCCESS in 1h 57m 44s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e60b304e0d074b6185c7e46000b77f59 : SUCCESS in 58m 28s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/452512399d4249369ef295d80945aa62 : SUCCESS in 1h 49m 31s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2fbddfc322d949189159d83457679cf5 : SUCCESS in 1h 27m 33s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0fbb3a0714cb4b8e8ae0040438a86dbf : SUCCESS in 1h 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"69059006d1321a258dea09414ee8c79eaff97534","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:22:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae96943296ea3b421e1789de108bc760489c9850","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba8c093a890a4992bd0d906dbc135bd5\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/71888aa74c9747738b5e6bfec74d5ba3 : SUCCESS in 5m 43s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/01b7d2dfca5745c1bec4a0d2f78073b5 : SUCCESS in 6m 44s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/75cdfddacbde4824b01bb64c8f8a51f3 : SUCCESS in 6m 29s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/15b10cb0360a43d9989834615ef036b0 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e7f4d3aaa36e236913eb281232653d064381466","revisions":{"4e7f4d3aaa36e236913eb281232653d064381466":{"kind":"REWORK","_number":1,"created":"2025-10-31 16:47:29.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/23/965823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/23/965823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/965823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/965823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/23/965823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/23/965823/1"}}},"commit":{"parents":[{"commit":"132501bd9bb0f5c989690e1ae3d4d0dedeaf6f78","subject":"Merge \"Build skyline image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/132501bd9bb0f5c989690e1ae3d4d0dedeaf6f78"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-31 16:43:59.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-10-31 16:47:23.000000000","tz":-300},"subject":"Disable unnecessary jobs","message":"Disable unnecessary jobs\n\nThe builds are not stable due to infra, so\nlet\u0027s reduce the number of jobs.\n\n- 2024.1 is unmaintained since 2025-10-24\n- 2025.2 Ubuntu Jammy is not needed\n- Master builds are unstable. Let\u0027s comment them\n  out.\n\nChange-Id: I1a82bb1cc70c33d49c79ff7c724ab6842467aa11\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4e7f4d3aaa36e236913eb281232653d064381466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4e7f4d3aaa36e236913eb281232653d064381466"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
