)]}'
{"id":"openstack%2Floci~966622","triplet_id":"openstack%2Floci~master~Iff223e795ed727551fbc496559bf74e44f793b33","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-11-18 23:26:24.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Iff223e795ed727551fbc496559bf74e44f793b33","subject":"[WIP] Use uv instead of pip while building images","status":"ABANDONED","created":"2025-11-10 17:35:47.000000000","updated":"2025-12-30 15:46:13.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"093509b7c7fc175d5e1240807a3ca0466e87b1ad","_number":966622,"virtual_id_number":966622,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-26 07:42:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-11-10 17:50:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86da1dcc0ade7a53d680da06483ba63dfbc6edc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-10 17:35:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e44b89ce8a09cf70d1289941a37c085df78ce97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:50:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fbed353d86f4ad5be612024ee8d30bf\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6e123f41deae4bb6adbcdce90b5c0b98 : SUCCESS in 13m 24s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6cbb0ffd3adb4035bfa6d6218513239b : FAILURE in 5m 36s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ee08dbb1e539470e8634218c229f4f18 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ae7d3540fb4040f18ad1eec7a90bd02a : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9261aa656f3447b6bb538745c384abbc : FAILURE in 5m 37s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d18d1ce7f7494ce9a402ed3d194d85b6 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4e428463f50e451198af91b348617606 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/979415dfa24d470aaf70537f8f8b4a6e : FAILURE in 4m 54s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/af7026fb72ae4d419adc808a963537fe : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/339b543b486a4b578dd2ddd41b0a54c8 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cfbcd3d244f54c24a7d894820cc4137f : FAILURE in 5m 47s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e727f4120d4b4278a1a056c15f54a1d3 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ce906220adea43acba9789e33b7627b7 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble","accounts_in_message":[],"_revision_number":1},{"id":"d58897902a323be0d20e87d0559aef98ce4f5e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-10 17:56:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a054e5b0bab12d96e3e0afaf96ae0d94b6be00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 18:24:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe1883bda2d34fb3a5b3e8b309a86209\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/30a62708a89d44cea1cbf2543e9868a9 : SUCCESS in 27m 40s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f6bac04a983244ba95decf774415cfb1 : FAILURE in 12m 38s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f5a64d33fced46c685ca64fb46aa80d2 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/57c61ecd771041b49086fd860eefd892 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/409a0bbcdfb04ae3b1ae26500c06bb4a : FAILURE in 13m 00s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ec7747aa25ef4ce796ec68bd9b8ee6bd : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c8c4fc21eb96438bb72e5250008dd3a1 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/276315790f514cd8be274b6ae2a59355 : FAILURE in 19m 58s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2a40ba9cfa3844c3a918ef9a141ebf0a : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e570046d44a64456855a571167312caa : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/29d43bec18cf49199e388618545ef21d : FAILURE in 17m 57s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9ac86861e52c4dfc98ac8a467b76bd19 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/009297a8c26644a8bbd01d4c8a23c0fe : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble","accounts_in_message":[],"_revision_number":2},{"id":"d48c32a08c0b57aa6480d3a7dbba384939e1197b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-10 20:26:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"473d39fb5a740fdb6c32d500e7d8258468874878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 20:58:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5b7e64f39a84a1280c99de790161444\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2458ca3435ec43b1b490f14407a361c3 : SUCCESS in 26m 13s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba2df61e1a44480398aaf81a5874a136 : FAILURE in 18m 37s","accounts_in_message":[],"_revision_number":3},{"id":"58f4dea38b2f309abc14c855e703f71e8706b98e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-10 21:13:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"941166b9126fa2b79ee8f1736d845ac313c0eec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 21:23:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/840092914a0e44f989163fcfa66faf2e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c6d7a7f551bd4cf08b1891277d60309a : SUCCESS in 8m 45s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d59bcc66a9ab4621b9ca50057a931c76 : FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"20b49bac380fd44e5c3ca64ee1e5c1aa736ba131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-10 22:15:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"93028e446c9886e397cf20b4735254c65773f03c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 22:37:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ac5525838184ca094fe117bf20ccd93\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2fb315b7a037477ca5da18bd8c2886b5 : SUCCESS in 17m 21s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d7819e821349432fbd5bf3f2a68d19e4 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":5},{"id":"73d626bb08a188098948c4e3ef8cd66af7dd29b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-10 23:26:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5f2e67ec269c22d17e33717c8e51d5865cda7b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 23:58:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/274a9b78a6804c078e9697a2179ab0a5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1c542bd006f74fe6a1bd2e0d6f122f3a : SUCCESS in 29m 49s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0b4fe12cc2c941a08c3e49a725b2f518 : FAILURE in 22m 12s","accounts_in_message":[],"_revision_number":6},{"id":"6e31edf9767c7d03dea7253a556c1bfea43963cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-11 03:05:23.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7da20016dba1bf845cea4a532e45fc3d9ab3928a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 03:43:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c77d6afe2b33484ebad9467bdc178f19\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d1fce538408f49aaa740d0878ab84294 : SUCCESS in 18m 34s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3c9fa0439ac948f7ad399e7d95125235 : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":7},{"id":"a10b20e918b4eb8a680f6636e22e8d9679218c4b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-18 23:26:24.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"0363e93d63d2dfbdacf22549d5b13ea755551bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 23:41:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd61b6aad22a4daeb19793ad41a25555\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/481e482f0557459bb2b800db82c20a77 : SUCCESS in 14m 07s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e6fc93c1b89a41998ef26350aa3c8ffd : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":8},{"id":"45c22d3b013c07faaae7072d8635275841217d27","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-19 00:53:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"d68e92bb5ed6eeae241a51c6b35acd5b89833979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 01:18:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/595295bce5f546b38d98a1e76be08bc2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1ac22513bede42df8376cba597836975 : SUCCESS in 19m 07s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c26d8a6da3844954b857b325e0583b6c : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":9},{"id":"115e689ec40f74fecc1228d167d450efc73991bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 04:49:56.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8494b025f328b3a1af18e7b1c9d189c9daae06ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 05:31:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d0fd30dbdc04cc6a096fadd2e6dfd4f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/719479e4081f4c7f901b1c9987b32a3e : SUCCESS in 40m 57s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6772e611b6694a9f995cf9695eee87b5 : FAILURE in 34m 21s","accounts_in_message":[],"_revision_number":10},{"id":"4f714899e29ee60972af624ef4ead81b04900710","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 06:34:32.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"a0eea9980b3f2f5e0b4d206834c8c9b49494aec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 07:58:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f77d835e22f4acf939baeaee4cda873\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2634d8880c5542ef83382ef876a7ef81 : SUCCESS in 1h 18m 30s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9f2ea4bbe9c54f5596a9bd262621d613 : FAILURE in 1h 10m 24s","accounts_in_message":[],"_revision_number":11},{"id":"19f34f7bb2137cb98ca37f9ee32177c9806cd85c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 16:31:58.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a57b503a5802c78bb7f28ea1c1d2292221f331e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:10:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d863ff218f8246c7ae79a1d09afa7bb4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/718c0d1702644c75a1cb4630d3f6eb76 : SUCCESS in 37m 40s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2ec4e21ea44347ce9d8e83db9bb24aa5 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":12},{"id":"42cfa397d52287ba793b8b8f216ac3321a6aec86","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 18:09:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9cdaaefc79b37ca6d603116ea785700136d36117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 19:09:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2512a4c70c044f09c6d8b1aebc59adf\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a417e570c57f42be8ede47a6ac63a63d : SUCCESS in 59m 22s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a6645f4004e34caf955a9258d98ee9dd : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":12},{"id":"249c39aa4b5c9816b3eaa53b8cc0cd0a90397a19","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 19:45:21.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"964bb446eb11df1fa3b7af13e5606a2bfff6ecec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 21:17:11.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a5d3b6b8c2401fbf41227c4b518453\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d27c0e54068a4ba6a647f9efc2e9efd6 : SUCCESS in 1h 27m 05s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/95a86459d67e4962bbe72b453e51ab86 : SUCCESS in 1h 03m 55s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3b661bd7e9444acb9ba1fc9a39ed6c85 : SUCCESS in 42m 19s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a3524c713875400badc69d8adbb2fc3a : SUCCESS in 1h 19m 32s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f300d807017848c886b72a28393e9406 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":13},{"id":"b2a0b471c9ef9321938aaff68e3db22d8079ef1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 21:36:15.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"58751cfaf31a40b72f30a85a68ca57460c8fe047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 21:56:07.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2204bf6b26894eae8d50b49f1a81378e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/44b8e29e17ab4ca3a1c4aba7ce4e66e6 : SUCCESS in 14m 15s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5ee1aec0ec574ebdbc0289c565430042 : FAILURE in 5m 53s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7c886a0694d84fa3a8a17a371881c34b : FAILURE in 7m 28s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f79418233cf34807a4f19315822786f1 : FAILURE in 5m 44s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/984f9b9a7eb74bbba298bd76daa28095 : FAILURE in 6m 02s","accounts_in_message":[],"_revision_number":14},{"id":"74487b498374f604386728eed6b3e1fd42aff85f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 22:26:55.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"6e6111fd8cc8c0ecda236dd6590105eee0c3202f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 22:30:01.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"dff25a00f911b587788a06870dcf06e8651af6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 22:37:56.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4420750455d54c049521f23dc6d34e62\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f10222c8110e4d36b2b03ce8786f1ed7 : SUCCESS in 6m 25s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/af1ac095d71747e9b1b83fea61e4e72a : FAILURE in 2m 51s","accounts_in_message":[],"_revision_number":16},{"id":"e054d3c772890c3120bd84ac63b584036bbdfc39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 22:55:49.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"0110e963ce491af4e89c4a037f97e6959a8d3185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 23:07:08.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b72e5ebab82d4d25b8fde81aed1aa015\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/94d41e42ff9e41c9b0081ce1681848f2 : SUCCESS in 6m 10s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1b7f1662c9ff437c9f840aa34235c8ed : FAILURE in 2m 28s","accounts_in_message":[],"_revision_number":17},{"id":"d8de00079967f1a3d2d9851eace119f3f7b69a3e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-25 23:35:54.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"336f6bfc250694993d8bd7a9cf64b710166ecea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 23:49:11.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18154298a774ba2b089caa8bfd04e7f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4ee8b22371ae4b44ade995ee95708cc9 : SUCCESS in 12m 22s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/66b8893eb55e4f8e88b8b0d2129bdc1f : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":18},{"id":"c06243fa7358bf941bfdf85aefd20d8be4769e5d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-26 05:17:48.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"9431ca4c9b667b793ec1b9843ffdab05fe102f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 05:36:03.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e53f44b2205046d6ba0734ae0b7ed21d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4a37f3a2305142a297de8f5840e8eefa : SUCCESS in 13m 32s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/986b53d557f9449e99d036013f7c2025 : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":19},{"id":"e17e10578cb1fbe6709f46961419e76c6dd23806","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-26 06:18:39.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"3a316ec9e2d3cd3af9c0c7bf71b7e16bbf8e6077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 06:29:37.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/804d4b10c2244417bf8aa93a6667d599\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f99c07f717a645d89616d509610d62f0 : SUCCESS in 9m 35s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ce2e571da0c046fc841182f8cebbe2ea : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":19},{"id":"6f35d05cb70cbd5ab45de3b880d91497dd572e53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-11-26 07:08:59.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"dce0e3179f75cfe9674ff97f06e626c974c3ffb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 07:42:43.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d95fb47d9e94b1f92ed96bb4466260e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/905f80dd536b45a1a5752eec590eacaf : SUCCESS in 28m 38s (non-voting)\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f0ddac3af72a4e42be2a7b3dc6bc8f7e : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":20},{"id":"093509b7c7fc175d5e1240807a3ca0466e87b1ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-30 15:46:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"3847296c0b5aae74714ec72d2233c151c613af03","revisions":{"2c37ea8370cbc1d93f4c2d77cc82cd7b55b95565":{"kind":"REWORK","_number":1,"created":"2025-11-10 17:35:47.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/1"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"subject":"Use uv instead of pip while building images","message":"Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2c37ea8370cbc1d93f4c2d77cc82cd7b55b95565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2c37ea8370cbc1d93f4c2d77cc82cd7b55b95565"}]},"branch":"refs/heads/master"},"de57895fed9d3792e08a47e1feb7e45de4650eb1":{"kind":"REWORK","_number":2,"created":"2025-11-10 17:56:29.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/2"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:56:23.000000000","tz":-360},"subject":"Use uv instead of pip while building images","message":"Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/de57895fed9d3792e08a47e1feb7e45de4650eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/de57895fed9d3792e08a47e1feb7e45de4650eb1"}]},"branch":"refs/heads/master"},"562bb425cb964950c923f305f9491855b6fcb2eb":{"kind":"REWORK","_number":3,"created":"2025-11-10 20:26:57.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/3"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 20:26:39.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/562bb425cb964950c923f305f9491855b6fcb2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/562bb425cb964950c923f305f9491855b6fcb2eb"}]},"branch":"refs/heads/master"},"6353a96e2cf93505e58b7c764b641016d510eadf":{"kind":"REWORK","_number":4,"created":"2025-11-10 21:13:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/4"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 21:13:35.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6353a96e2cf93505e58b7c764b641016d510eadf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6353a96e2cf93505e58b7c764b641016d510eadf"}]},"branch":"refs/heads/master"},"1891ff94b05d0121a17da9f658bde08103a6c303":{"kind":"REWORK","_number":5,"created":"2025-11-10 22:15:01.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/5"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 22:14:41.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1891ff94b05d0121a17da9f658bde08103a6c303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1891ff94b05d0121a17da9f658bde08103a6c303"}]},"branch":"refs/heads/master"},"efb54e2ebdf8d372852b8b33805d8a92ea4646c4":{"kind":"REWORK","_number":6,"created":"2025-11-10 23:26:27.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/6"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 23:26:19.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efb54e2ebdf8d372852b8b33805d8a92ea4646c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efb54e2ebdf8d372852b8b33805d8a92ea4646c4"}]},"branch":"refs/heads/master"},"37a8955e9c089895af9905ebb707b36126724e58":{"kind":"REWORK","_number":7,"created":"2025-11-11 03:05:23.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/7"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-11 03:05:17.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/37a8955e9c089895af9905ebb707b36126724e58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/37a8955e9c089895af9905ebb707b36126724e58"}]},"branch":"refs/heads/master"},"4b710e13f77b438ef7efbb3d3bc53dbc261271ca":{"kind":"REWORK","_number":8,"created":"2025-11-18 23:26:24.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/8"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-18 23:26:18.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4b710e13f77b438ef7efbb3d3bc53dbc261271ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4b710e13f77b438ef7efbb3d3bc53dbc261271ca"}]},"branch":"refs/heads/master"},"3eabb8a94e76a63f2498c7db4c0ec36bea456c35":{"kind":"REWORK","_number":9,"created":"2025-11-19 00:53:49.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/9"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-19 00:53:43.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3eabb8a94e76a63f2498c7db4c0ec36bea456c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3eabb8a94e76a63f2498c7db4c0ec36bea456c35"}]},"branch":"refs/heads/master"},"0f0690f29ddcf987ca68b4eddea8edc66ee09222":{"kind":"REWORK","_number":10,"created":"2025-11-25 04:49:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/10"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 04:49:51.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0f0690f29ddcf987ca68b4eddea8edc66ee09222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0f0690f29ddcf987ca68b4eddea8edc66ee09222"}]},"branch":"refs/heads/master"},"c43d3c798a49ddf049b65fbde361abe8cf915bb2":{"kind":"REWORK","_number":11,"created":"2025-11-25 06:34:32.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/11"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 06:34:27.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c43d3c798a49ddf049b65fbde361abe8cf915bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c43d3c798a49ddf049b65fbde361abe8cf915bb2"}]},"branch":"refs/heads/master"},"389145097f1d7e3d3982fe9ad7703f90e493eda4":{"kind":"REWORK","_number":12,"created":"2025-11-25 16:31:58.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/12"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 16:27:15.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/389145097f1d7e3d3982fe9ad7703f90e493eda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/389145097f1d7e3d3982fe9ad7703f90e493eda4"}]},"branch":"refs/heads/master"},"c5b416b92ba59dab73f7150e8cb84f37ed47c18c":{"kind":"REWORK","_number":13,"created":"2025-11-25 19:45:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/13"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 19:45:13.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c5b416b92ba59dab73f7150e8cb84f37ed47c18c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c5b416b92ba59dab73f7150e8cb84f37ed47c18c"}]},"branch":"refs/heads/master"},"3d3e5777cb809e8b140e6a8a9aa48c4880e90462":{"kind":"REWORK","_number":14,"created":"2025-11-25 21:36:15.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/14"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 21:36:09.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3d3e5777cb809e8b140e6a8a9aa48c4880e90462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3d3e5777cb809e8b140e6a8a9aa48c4880e90462"}]},"branch":"refs/heads/master"},"7f89d1f90130d5acd279527232d3845b40220720":{"kind":"REWORK","_number":15,"created":"2025-11-25 22:26:55.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/15"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 22:26:49.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7f89d1f90130d5acd279527232d3845b40220720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7f89d1f90130d5acd279527232d3845b40220720"}]},"branch":"refs/heads/master"},"184958bc37fb777d1cf23489fd431611abdf453b":{"kind":"REWORK","_number":16,"created":"2025-11-25 22:30:01.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/16"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 22:29:56.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/184958bc37fb777d1cf23489fd431611abdf453b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/184958bc37fb777d1cf23489fd431611abdf453b"}]},"branch":"refs/heads/master"},"ab7e39955d9f2f6624188d1c93f2e606a4b0cae3":{"kind":"REWORK","_number":17,"created":"2025-11-25 22:55:49.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/17"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 22:55:43.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ab7e39955d9f2f6624188d1c93f2e606a4b0cae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ab7e39955d9f2f6624188d1c93f2e606a4b0cae3"}]},"branch":"refs/heads/master"},"17b897fc16cba06709728cdff634f689d1047c38":{"kind":"REWORK","_number":18,"created":"2025-11-25 23:35:54.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/18"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-25 23:35:48.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/17b897fc16cba06709728cdff634f689d1047c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/17b897fc16cba06709728cdff634f689d1047c38"}]},"branch":"refs/heads/master"},"d02f4d6e756b297b1a53d8eb9338002433e80a33":{"kind":"REWORK","_number":19,"created":"2025-11-26 05:17:48.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/19"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-26 05:17:40.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d02f4d6e756b297b1a53d8eb9338002433e80a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d02f4d6e756b297b1a53d8eb9338002433e80a33"}]},"branch":"refs/heads/master"},"3847296c0b5aae74714ec72d2233c151c613af03":{"kind":"REWORK","_number":20,"created":"2025-11-26 07:08:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/22/966622/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/22/966622/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/22/966622/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/22/966622/20"}}},"commit":{"parents":[{"commit":"8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd","subject":"Rename skyline project specific dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ae72fda7bffaf65a1c9ca3b3c79269721e1d6dd"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-10 17:35:20.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-11-26 07:08:54.000000000","tz":-360},"subject":"[WIP] Use uv instead of pip while building images","message":"[WIP] Use uv instead of pip while building images\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Iff223e795ed727551fbc496559bf74e44f793b33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3847296c0b5aae74714ec72d2233c151c613af03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3847296c0b5aae74714ec72d2233c151c613af03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
