)]}'
{"id":"openstack%2Floci~970466","triplet_id":"openstack%2Floci~master~Ic4ac79093abe3709f4ce8cbc4ab322ad3753bb89","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-12-12 18:28:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4ac79093abe3709f4ce8cbc4ab322ad3753bb89","subject":"Remove platform:suse from the bindep.txt file","status":"MERGED","created":"2025-12-10 18:26:05.000000000","updated":"2025-12-12 18:36:50.000000000","submitted":"2025-12-12 18:28:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970466","meta_rev_id":"d5f067e8772769dbec465ac8d24c039352266ca7","_number":970466,"virtual_id_number":970466,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 18:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2025-12-12 16:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2025-12-12 03:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2025-12-12 16:50:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-10 20:24:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-12 03:25:22.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-12-12 16:50:31.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"260cf26bcf4fbbf14036564c83b124af82ae95d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-10 18:26:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c64bdf171c2161a196bff409adff9ed25d819718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 20:24:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44eb825fd7f04567bf11c9f4c7e94b03\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/32b36d243aa74db290aac23d0f45a1ac : SUCCESS in 1h 57m 58s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/33654b3a601541e09d09542bf886c30d : SUCCESS in 34m 50s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/07d6e2426f2e4ba592be3c5059195209 : SUCCESS in 43m 15s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2cf482d8883045bf8de8ada7af7f96c1 : SUCCESS in 28m 07s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ea0780aa554e43878d408bfa15e41f6a : SUCCESS in 36m 43s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/065e4fcb67964c7496d14d14ceb784e5 : SUCCESS in 1h 14m 05s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f53c94114267412882a539bbebd5f71d : SUCCESS in 28m 27s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f6bd1d4421a248e99e6b4664ff3fa0e1 : SUCCESS in 38m 40s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d19f429db95d490597648f3090465510 : SUCCESS in 1h 02m 32s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/73996fe54ab34f6ba8eb4f61b4092ecb : SUCCESS in 44m 04s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d3e235a1ea13414999253814d0ba5ca6 : SUCCESS in 41m 34s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/af4013c73414465aa627fb66c64e3e5d : SUCCESS in 50m 13s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/beb74b6c87af4e4f88a9f7d5e427e2e0 : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fcdc005edc6db10fcbec987e12819e193018aff7","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-12-12 03:25:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2724389d25872d9975849d1181dc04b34d3d1fec","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2025-12-12 16:50:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c41cb92c26e285bf01f850b9c3fa5504cc8dbd12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:50:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44615a2785f4b16a86e3cd9f223bed41d195b9cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:28:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f445cce0d357444186a5a1228171e2b8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6e0691d192ef4a3a832a01a095c8cb83 : SUCCESS in 1h 37m 24s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c7208935cce64542bf046e85cb46f326 : SUCCESS in 37m 51s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/41f24992cc294b449a023606bfb4c3a5 : SUCCESS in 42m 09s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ab3124ef08fd4ff28caf461f27804f3b : SUCCESS in 1h 29m 18s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/733ac92762344b2fa8b955b847b428e2 : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":1},{"id":"93a7a3c36c8a16ca835937cbc4e58a19103642e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:28:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d5f067e8772769dbec465ac8d24c039352266ca7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:36:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/077520934b4d436c839c05f7b77fccac\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/af6c63ad6908411c9fcc6a972638b4d8 : SUCCESS in 6m 29s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/811a0689c47e4949be4bf10a410f4a72 : SUCCESS in 8m 14s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2b0f93481d64482193d8c1bc11ad84c6 : SUCCESS in 7m 25s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b13c26dac8d746caba57683bff4fb606 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1b127eee319c26ad96d8ed50e9eba825cc36c59","revisions":{"c1b127eee319c26ad96d8ed50e9eba825cc36c59":{"kind":"REWORK","_number":1,"created":"2025-12-10 18:26:05.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/66/970466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/66/970466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/66/970466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/66/970466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/66/970466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/66/970466/1"}}},"commit":{"parents":[{"commit":"3a96a9b6729cdf6f45c33b8e49b5d7b8e398ba24","subject":"Add Sqlalchemy to openstack-client image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3a96a9b6729cdf6f45c33b8e49b5d7b8e398ba24"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-10 18:24:00.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-10 18:24:00.000000000","tz":-360},"subject":"Remove platform:suse from the bindep.txt file","message":"Remove platform:suse from the bindep.txt file\n\nA while ago we decided to reduce the code base\nand to stay ubuntu centric.\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Ic4ac79093abe3709f4ce8cbc4ab322ad3753bb89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1b127eee319c26ad96d8ed50e9eba825cc36c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c1b127eee319c26ad96d8ed50e9eba825cc36c59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
