)]}'
{"id":"openstack%2Floci~970899","triplet_id":"openstack%2Floci~master~I4e4dee8b172623f5be31a4b3aa5fc8a58d25999e","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-12-13 00:01:03.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I4e4dee8b172623f5be31a4b3aa5fc8a58d25999e","subject":"Add curl and jq packages to the openstack-client image","status":"MERGED","created":"2025-12-12 19:56:42.000000000","updated":"2025-12-13 01:39:38.000000000","submitted":"2025-12-13 01:31:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970899","meta_rev_id":"187406c95f21595976a0467da9eab13299308e8b","_number":970899,"virtual_id_number":970899,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-13 01:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2025-12-12 20:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2025-12-13 00:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 20:35:09.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2025-12-12 23:05:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"babd2ede17e05f48c31defe916971431de32d0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-12 19:56:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3be572cc4ff5604946c1682e327e5afd284d48cf","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2025-12-12 20:35:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1d2c5919be96d1b7eefe44df093253df4f4a15e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:05:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/763e2775a4f941c0b80d3c3fd8302739\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/749cb851cff742749baa74979618fa15 : SUCCESS in 2h 38m 22s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fdd65032d93d4df48b37626a6210b0c9 : SUCCESS in 30m 55s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/89db30b49f4d44ce82bdfe7373f258dd : SUCCESS in 1h 20m 09s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cd42ef42d4eb4cc59f30d6bc5763a703 : SUCCESS in 42m 08s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3a7e64d6f80f4b66accb5918cca0b5a2 : SUCCESS in 1h 11m 56s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/620c265520ac4e3095e606c584b90d5e : SUCCESS in 1h 11m 34s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7d75d1eaa8b449579e1d277e9faf055b : SUCCESS in 35m 17s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b674ee8923064ffc8d4af8cb33c05f30 : SUCCESS in 36m 56s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fba6115609244c90981a97f5f80d2c2e : SUCCESS in 1h 04m 12s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ec142e22b48a486883496a42b9c31bb0 : SUCCESS in 44m 29s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/291f3aae4f6f44758c8e51f99d410653 : SUCCESS in 1h 08m 08s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/01b4b3eb18364abbab935d305fdd1b90 : SUCCESS in 1h 00m 32s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9f07b974c3024eb084e91da6d0329c53 : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c40f4e7de0928018a87ee1dd49484044f1aa8b20","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-13 00:01:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02d7d92463b7efab7f6e5d4bbc6bd5216f005dad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 00:01:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c9f19d693dfa2b832205135d27ab90aee7b3f49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 01:31:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed167fbd85d4c8c97dd04e417277ead\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d4bd71c7fc624a8a91d913edb8c3a623 : SUCCESS in 1h 25m 53s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/eec0f2600ff942068a2bd332877aba87 : SUCCESS in 1h 05m 48s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d8f4b71973644d8b85a12b4363715225 : SUCCESS in 1h 14m 47s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5dfd92bd051a48dc861c0e946ec52309 : SUCCESS in 58m 27s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1e177d5895e24de18848b581698ca241 : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"603924d9766ede445f3ba27cf5bc412a475f84df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 01:31:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"187406c95f21595976a0467da9eab13299308e8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 01:39:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4565cd5333764a548a02ddd1f27013e5\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/890872a7e3bc40a694430e7600a2cb6d : SUCCESS in 6m 28s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0dd1c025c6284374b8c435456315994e : SUCCESS in 7m 11s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f92c48ad0bdb4209bd853dcc04d26d0b : SUCCESS in 7m 29s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/63b501eb7fe640789d9fbb8f90916686 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b8abf47870fca671a773e1b6d6819b11c1dc934","revisions":{"1b8abf47870fca671a773e1b6d6819b11c1dc934":{"kind":"REWORK","_number":1,"created":"2025-12-12 19:56:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/99/970899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/99/970899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/970899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/970899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/970899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/99/970899/1"}}},"commit":{"parents":[{"commit":"5e7e2faaa0cd1575993ded171db3d01facf91bd1","subject":"Refactor build scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e7e2faaa0cd1575993ded171db3d01facf91bd1"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-12 19:55:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-12 19:55:49.000000000","tz":-360},"subject":"Add curl and jq packages to the openstack-client image","message":"Add curl and jq packages to the openstack-client image\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: I4e4dee8b172623f5be31a4b3aa5fc8a58d25999e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1b8abf47870fca671a773e1b6d6819b11c1dc934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1b8abf47870fca671a773e1b6d6819b11c1dc934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
