)]}'
{"id":"openstack%2Floci~971953","triplet_id":"openstack%2Floci~master~I2629d39383aa9b75fc8247169bcd3ae5e1284709","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-12-30 16:32:35.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I2629d39383aa9b75fc8247169bcd3ae5e1284709","subject":"[WIP] Add arm64 architecture","status":"ABANDONED","created":"2025-12-30 06:41:20.000000000","updated":"2026-02-02 02:53:02.000000000","total_comment_count":6,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"0a5204bf933398562cb7710bdcf686d5d3b76f90","_number":971953,"virtual_id_number":971953,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-26 18:18:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-12-30 07:12:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b51e39921f28344ee8a6cb49da2587212d6825b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-30 06:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25a6c7236e8680d5f8cfa1eacb023b38a1c88e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 07:12:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fb7c775c9a14cf6b6e1679318fdf6de\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/5995993c9f584a72b7ba651e9a145ec9 : SUCCESS in 24m 11s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a34a77652cf9451782fabd3d8be057af : FAILURE in 7m 34s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0df0abe7b097494d96fd1788b06c3d5d : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8a2fc2e6939d42ce9e92086f08df0a25 : SKIPPED Skipped due to failed job loci-build-2024-2-ubuntu_jammy\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/77018833b3c24bbdafdaeaaf05ec5250 : FAILURE in 8m 53s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/64ecb9f9734a4a6fa313179409afdd6a : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6ba00aa492714209b46287a9c201800b : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_jammy\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e5a9cb06a8c84d2581c4d324eec715f4 : FAILURE in 15m 54s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/bc5db8b8b7cc4ce59575f65550f46beb : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d306b26126144f0fbbb446c4545946a7 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a47db8db5d4a4f8f9b8d19a5afe485c9 : FAILURE in 13m 34s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4eef6873e4be4312bb7aeaed7626fc3d : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d7ccd7c8db1a43829cb9720ce8a22c9a : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6df268b429f54f09867545e72a133eee : FAILURE in 14m 18s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/b5a6ebb80661464eaef8f611582eebe2 : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble_uv\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/aec0ddb88306400098ffaf451068d05a : SKIPPED Skipped due to failed job loci-build-2025-1-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/efd005974134466c8a9c68517be7d12c : FAILURE in 16m 29s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/53a1a399532c4f7da6747fe23d80a2ea : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/411604eb07b14c1f878665a226c6e9b2 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":1},{"id":"7f73f11fed5ff4639379f113b106b9bbdb4e3ad3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-30 16:32:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d76be4ce3f05f065d8188080660fe3037ba2792e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 16:33:15.000000000","message":"Patch Set 2: Verified-1\n\nUnable to freeze job graph: Job loci-build-2025-2-ubuntu_noble_uv depends on loci-buildset-registry which was not run.","accounts_in_message":[],"_revision_number":2},{"id":"df11c5657ff736edc46cbd964bdfaa7f51f760c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-30 17:04:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bdfeeee86150ea5a78dc429c1e745ae9fd7c95b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 17:05:11.000000000","message":"Patch Set 3: Verified-1\n\nUnable to freeze job graph: Job loci-build-2025-2-ubuntu_noble_uv depends on loci-buildset-registry which was not run.","accounts_in_message":[],"_revision_number":3},{"id":"ee0f49559ef6984172bb1ccc50086a85aa2a37de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-30 17:13:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b069887d9644c7dfa863b19f3aaa3852cb277d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 17:24:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a42ffd6907940edb79f9290790aa098\n\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/bdc2209bdcf34b3d9855843a656198a7 : FAILURE in 2m 23s","accounts_in_message":[],"_revision_number":4},{"id":"0891d4f43b35c2d0d0d67badf8641a750b6ee876","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-30 17:48:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"582a5263bba74f014385dd89a660e5bc6ff1db6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-30 19:55:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb1f396e0db34616bf1147b35c1b5245\n\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e1fea3cdda874295902d7afd09d1bc01 : FAILURE in 2h 05m 36s","accounts_in_message":[],"_revision_number":5},{"id":"c6673ec75060e5c387f4d64cb8fef202ffcb39da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 05:02:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"69bac400b325a8dd86f178ebf92e2370def12152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 05:11:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcc2ddbe8cac4049b6acc26ce5d82e3c\n\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2bf2d55006834ffa8bc84947254a18be : FAILURE in 8m 16s","accounts_in_message":[],"_revision_number":6},{"id":"e81471d8fa4029aecc278d1fdcda8b88db689b29","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 07:12:06.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8ded9963336745dc27f9432756832576b8ab6231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 07:30:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d9a8350de4449428039ea1a3e8daa86\n\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9337c5d125584ebcb3835312eb94481e : FAILURE in 13m 03s","accounts_in_message":[],"_revision_number":7},{"id":"124e2deb259415c9fd4b7703a15e4b40bdcc49ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 14:02:32.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bf28d03403322ba442e8e9eb0655c3f8f158cc98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 14:15:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/560486098afd4ee1b09f04d4e28883c2\n\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6e40f14a76bf4ebb9f8eabb42ac1a064 : FAILURE in 10m 30s","accounts_in_message":[],"_revision_number":8},{"id":"d6be90818dd60155d3cde21d8e375101023dabb0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 15:09:12.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"baa4651f2ef4cb3330e22c9e1747e1b66ceb1795","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 15:48:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"08687817b1882ac81cc2c1b1976a239089238217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 15:59:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/054945fc2df143b681506ee228bd018e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b0901d7fffe245caac9fdc7021cae9d4 : SUCCESS in 6m 30s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/06ec30854e5742c4b7bf83ac4d6b3dc1 : FAILURE in 2m 51s","accounts_in_message":[],"_revision_number":10},{"id":"c033f8c02aa6cc3d14d06b060ff22aaa1c385da7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 16:15:42.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"375aa8c69a1460e598fb21129dbde19fd8e55d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 18:03:16.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f79485e8f9447388b97a6af819fbed4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/98cd60a211ea4ba7885fad9f9d07df4a : SUCCESS in 1h 46m 53s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3024b546fb1d4e57a75ad656bc0bf73d : FAILURE in 1h 39m 11s","accounts_in_message":[],"_revision_number":11},{"id":"663c22323b0d9c1c08f98ce550eb645a701edb27","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-12-31 20:12:06.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"7036a41624648b08d4dddc0bd9384c385d0a6e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 23:24:47.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b11732cb52ff4a929b059ab353af25a7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f7a1e1d2e3394e2dbd1bdc045d85c158 : SUCCESS in 3h 07m 47s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ea3c4b4fd8b04c6fa5c18342294e85ef : TIMED_OUT in 3h 00m 43s","accounts_in_message":[],"_revision_number":12},{"id":"da8f86185853bda0a021839f8a6592f41e05a871","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-01 00:42:40.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"d66a1750619fcc29631746657b07fd8d7c208ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-01 02:20:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5268ebad4d484e3d9f799ed44e6ed6f7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9320ed94e4a74baea3c10569dd23c52c : SUCCESS in 1h 37m 02s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9bb86258ca554614976ea6db1b796ec3 : FAILURE in 1h 28m 55s","accounts_in_message":[],"_revision_number":13},{"id":"ee6d7cc9d7f98b6ae32d00d536ac53b7be821ea2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-01 03:56:47.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"5a2cea0823b1a09ab951ce1b31019c4b3efed9c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-01 06:27:08.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d162c444da4444c993ebb2a405c1938\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/722f2349edd048838340bd4bfa5ad49e : SUCCESS in 2h 29m 29s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3d18c083fca44f4e8c588efe1a49a2dd : FAILURE in 2h 27m 02s","accounts_in_message":[],"_revision_number":14},{"id":"fe32d8b662d763805435aaeb05721fd6f89d88de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-01 19:55:42.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"4658323c7324bbe81f9e32954826676cd32f047c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-01 21:23:47.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/def71d0747db4e6fa943e25ee2f3ae17\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4ec9827767d04e9385d9fe2f8ab4ca5d : SUCCESS in 1h 23m 31s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/14f56fc56abb469382920b6516dbbc1c : FAILURE in 9m 29s\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/8dddecda035e4684ba9442a1f7d27a1b : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":15},{"id":"8c23fe086f94b246deab1917e4c64546dd933e80","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-01 21:34:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cbef6a62af0cf0569d4d0b5d32cdea6053242546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-01 22:28:59.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e360699731f4fd0a7d5310ad59b414f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/47fd1879013148bfacd774d7169e3786 : SUCCESS in 46m 44s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/96347ec23e474408b3c1d35046d97691 : SUCCESS in 38m 16s\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/ec4b0c5e7e884361955bf197219397fe : FAILURE in 8m 31s","accounts_in_message":[],"_revision_number":15},{"id":"ca0616c3a80f03eb291c9fcaf0e8a678530801e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-04 21:14:21.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"171ae547306f3ff1a1d746531a0d32d7d1ea1763","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-04 21:21:35.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"43c601e5bd91f4f61147444a75bb454b4f52f70d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-04 21:51:45.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01c024bb8242403aab7e76258db1b8b3\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/219b3fdc12d0488484c214aa1e9af311 : SUCCESS in 29m 17s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/9ee3f89ada564d0bba5854f93ade36e0 : SUCCESS in 17m 38s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/1e2dbe4e52f849209dd43e18407c8a37 : SUCCESS in 10m 19s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/7dc039a1d2144064899f12aa25730f37 : FAILURE in 1m 40s","accounts_in_message":[],"_revision_number":17},{"id":"f0e6eb560b2eb8e7dc848c6324c81a1ac9d0cfd8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-05 16:17:10.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"1cd383b1826cfedbae61c4019b61d8d2d55760ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:27:22.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c5b3a9507084f2091105f9cc87f9283\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ac0032584ec74ee9b9013557a4de59eb : SUCCESS in 37m 57s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/9f0ce23b967e4a8d98507f7b77341373 : SUCCESS in 14m 05s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/f52d59d25f824d0494a4573bc2051867 : SUCCESS in 11m 08s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/105dc33629324356bed5ad76ed64a3fb : FAILURE in 1m 48s","accounts_in_message":[],"_revision_number":18},{"id":"0033900c08136ebbe56cf2050494893c6b786b9f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-05 20:25:32.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"1e8cabcf39202ba5c67249e20b6f7abe5eff2284","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-05 21:01:43.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"c54dd404e075de945d1e5f2eb1ba4fa02fbe9d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 21:57:58.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d3a9985afe1418ea004b9642c0d4734\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/8417075b701547718a62fc4f3a7d930d : SUCCESS in 35m 15s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/5403aff542ed44f5b1b98d40997c9cbc : SUCCESS in 17m 04s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/edb8e4cd85bd48a79a057b6c490e8872 : SUCCESS in 8m 04s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/1c04cb9935ac4d06bf3d3f3a11ae7b17 : FAILURE in 1m 52s","accounts_in_message":[],"_revision_number":20},{"id":"1bd7cb568ee0bf5b79f1a59149a1c4a2a95340af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-05 22:14:23.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"fe70fa3285fa90e3da0dc6d65a5cfce7c706570a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 22:55:03.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db5cd45fa89a4b36a110cc68f67ece88\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/73d9ce99e538487396398e6509662e4d : SUCCESS in 29m 47s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/6ad0d52d042c4403a296126fcbe951fd : SUCCESS in 13m 07s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/1f6ab3ab51214ebb93b4a6173fdbfc56 : SUCCESS in 8m 22s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/095c6d3c23814f8c8dbd06365f677cc2 : FAILURE in 2m 11s","accounts_in_message":[],"_revision_number":21},{"id":"95fbd2afd4a96e6e1aa03793be3e08837f240171","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-05 23:21:50.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"5a80b821512190bd26ef692445e846a3838e339b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 23:51:46.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dc8e06d9712469fb8b40f02f94511e8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/65fa13688ba44df98290ef7cd3a433e5 : SUCCESS in 24m 08s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/98425954b69d4440856c63936a3c0741 : SUCCESS in 17m 12s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/4fcfcef0806747448148bc77116e7e87 : SUCCESS in 7m 54s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/870d67fb1687458583b819e51736286d : FAILURE in 1m 10s","accounts_in_message":[],"_revision_number":22},{"id":"2f8f15cf6ecc07bf566fb0bf72d0b77a80360598","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-06 02:06:57.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"809af78fe3395984b0916c4590f5e78f59d51dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 03:55:08.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/363a89028d4a432785656f092f0bb1fa\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ec24eb26bcfd49c7a506eb5a81eec885 : SUCCESS in 1h 38m 25s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/2d6550dc7b394511b70d9d57ef5a31d3 : SUCCESS in 13m 29s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/886c96a6cb3749cbb47fbcc3588c41e8 : SUCCESS in 11m 26s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/daa4b0cff8e748cba2c162990a67ad60 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":23},{"id":"1f8ed6ccf9aebe805c4f84ab6643636a2aa7257d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-06 04:40:01.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"dcea1ba125f3e4373c8ff4f0fcbf92701842eb68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 05:17:25.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/223f0b9c301c49f4ba678f958febaffc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9c49cf12ff4049f3b950a8c03c2c753c : SUCCESS in 28m 00s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/49d80f82f5014efcb243ad4f0ff344c9 : SUCCESS in 15m 57s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/09ec131eb2414729aa79a2ff80fa6c7b : SUCCESS in 13m 30s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/faf84ce6fa4f47c3b5dd1b0c0129b4af : FAILURE in 2m 38s","accounts_in_message":[],"_revision_number":24},{"id":"4327315e4dcf9bd5613623c21fbff5157c05a0c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-06 06:21:17.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"a4d9ab0edbb92041cb192ac7ec19ac289436918c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 06:59:08.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2865731834c47cd9b4453b02f494fe2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/aa94c1af656a4876a782a2af7f348d15 : SUCCESS in 29m 41s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/84d9d36dac0c4514a32aac1e018eb834 : SUCCESS in 16m 40s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/14c223d89f0847d29a3090086fc896f5 : SUCCESS in 5m 55s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/c45f6b7746734a459cee0d3e26d6e6c6 : FAILURE in 4m 11s\n- loci-build-2025-2-ubuntu_noble_uv_print https://zuul.opendev.org/t/openstack/build/060d2763abab4acb9fc86fe318756f9e : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_merge","accounts_in_message":[],"_revision_number":25},{"id":"27ca95d3ff326e456fd3a4abb20d6eaeb8719063","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-06 19:40:09.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"aedc9c4481d83461575e96e9ac353831faa8e6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 20:17:43.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5607942d26434181aa4032a237e64e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/5c7d65f4f7c64db8ab1af38ef3f1b1be : SUCCESS in 36m 38s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/fb14fb32e6f84a80ad5a56bd6ddc89ed : SUCCESS in 11m 49s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/306baca18e01400aaa6c71daf70fb4be : SUCCESS in 10m 23s\n- loci-build-2025-2-ubuntu_noble_uv_merge https://zuul.opendev.org/t/openstack/build/dde09ac8a6a3499fbda4c58c3e236e33 : SUCCESS in 3m 19s\n- loci-build-2025-2-ubuntu_noble_uv_print https://zuul.opendev.org/t/openstack/build/d5648dca0c3244809c2d1e791232f04d : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":25},{"id":"19ced2dd6932db7e13e4b16ff2d0a5ff354e8cbc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-07 21:49:09.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"f3800d5613b6c717587f05bc684c326a08b5fcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 22:22:47.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f11dd788f7545dba64a9ccde308cabf\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/681b2b1f8e9249ed988375eafc32bf0b : SUCCESS in 26m 01s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/ae3b8057da82475f86031e822456a825 : SUCCESS in 13m 09s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/fddbd32ff562403db20f15c86a51fdba : SUCCESS in 10m 46s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/ad311ade477a4a8983e560fb05270e97 : FAILURE in 2m 37s","accounts_in_message":[],"_revision_number":26},{"id":"23dac53e509a85ad18b4b4dadc4c3eb69376bcaa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-07 22:57:33.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"72147eb38b3beb1d78c1891edcf8888f88201dcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 23:33:20.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38f0334088a448818d93a9d2f3549707\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/99598815b93c43beb74278f74143a53b : SUCCESS in 33m 55s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/85789b5f2d2842758ba99e285d8c929b : SUCCESS in 11m 50s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/169463f81b42402996767b706bcf611c : SUCCESS in 11m 55s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/9aee5ae96e87418c93587894ccc96fc3 : FAILURE in 4m 07s","accounts_in_message":[],"_revision_number":27},{"id":"580a359a53580ff0000f97c9e812b43f71c5aae3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 00:22:10.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"746bbffbb6d04b36232d98fcae15ce49703897e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 00:50:21.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0712d02b42f4e8899c8ffedaf611466\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/49d421a20e9f45888aa6391a695bf110 : SUCCESS in 20m 54s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/d9d8a9421e0d447c8cbc7d048469ad67 : FAILURE in 10m 09s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/a8f99b8254cc4b689741a75cb74e2fd9 : SUCCESS in 10m 33s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/43b25f286e614e60b431d8facaec91cc : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64","accounts_in_message":[],"_revision_number":28},{"id":"77c2b0602d2684c6e7abea66ef2831250c9ce201","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 01:49:47.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"c09a2d971fc9a3844a242e63b3da1325800f7218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 03:18:29.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1de187636c0145778eae31f4c92b7ba3\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0fc48c86c98c48cb83b547a13b3f273e : SUCCESS in 1h 28m 16s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/feeadae7653a48e0bc09bf6a22c96e10 : SUCCESS in 13m 00s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/5e8f53319999457b9b376183c54ea4cc : SUCCESS in 8m 28s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/362642a836bd48739b9187ba09309bb5 : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":28},{"id":"eee93f8c008a9f7c2f3ced0b3c2add82fbd65715","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 03:56:41.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"8fa661b107f65fffda6ce9b0f856a70ab9c46177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 04:23:00.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7b026a8f8524824bf7703cf050ee609\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ad0d0e526f2b4f3c965468f025e6be84 : SUCCESS in 24m 58s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/1b0bc9cf558445f48cfb44163405508b : SUCCESS in 11m 26s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/13e4cb6e0f0b437598347724fdcec719 : SUCCESS in 8m 06s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/60d13cfae7ab4acc8939104f823ff232 : FAILURE in 1m 53s","accounts_in_message":[],"_revision_number":29},{"id":"d0db12915e1dddc51c342f8155e0f774718f5a85","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 04:57:13.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"4c8eb1fc064e334ffd1e53dec467de7d0d175d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 05:40:47.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e32ab76eabe04dcaa45dc041c63ed39b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/282c39b7af3e4c71ae7ab58b46df2ce6 : SUCCESS in 37m 12s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/38bd813c3a404194afb3671d77f893d3 : SUCCESS in 16m 28s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/2d134f57bed94a4fb25ec20ac954025e : SUCCESS in 8m 38s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/ccfaf649e064440fb009c78ea2fe032e : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":30},{"id":"c79e81ff6ba2ae0ed99302a8683c395c1dab40a6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 06:54:38.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"1918a5fcc14b9ae45f1008fcfe0c0e676c75e21c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 07:30:31.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae31bb360be544859c68ca5dc8d67bd2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a7d84a764ad7490581cb58c52a2e5484 : SUCCESS in 32m 47s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/b4c6deeb708049089ee57062f63e48cf : SUCCESS in 16m 03s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/6559f4a06f1740a482f2f2b8984c5534 : SUCCESS in 10m 17s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/fd2bd1ae1ace4002bb708261d31d16ab : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":31},{"id":"deca1384df660a679ccc74b3d3dd88265a7de8d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 07:44:15.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"fde4852ad66afce8f12498296aad41e2e564e0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 08:10:47.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92f8f2f996f841d686b506648aafbe4c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/409615aac7854269964ac94cd20377b3 : SUCCESS in 21m 45s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/1576bb6a9ba5491cb0b2a0191c7e66c6 : SUCCESS in 10m 50s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/a53cb5635f214968ae5e0ffa11b808cb : SUCCESS in 8m 07s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/76c6eac4d8fa464599f559b5e689583e : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":32},{"id":"a878fe68754af29fc86021d9e0617557430a5fef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 20:28:26.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"b20491ad21fb6bde74b579c01c3452b516523fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 21:29:07.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f5ba723a8274a60ae537b207ea6328e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f7f82e09c7244148a3999a9cfab8c205 : SUCCESS in 43m 00s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/c104d2fc1c1b4c8ba00879c2e1e8bbd4 : SUCCESS in 11m 59s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/3f414c7c865744a5b12b86bbef2d6ab0 : SUCCESS in 10m 10s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/245c20703c2e429a89eeb495c268b583 : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":33},{"id":"5536e10a6e90a9a98db919c3f4bd876e5b0d9638","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 21:39:27.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"70398a7ae269c54ad3d4bc80f2b7676de82a0abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 22:41:40.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/430bf33d7a16459fa9da2a8fd8a941c8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/252dbb771934416a88cbe62b72229b8d : SUCCESS in 51m 27s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/c26856ab5d7a4dea80cdcb5e3760c8bf : SUCCESS in 12m 15s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/bcbd1993c58c4dc1b758a8ab3bbd1655 : SUCCESS in 11m 23s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/7667228863fd4065a12c134252f57670 : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":34},{"id":"f53bec13fe3fe0a6df5e765fb2d66baf9e813530","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-08 23:51:30.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"f8c251b6e565dbd1a877cfa4ef352680bec52f43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-09 05:19:26.000000000","message":"Uploaded patch set 36: New patch set was added with same tree, parent tree, and commit message as Patch Set 35.","accounts_in_message":[],"_revision_number":36},{"id":"822b62bea2eff2dd592e9fad64af4c371225da22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 05:19:49.000000000","message":"Patch Set 36: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"loci-upload-2024-2-ubuntu_jammy\" job stanza:\n\n  job:\n      name: loci-upload-2024-2-ubuntu_jammy\n      roles:\n        - zuul: openstack/loci\n        - zuul: zuul/zuul-jobs\n      timeout: 10800\n      dependencies:\n        - name: loci-buildset-registry\n      pre-run:\n  ...\n\n  in \"openstack/loci/zuul.d/2024.2-ubuntu_jammy.yaml@master\", line 18","accounts_in_message":[],"_revision_number":36},{"id":"8098c6f33578e6ad9a7539c3fb7f2b9ec84c6427","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-09 16:38:42.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"968cffd9dda8f2f5ab72d52ffea9feb64ef62c86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:39:43.000000000","message":"Patch Set 36:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Decryption failed\n\nThe problem appears in the the \"loci-upload-2024-2-ubuntu_jammy\" job stanza:\n\n  job:\n      name: loci-upload-2024-2-ubuntu_jammy\n      roles:\n        - zuul: openstack/loci\n        - zuul: zuul/zuul-jobs\n      timeout: 10800\n      dependencies:\n        - name: loci-buildset-registry\n      pre-run:\n  ...\n\n  in \"openstack/loci/zuul.d/2024.2-ubuntu_jammy.yaml@master\", line 18","accounts_in_message":[],"_revision_number":36},{"id":"76d5bacf8a9924eb191501d032f85814f550da8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-12 18:01:03.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"1a6c8ff998fcd73982974a9331e8cdaa4d50d2f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 18:43:19.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fcaec1c77084b97ae6bfeaf9b9bfd5d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3a723575212b42f8b0f3065037e31383 : SUCCESS in 27m 46s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/7882b9d4c4c042d6b18077f4d78d2275 : SUCCESS in 13m 45s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/2b126dd988d142638154e95f57067f9b : SUCCESS in 5m 40s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/582bcc1301454d01b5946dc7bb32d1be : FAILURE in 6m 31s","accounts_in_message":[],"_revision_number":37},{"id":"48d4518bce8c1ca3f06d3b6bb2f99c10f58f4ac3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-12 22:04:03.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"7dfb41a6b84792c1d67e08ef97b9abeefff35428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 23:09:48.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47cc599758443b59bbde6211ba02adc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7c4f41e11a1c4a9bb06f1e8ef5f8ed06 : SUCCESS in 1h 00m 24s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/f853933297be458b8ece3c8ffeae95ec : SUCCESS in 20m 01s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/10671f79f1b045059d6ba344d5b81c99 : SUCCESS in 6m 34s\n- loci-build-2025-2-ubuntu_noble_uv_upload https://zuul.opendev.org/t/openstack/build/2727f3054b9643619158ac05d01e6c78 : TIMED_OUT in 30m 48s","accounts_in_message":[],"_revision_number":38},{"id":"36b1725714e990eeed157c062fc5668ec515e20e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-13 20:15:53.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"c116b29c73b5144eae8c9aa2b2fd14dbbe00f827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:16:13.000000000","message":"Patch Set 39: Verified-1\n\nUnable to freeze job graph: Job loci-upload-2025-2-ubuntu_noble_uv_arm64 depends on loci-build-2025-2-ubuntu_noble_uv_amd64 which was not run.","accounts_in_message":[],"_revision_number":39},{"id":"c257d3f8be014da255a02267d3ed3c388e39545d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 04:04:40.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"b6f5173afa6e27190b332cb0ec67c503053f36c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:05:00.000000000","message":"Patch Set 40: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Job loci-compute-kit-2025-2-ubuntu_noble_uv not defined\n\nThe problem appears in the the \"opendev.org/openstack/loci\" project stanza:\n\n  project:\n      vars:\n        helm_version: \"3.18.1\"\n  \n      check:\n        jobs:\n          - loci-buildset-registry\n          # # 2024.2 Ubuntu Jammy\n          # - loci-build-2024-2-ubuntu_jammy\n  ...\n\n  in \"openstack/loci/zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":40},{"id":"65835a1c086fc8e26b2d2577ff8da83e47b3b2fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 05:09:39.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"6cfe514d392ae866ed21931b4dc2e3e0de0c7191","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 05:14:14.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"5dc70759eab41be93b607ca5e02eef07558d7733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 05:41:34.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60f9666f954e4f3fbf822aa18fa1a7db\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/75e72c54dbb14bfebfe92b738554b726 : SUCCESS in 23m 59s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/a693e884caf446a697d63492c7de6b7b : SUCCESS in 17m 45s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/b9ea43cc77004c8d89a65a044c2c38c5 : SUCCESS in 11m 22s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e9c218543e254a8cbbf6cd7d1a09c6aa : FAILURE in 2m 13s","accounts_in_message":[],"_revision_number":42},{"id":"c8414d5bbd649c6400d42d775bd1d898b7484756","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 06:00:20.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"03fed5f33b214c25a2e501394747c7a9ca4b8c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 06:23:37.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2499b9d2d1a47fda321ded77adff030\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/24455d9f5f6040638e9fa83eb56b0e2a : SUCCESS in 22m 29s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/158db35092494bc4bb66057aaf1f34ff : SUCCESS in 16m 59s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/268136e303794631aa84e9ec58ca9ed7 : SUCCESS in 11m 17s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/7fbd30c84be343baa6d3669a77525b52 : FAILURE in 33s","accounts_in_message":[],"_revision_number":43},{"id":"46676e751a9989ac37f5e8b755e61dc30ccf9137","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 19:09:46.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"b066cd2915dfcf21f35836dac4cc10962bb3e056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 19:41:30.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d27c74fcd5743dcb2685b01853037a4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9e160ff4b42a46ddbbe6f2ab76b7736f : SUCCESS in 24m 56s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/16d1a4ea45424441a82a0cc79ba8ebc9 : SUCCESS in 18m 30s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/a7b80e17424d4d4f84a52b7d5358ceaa : SUCCESS in 7m 06s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f6fbef7d55a348f58bb83f2594a277eb : FAILURE in 44s","accounts_in_message":[],"_revision_number":44},{"id":"094234607d37f351153a72bc928c1a7f29249b9e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 21:26:09.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"d87519264bdf83cd9e01a40a8d65b778768f6ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 21:50:01.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71f881a894cb4af99c5d5456995092c7\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1b5dfaed5e2c428b9522ae259603aff0 : SUCCESS in 23m 05s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/be967882d79e40f19847ca514761a722 : SUCCESS in 17m 38s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/506955d2910941d2b83d42dc430ad3a6 : SUCCESS in 8m 37s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/8f0009dce2b34df894b2ebc060f0a040 : FAILURE in 42s","accounts_in_message":[],"_revision_number":45},{"id":"799ff16a530e7878a5688700655174fc5f8db894","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-14 22:27:00.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"d9aa2c7532e47d260520448f0efb300fffbddaf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 22:50:33.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/514217437c034574814e6117237f4cf1\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/80168ffe94b749f4a8fda15b17b08e0a : SUCCESS in 18m 50s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/4be56b8537dd4aa3b53a6b412ef2ad88 : SUCCESS in 13m 30s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/fcc6ebaa195349199db8f574fcc314ac : SUCCESS in 8m 11s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/af05488b46d54d8fb67160cae834786c : FAILURE in 44s","accounts_in_message":[],"_revision_number":46},{"id":"72d471fa71ba7c16b600b3e5a0400785ea8438c1","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 00:12:54.000000000","message":"Patch Set 46:\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"f10e84cc3ed3a3655fa66af55280e7a32f57edfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 00:38:03.000000000","message":"Patch Set 46:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2783b94ad58b432ca64be5e85f00993e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ad0282272a424af3ac87f6f2ae4d3f9d : SUCCESS in 23m 42s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/8570ce13189e48bc9ce963c31f8e3d4b : SUCCESS in 13m 52s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/b893cbba91114f308ce5d164c093999f : SUCCESS in 10m 54s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/104ae2cc5540439db1340ddeda5cb7f7 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":46},{"id":"addcbe78f84cd082e1af8ab67e5da0e7a86bd3f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 02:40:53.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"a1e863d29d7efa62c642ea1fd631ec3aad43869f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 03:55:12.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b7dceae1da541a5bacaa5d08e78acc5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/4eb6fc6f4d834de3baafa8d9b7405dc2 : SUCCESS in 57m 33s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/72ecf7a99540455683469ee00ea0aaac : SUCCESS in 20m 07s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/fbcd03fca8cc4088a85f94455eb4358e : SUCCESS in 13m 07s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e815a7e8a0d84c52b7e8f1fe16aa3a67 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":47},{"id":"ca1b5b9a1d51bbcec8c3b39717c8e49970a9ac3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 04:57:31.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"bafd44aceefb46b0b53014c8c9c7bb4360ed8da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 05:33:04.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/30777fcc797440be98ec2c05b15481c8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bed3a83c23874a3e83c7c98ee1158e0e : SUCCESS in 30m 46s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/bbca4c07abad4e3dba9ded103b51046e : SUCCESS in 16m 31s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/6ca0f165c2214346a6cb58e9710b21e5 : SUCCESS in 9m 19s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3d690b1bc8fe46b69a065631e3260004 : FAILURE in 4m 09s","accounts_in_message":[],"_revision_number":48},{"id":"865add7e611306e46b7536204072e47bccf35ec8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 06:41:27.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"ae7f366d034f6dc630f1c05e5397e8fb72f96703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 07:23:25.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7c9a6b2334542bdbb704a1a09daffa5\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6e950f02e2774acda77e7799894ebac8 : SUCCESS in 36m 39s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/c749d36ce790436fa11b6dd433557764 : SUCCESS in 20m 03s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/7ad6ea6c462843baad0088ffa472d1c5 : SUCCESS in 11m 26s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f06b1b7956104ebd8c5628208c44a70b : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":49},{"id":"1b87b5daa78242f45a8030d9047afcb51d722510","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 20:24:07.000000000","message":"Uploaded patch set 50.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"d150e0c7ca67cd6253edc43c9657cfae3ad0e74f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 20:51:29.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41f95185b95f487e98abdeab8a494825\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/ae3c4d8acd984a98aa44b3db58707cce : SUCCESS in 26m 32s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/76e950269e5147a0bdf8726440a79c07 : SUCCESS in 13m 10s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/7f5e69cd3f684cc582f94a50848fb1ae : SUCCESS in 11m 53s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/5261e7c173f84632ae51ceaa28f04d3c : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":50},{"id":"b6db4bad716230559598fde704273d7297b5c0ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 23:04:14.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"93e33f6a463161a48ba52b269ef114bd26c1642f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-15 23:49:58.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"16bf4f1a8a86f8e0a2bbf7b979df1621356868a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 00:26:52.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca9456eea05428288bd4bb26b6fd173\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/3ade3bcab52640de8489f9b81170819c : SUCCESS in 31m 20s (non-voting)\n- loci-upload-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/3d156cd4da3143489a3175b15118d046 : SUCCESS in 13m 39s\n- loci-upload-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/1ca328414a294e8b8c9d61934ec1a1b4 : SUCCESS in 12m 43s\n- loci-promote-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/014ca682643d4a589ba91baef6689d4e : SUCCESS in 1m 19s\n- loci-promote-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/39f345e07fb0484caa81878a6065f0f6 : SUCCESS in 1m 23s\n- loci-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/1629966f66e44104844e2f54ff65ccd2 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":52},{"id":"ca6ef02a43c105266fedcda260ffaea2bddc0f14","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-22 21:04:58.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":53},{"id":"4576f49f1f47c6fff3415bb0fcd0222385cb36a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 21:05:19.000000000","message":"Patch Set 53: Verified-1\n\nUnable to freeze job graph: Job loci-build-multiarch-2025-2-ubuntu_noble_uv depends on loci-build-2025-2-ubuntu_noble_uv_arm64 which was not run.","accounts_in_message":[],"_revision_number":53},{"id":"304508f9bc556df114239f035197168e1f168983","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-22 21:35:59.000000000","message":"Uploaded patch set 54.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":54},{"id":"6977cb53260a8fb11d5e4440c606ad5cbcf3ff97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 21:48:05.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d4007170421479395ec6f21804fe929\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/920cf23287d648fbb763fd8977112d62 : SUCCESS in 11m 08s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/41d1d0a8f3eb4ead8cedc1d792a4f809 : FAILURE in 5m 36s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/34a0f4516260459a83da8a64ad9cbc01 : FAILURE in 3m 02s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e6a5944c39454b87a8ebcc16f5d16776 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":54},{"id":"e3ef7d73e9a5f69f1cc0429a7230b6c4652d188e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-22 22:09:13.000000000","message":"Uploaded patch set 55.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"7d4ae8ca9a1e43a9f0e81273025f8878556bbb93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 22:20:17.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fa427563d9a4f86bba49c2474a6236e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c3d94338d9a84d4d89dfac03d15ecae3 : SUCCESS in 10m 04s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/648c983ddb1b4273bfe9b70895d9ff4c : FAILURE in 5m 16s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/0ec3b1ec749a4de1b52fa8c6913b7741 : FAILURE in 2m 20s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/4765289d08d8402997708872bb6988fa : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":55},{"id":"e33dc516d2f77ff5049348d8d7d2e21e56435500","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-22 22:27:22.000000000","message":"Uploaded patch set 56.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"188bec6907364cf492e5bb58ea446ab965a79561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 23:09:40.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2ca5cddb95b4162bd91465945ad863e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/2d20ff11326c437ab85c7ec52f2074d7 : SUCCESS in 41m 25s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/d9732e06273f4782933a2bd3f838aae0 : SUCCESS in 24m 22s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/7dba26705f8d4f2b804c252c21daf1cd : SUCCESS in 8m 39s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3c1122e0b1f248a785c31b8e89b4e2b8 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":56},{"id":"431e2043801735f458780f092892b46586a466f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-22 23:23:53.000000000","message":"Uploaded patch set 57.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":57},{"id":"385d3065640075bdc39093decd166366f8602e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 01:27:12.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc709c848a234284949e73ffaa4bfef2\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0027adcff186447fbe69d1ab8dd3171c : SUCCESS in 1h 57m 51s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/cae12121b9bb4bb1813fb4126de6f721 : SUCCESS in 1h 20m 37s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/99f557aecd8f46668de102ffe9b5be85 : SUCCESS in 6m 33s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a98e5aee05e44e39b4da2957fe92fcf7 : FAILURE in 7m 01s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f89a9383c4dd44b68d835b818ab47698 : SKIPPED Skipped due to failed job loci-build-multiarch-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/7599ed3d1d484d309c539d712b87da7f : SKIPPED Skipped due to failed job loci-build-multiarch-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":57},{"id":"573716b2ef731cf13fc97b6634d51cfd00cfb448","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 03:44:21.000000000","message":"Uploaded patch set 58.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":58},{"id":"2d593c2344f7ae2fb16f7d8d77927c1d44c23279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 04:04:34.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4df0a9918a3445e8297e7315bda2122\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/12fcceb9cd1c400aabc58b295c214849 : SUCCESS in 19m 22s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/315a40a7a8514d619f71960eadcc10c6 : FAILURE in 6m 38s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/8e40db51503a453cbde05eeefbb8114b : FAILURE in 2m 00s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/013c38328cba4792850124c4c54943b8 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d5f49e50256a4818b5a064e2943d5358 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/574fe0392ea048dd8fe8075e2f15e68b : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":58},{"id":"a6eb3af8c78bdcdaaa409c65ebd2f2dc420ea322","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 05:18:23.000000000","message":"Uploaded patch set 59.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":59},{"id":"13db3f2d025de8a47928531bb62710f62726131f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 06:28:25.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"2a6da22f78b285401ac65f3056b9635fd2d3d3d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 09:35:35.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c2adf8f09b64713b6f4c1b5750cdb56\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/dfb41983e5b7451e9ec8eda8d51f2d52 : SUCCESS in 3h 04m 55s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/cf794906dd8b4a72b9b9a8f9a9aba8a7 : TIMED_OUT in 3h 00m 54s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/a8e4164a000f456ab69280567a64172d : SUCCESS in 28m 17s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a2d7121463674436a700cb9eabb30466 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/7286ed00ec8c4839b0131931384b174a : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/703d145235c4482c915c2481c03325ee : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64","accounts_in_message":[],"_revision_number":60},{"id":"1ac9e606675ff6c3ddf678e30f4357f0d7945489","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 17:27:14.000000000","message":"Uploaded patch set 61.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":61},{"id":"4a47137263413ef6991091a3647705c3a452cb71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 17:31:42.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"63fd165d80f1418bfa015be61df5b10e104aa91e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 20:44:52.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/407b927d0d3c459b97b8452666f9080d\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1ba97b9b45db4c66aabae5433d52c185 : SUCCESS in 3h 06m 57s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/6aee597f724242f2929146678b1ccd4d : FAILURE in 2h 45m 35s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/2d48d0826d3641eab6d7da54b23472ef : SUCCESS in 1h 01m 57s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/063bab91e2fc436d819ecf2ab09ab557 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/4d941915c6534d51a6099c66c90adfec : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/0984acbb047549a68b2ddab9c7d7b968 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64","accounts_in_message":[],"_revision_number":62},{"id":"d47502a4b6caebdcbbfd9dc5cc771b742225eac2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 21:06:00.000000000","message":"Uploaded patch set 63.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":63},{"id":"e057f70f1d1046753d53f4fe6e799a4a670d711d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 21:58:43.000000000","message":"Uploaded patch set 64.","accounts_in_message":[],"_revision_number":64},{"id":"793a3604b33a73352fcf8b0ca947e2c12230f193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 00:57:43.000000000","message":"Patch Set 64: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00e8f9515623402690fef1922153fe5f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/147e3ee7f936408f8b339ad95b98c701 : SUCCESS in 2h 56m 56s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/a83ab54bb0a34685a5a79dd8b63b6d8d : SUCCESS in 2h 23m 35s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/68d6b6ceb9c4451aa2607bc5cd96a576 : SUCCESS in 1h 02m 13s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a6d484f65a9c4b39b645510a7da2d1da : FAILURE in 7m 06s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/109066b9693d4f1bb9c99241ca3a2ff9 : SKIPPED Skipped due to failed job loci-build-multiarch-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/68c8ea343a9040faaeedbb57dda295d3 : SKIPPED Skipped due to failed job loci-build-multiarch-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":64},{"id":"ff056945a08d81b381bbb9802b64842cd1cfbaaa","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-24 18:11:01.000000000","message":"Patch Set 64:\n\n(1 comment)","accounts_in_message":[],"_revision_number":64},{"id":"5f0c7f649a96daeb107057227f54828144b191e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-24 21:32:17.000000000","message":"Uploaded patch set 65.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":65},{"id":"004c08ea1434972c2aa9c0b001fec13f2304fd1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 21:32:52.000000000","message":"Patch Set 65: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/loci/zuul.d/uv_2025.2-ubuntu_noble.yaml@master\", line 66, column 11","accounts_in_message":[],"_revision_number":65},{"id":"aeedf90ee76b5a711835167417530694a7927b9a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-24 21:42:39.000000000","message":"Uploaded patch set 66.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":66},{"id":"b1cd6388d16e34c3c4f592c79c8f1730ea27dfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 21:52:11.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a5c5bbb974f4f3e9b5dea4d27702bcc\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6a5bcc9bfb3a4d04a8d7e507fbffda6d : SUCCESS in 8m 34s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/dc9a8ca06a674af1836a100154f385fb : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":66},{"id":"d921059d29e47d9f9e5eef9a6d1350149bbc25a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-24 23:45:20.000000000","message":"Uploaded patch set 67.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":67},{"id":"80c5499777a19bede01474fd14cb873d9a4c9a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 23:49:15.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3d1f7e405784667b063ca44ceebe7fa\n\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/662d4954380d4c0181d622cf470180ae : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":67},{"id":"a494f6db9e376092c56e73ca9345412388930870","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 00:28:48.000000000","message":"Uploaded patch set 68.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"727f8bd76b08501b4a51baf5df07affea22231ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:29:08.000000000","message":"Patch Set 68: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  expected dict for dictionary value @ data[\u0027nodeset\u0027]\n\nThe problem appears in the the \"loci-set-current-date\" job stanza:\n\n  job:\n      name: loci-set-current-date\n      nodeset: []\n      run:\n        - playbooks/set-current-date.yaml\n\n  in \"openstack/loci/zuul.d/base.yaml@master\", line 7","accounts_in_message":[],"_revision_number":68},{"id":"e5076e572ee8a5adc1304d8a9c1ffb6b1064e94e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 00:33:02.000000000","message":"Uploaded patch set 69.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":69},{"id":"98dabd53f68dd3e9b68d578f09af2bbf44d39f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:33:21.000000000","message":"Patch Set 69: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  expected dict for dictionary value @ data[\u0027nodeset\u0027]\n\nThe problem appears in the the \"loci-set-current-date\" job stanza:\n\n  job:\n      name: loci-set-current-date\n      nodeset: []\n      run:\n        - playbooks/set_current_date.yaml\n\n  in \"openstack/loci/zuul.d/base.yaml@master\", line 7","accounts_in_message":[],"_revision_number":69},{"id":"7d9833f2259e019d639b1f100bcd667f96b02b1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 00:35:56.000000000","message":"Uploaded patch set 70.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":70},{"id":"97a6c95b48ff009b8ec80547081284bef7c1b2d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 00:36:07.000000000","message":"Uploaded patch set 71: New patch set was added with same tree, parent tree, and commit message as Patch Set 70.","accounts_in_message":[],"_revision_number":71},{"id":"1a3a2d652a9e62014b3c72f457225ed6085f9cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:42:54.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/724da813fa3a4f8782d2cabb864baf17\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/571f5368ff9b4c71a440261203cf377e : SUCCESS in 31s\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/345fd0434cdb493e8938ad485bdbbf5c : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":71},{"id":"4ab6aabb4ca8fc9dfbfc6fbf42f56e05dd4aaa48","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 00:51:39.000000000","message":"Uploaded patch set 72.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":72},{"id":"9d11893dadbd161e9e02815068e8782a6ad36b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 00:52:58.000000000","message":"Patch Set 72: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4952206c29a49c4a834315ecf516151\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/036802a3413c4007aa793a0b8de2d6b7 : FAILURE in 26s\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/8f76e739e82640dd8cfcff78a0d953ca : SKIPPED Skipped due to failed job loci-set-current-date","accounts_in_message":[],"_revision_number":72},{"id":"20f8309bbb99594124b26dd8cf4b4685e6237bc2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 03:05:45.000000000","message":"Uploaded patch set 73.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":73},{"id":"871a0c91ecd048902119a09528001d11d172d384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 03:17:29.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/801bd9c856094fde923008544636f93c\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/a8ad4b98d7f446dd86486d20ea542459 : SUCCESS in 29s\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/2380cb585336457cbf5f5b4a7fb99ea7 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":73},{"id":"d9374cf1694e42acb46099cf64ac650c1d064ae4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 03:26:00.000000000","message":"Uploaded patch set 74.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":74},{"id":"6935c8c86f5bcc9b89188c3f3ea69292a9fe1108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 04:30:38.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a415cfb854f4fbeba45d358e2280752\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/12d20db03c5f452e90ad56c4cc7912f5 : SUCCESS in 28s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/8f01445d7c0641fbaafed08ed1aa5413 : SUCCESS in 1h 02m 44s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/5490ce3181f64f33ad6360e35035d769 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":74},{"id":"b90c86c6024b4b0e29a00ec9614684727eaeef1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 08:37:46.000000000","message":"Uploaded patch set 75.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":75},{"id":"1c0d84921fb95d73fd70c4ec300e72992ffd0ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 08:38:04.000000000","message":"Patch Set 75: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Job loci-upload-multiarch-2025-2-ubuntu_noble_uv not defined\n\nThe problem appears in the the \"loci-compute-kit-2025-2-ubuntu_noble_uv\" job stanza:\n\n  job:\n      name: loci-compute-kit-2025-2-ubuntu_noble_uv\n      parent: openstack-helm-compute-kit-2025-2-ubuntu_noble_uv\n      files:\n        - ^.*$\n      required-projects:\n        - openstack/openstack-helm\n      dependencies:\n        - name: loci-build-multiarch-2025-2-ubuntu_noble_uv\n  ...\n\n  in \"openstack/loci/zuul.d/uv/build_2025.2-ubuntu_noble.yaml@master\", line 53","accounts_in_message":[],"_revision_number":75},{"id":"7f9696235a8d17269e90a78619f1a88f26035023","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 08:39:01.000000000","message":"Uploaded patch set 76.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":76},{"id":"9308f1e34deb82037ae7488471e09eb799615d84","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 08:41:18.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"2490531b1e6f7764c3fc0526cf1ffaaebaafa8b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 08:52:21.000000000","message":"Uploaded patch set 78.","accounts_in_message":[],"_revision_number":78},{"id":"288f20db89cc06716c9c597d10a690e2c778c60e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 11:30:43.000000000","message":"Patch Set 78: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21dab9e454094923931dc72c7c085ac7\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/0310284946d0419ea5758be9ce1018fb : SUCCESS in 25s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/8f9dc5b47edc4635a5ad79614e676206 : SUCCESS in 2h 33m 10s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/d1d4828e511f497e96a97edf138f6e93 : FAILURE in 2h 28m 50s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/788b0a7258af423aba5b11263ad98acc : FAILURE in 55m 45s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/c3cb5a183e8b4bdebc26c87c371f3efd : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6fd53d26b32c468bb004e22336f17c68 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a1c0010760044ce69e3006b750bf3ff9 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":78},{"id":"107acaf13e40050f0c4e69bdeef53e13ae2117ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 17:29:56.000000000","message":"Uploaded patch set 79.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":79},{"id":"1cee03c54bd7d994ee69d52cbcd2e8f118fc9eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 17:52:09.000000000","message":"Patch Set 79: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70c29049978843d8b690f6e9d8fb9fa7\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/28ef0865520c4e64af7a0681458802d8 : SUCCESS in 27s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/e666b80ee21045998522acf5f38588d0 : SUCCESS in 21m 25s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/b8bda94e16b74f489739c28e754131b0 : FAILURE in 17m 18s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/c664c2f0b16f4678af3d7913b4a8adcc : FAILURE in 5m 30s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/90ec4801480b40fc90d27d4676c57184 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/360eb39d29214f4dacfd76ce007a5767 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/028166c1d7c14efcad08269e58e0b46c : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":79},{"id":"cbba097213398f889312fcb40399d712a6ac50a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 18:48:19.000000000","message":"Uploaded patch set 80.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":80},{"id":"6c558312e583600f71e71380b223e40119398444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 19:22:31.000000000","message":"Patch Set 80: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8476a41998974c5e86351f394a6fd9e2\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/e4403cf474b642ec89e2dfae8aa4cda1 : SUCCESS in 25s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/27db4588d9e4413c906a4f9ba3624dd3 : SUCCESS in 29m 24s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/14c24311a38e497785fee0559f1a0402 : SUCCESS in 17m 46s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/0d019067d8e3457ca5b3fdc8a45a6590 : SUCCESS in 10m 59s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/04ce28266fac4640a19f411765414fb6 : FAILURE in 5m 19s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9fd25d9faa464935bf5adac29c084b7e : SKIPPED Skipped due to failed job loci-build-multiarch-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2d75e9c194174170970cfba62077b008 : SKIPPED Skipped due to failed job loci-build-multiarch-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":80},{"id":"11b475bc452940d164670a7763b6e920c1c9b0f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 19:42:02.000000000","message":"Uploaded patch set 81.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":81},{"id":"a64633c87d70f25fa33d97f1398cf57f85bc4ab2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 22:22:51.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"d29986d2747d53f7ce5697894f3e09219137e0fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-25 22:25:40.000000000","message":"Uploaded patch set 83.","accounts_in_message":[],"_revision_number":83},{"id":"8b315a26fcba5f228b85355e1f3845edfe00de96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:38:10.000000000","message":"Patch Set 83: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4f68a8627ba4dd4ab33f9d5ff907771\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/487801c5d4d841468f93ea76ca3fa25c : SUCCESS in 26s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1929b26df7e04ed2b3be291852be23eb : SUCCESS in 4h 10m 59s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/5d148f7734f140efbc7ecb27eae8a57b : SUCCESS in 2h 24m 58s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/1a02ddc0d0be4057beb42b9bfc2f01c5 : SUCCESS in 37m 59s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/bbd3d9672065491bada116a8f4725bcd : SUCCESS in 12m 53s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/bd48f46e0dab4150a3324e7ea873520a : SUCCESS in 1h 21m 12s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/97d21ea35e024f489ed8e002dea0e084 : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":83},{"id":"df564f3007220393c1cce46d3c55cff562da1fb0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 05:57:18.000000000","message":"Uploaded patch set 84.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":84},{"id":"e23b934e5fe6e7546a734b119d9a835aabc5b5b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 05:57:37.000000000","message":"Patch Set 84: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Job loci-build-ubuntu_noble_uv not defined\n\nThe problem appears in the the \"loci-build-2025-2-ubuntu_noble_uv_arm64\" job stanza:\n\n  job:\n      name: loci-build-2025-2-ubuntu_noble_uv_arm64\n      nodeset: loci-6nodes-ubuntu-noble-arm64\n      parent: loci-build-ubuntu_noble_uv\n      vars:\n        arch: \u0027arm64\u0027\n\n  in \"openstack/loci/zuul.d/uv/build_2025.2-ubuntu_noble.yaml@master\", line 26","accounts_in_message":[],"_revision_number":84},{"id":"c038b1778ab34a038910d8de2d27fb83d34c22de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 06:03:08.000000000","message":"Uploaded patch set 85.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":85},{"id":"44cf9aed8852ef77e1d7fe02a97fd8e3d829309f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 06:22:59.000000000","message":"Patch Set 85: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/644daedaf4c34ee180a6e5b5ec702ad6\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/32bc3f7f605b4563a496e994196c8cad : SUCCESS in 27s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/5ded3ed495be4bd880f1a2dd4929f07d : SUCCESS in 13m 53s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/b1f3ddc67ddd4cb38ae0c1606b9dd221 : FAILURE in 7m 40s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/5df0994db1e14a93abdb2bd02ee40b6f : FAILURE in 3m 46s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/0fc3dc404e8d43f48544dff4b95e3d65 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/679af75bb8d24fc89f70bc5e0ee29be9 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/94baffe7d11a474aae9f2a34ec771f09 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64","accounts_in_message":[],"_revision_number":85},{"id":"f2ad2a7edd095a1f90a1e9dc7b46497a9360e6a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 06:49:41.000000000","message":"Uploaded patch set 86.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":86},{"id":"b8e91779271eb01b8a76fb946dbdd02b9590991b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 07:05:24.000000000","message":"Patch Set 86: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5523c868f1243f2a1310fcf5a4cc21f\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/3dd624bedd9543abb2c631fbf0801682 : SUCCESS in 26s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/a6b5b58714bb4d04a239d8a3f53cda02 : SUCCESS in 10m 44s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/e0a03aeae0f64eff8017e22a51f9d376 : FAILURE in 4m 51s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/aad1acc595cb4dd9b839f09b94984a58 : FAILURE in 2m 47s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2e0798274aa547ec99daa34099af024b : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/5d31669f57584eef8e55e2f3b15e86ac : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/06eee636bdd14e1d8cd68f0da0483bb4 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_arm64","accounts_in_message":[],"_revision_number":86},{"id":"8ec355d57a6fdaa330f1ee39ff7d16e1326a8357","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 16:11:51.000000000","message":"Uploaded patch set 87.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":87},{"id":"f2b64c77e65491f94af655678c75247f15546337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:43:47.000000000","message":"Patch Set 87: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1bba7db368b47bb9c36474e93b80ef5\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/8d6fd1d6366349bdbd5d7b851c36279e : SUCCESS in 35s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/537ab3f960a649c18af12113b797d74f : SUCCESS in 24m 22s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/9d1d0f0b54d2401183f7942497d3c364 : FAILURE in 15m 04s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/b17c8bcae7c648d5a044ad8021951ec7 : FAILURE in 9m 30s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6e2834643cbf4403b03cbd4352b446b9 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2805c96bb4c14b7ebe2223ca71020874 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/807a5f913391463e8e62764e20ec58d7 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":87},{"id":"69190a100f085f03cce2aa95b155ccfd4f4c5f43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 16:52:00.000000000","message":"Uploaded patch set 88.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":88},{"id":"45706e02ca4cc3b3f34437c1e0be3fef150fac87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 17:25:56.000000000","message":"Patch Set 88: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d39fec48d82e4a468ebecd3ad0867f1c\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/fda83c9344554019810ab4387dd47bf2 : SUCCESS in 32s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/6482ba274ed94daa8b572ff1abd412ca : SUCCESS in 23m 01s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/c269bce4f20a47c3ba01218e8db585b1 : FAILURE in 13m 44s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/06cf4eb405f24a368f94ce218d07cd40 : FAILURE in 8m 40s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9bbc3dcff1864a1094f43a587a0ff5bd : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a68f9becfd8344c69cd39b8eb208b8f3 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9707e064f1ad4b0abc84a14f21e2e759 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":88},{"id":"071e822f9798978ebd7edab570401e8272f5411a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 17:39:59.000000000","message":"Uploaded patch set 89.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":89},{"id":"ce18bff61128e5a344737aa5daab33a02abbcf93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:18:21.000000000","message":"Patch Set 89: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/699ef92958274b829aa1b30075c92940\n\n- loci-set-current-date https://zuul.opendev.org/t/openstack/build/24c26255523b4717b483a65105965ab6 : SUCCESS in 27s\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/957d2521d9714ff9ae6ac6afccdeb68b : SUCCESS in 36m 41s (non-voting)\n- loci-build-2025-2-ubuntu_noble_uv_arm64 https://zuul.opendev.org/t/openstack/build/488f9525eed14324afc2284dbfd581f3 : FAILURE in 28m 46s\n- loci-build-2025-2-ubuntu_noble_uv_amd64 https://zuul.opendev.org/t/openstack/build/c9612d46b3964d47a639769010d166f8 : FAILURE in 16m 45s\n- loci-build-multiarch-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/87a57c5e92c541e2964864a567ea684c : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/65bcfa12949b4d698732dc40928c1743 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3bb4caa1aa0e42918cbc56011f1e2706 : SKIPPED Skipped due to failed job loci-build-2025-2-ubuntu_noble_uv_amd64","accounts_in_message":[],"_revision_number":89},{"id":"0a5204bf933398562cb7710bdcf686d5d3b76f90","tag":"autogenerated:gerrit:abandon","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-02 02:53:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":89}],"current_revision_number":89,"current_revision":"3e1da971c48230e4621cfa08ccad0a496020f523","revisions":{"3cdab12e6721550a783ba582645011d41a0e7f1a":{"kind":"REWORK","_number":1,"created":"2025-12-30 06:41:20.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/1"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:41:14.000000000","tz":-360},"subject":"Add arm64 architecture","message":"Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3cdab12e6721550a783ba582645011d41a0e7f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3cdab12e6721550a783ba582645011d41a0e7f1a"}]},"branch":"refs/heads/master"},"f22df70c30ce823d69a71e52d1d52d1e2b0fdced":{"kind":"REWORK","_number":2,"created":"2025-12-30 16:32:35.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/2"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 16:27:59.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f22df70c30ce823d69a71e52d1d52d1e2b0fdced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f22df70c30ce823d69a71e52d1d52d1e2b0fdced"}]},"branch":"refs/heads/master"},"02129289bb14a2642eaa3ebe2d419b9ac8b9269d":{"kind":"REWORK","_number":3,"created":"2025-12-30 17:04:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/3"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 17:04:46.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/02129289bb14a2642eaa3ebe2d419b9ac8b9269d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/02129289bb14a2642eaa3ebe2d419b9ac8b9269d"}]},"branch":"refs/heads/master"},"8ffdd901769e69ffa7632bf9625d8e1903193e1a":{"kind":"REWORK","_number":4,"created":"2025-12-30 17:13:39.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/4"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 17:13:34.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ffdd901769e69ffa7632bf9625d8e1903193e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ffdd901769e69ffa7632bf9625d8e1903193e1a"}]},"branch":"refs/heads/master"},"e984ca1d16bf38522cfebb4ad5efaab67edae80f":{"kind":"REWORK","_number":5,"created":"2025-12-30 17:48:27.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/5"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 17:48:17.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e984ca1d16bf38522cfebb4ad5efaab67edae80f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e984ca1d16bf38522cfebb4ad5efaab67edae80f"}]},"branch":"refs/heads/master"},"1fd1969948158f212c62e69239fe2c74762eff96":{"kind":"REWORK","_number":6,"created":"2025-12-31 05:02:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/6"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 05:02:39.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1fd1969948158f212c62e69239fe2c74762eff96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1fd1969948158f212c62e69239fe2c74762eff96"}]},"branch":"refs/heads/master"},"94403d9d006dbe7f6c7281a3a0979543deeecea6":{"kind":"REWORK","_number":7,"created":"2025-12-31 07:12:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/7"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 07:12:00.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94403d9d006dbe7f6c7281a3a0979543deeecea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94403d9d006dbe7f6c7281a3a0979543deeecea6"}]},"branch":"refs/heads/master"},"3adce47a69d3f83e3552badd380d456054c41b58":{"kind":"REWORK","_number":8,"created":"2025-12-31 14:02:32.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/8"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 14:02:24.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3adce47a69d3f83e3552badd380d456054c41b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3adce47a69d3f83e3552badd380d456054c41b58"}]},"branch":"refs/heads/master"},"5f01d4df20a83cabf7f1fb0e3aefb4b91829b56b":{"kind":"REWORK","_number":9,"created":"2025-12-31 15:09:12.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/9"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 15:09:05.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f01d4df20a83cabf7f1fb0e3aefb4b91829b56b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f01d4df20a83cabf7f1fb0e3aefb4b91829b56b"}]},"branch":"refs/heads/master"},"d6b7375c87f087d9b936ac04009fb0a6d5124bcb":{"kind":"REWORK","_number":10,"created":"2025-12-31 15:48:38.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/10"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 15:48:32.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d6b7375c87f087d9b936ac04009fb0a6d5124bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d6b7375c87f087d9b936ac04009fb0a6d5124bcb"}]},"branch":"refs/heads/master"},"e6a6e14c7d69825d391de9f997c3086378db93ed":{"kind":"REWORK","_number":11,"created":"2025-12-31 16:15:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/11"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 16:15:37.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e6a6e14c7d69825d391de9f997c3086378db93ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e6a6e14c7d69825d391de9f997c3086378db93ed"}]},"branch":"refs/heads/master"},"cd3288f370f69522f2acfdac64eb304dfdbeac18":{"kind":"REWORK","_number":12,"created":"2025-12-31 20:12:06.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/12"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-31 20:12:00.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cd3288f370f69522f2acfdac64eb304dfdbeac18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cd3288f370f69522f2acfdac64eb304dfdbeac18"}]},"branch":"refs/heads/master"},"d89a3b45ad2a26c15d92bb729f78b0bc02e1f32e":{"kind":"REWORK","_number":13,"created":"2026-01-01 00:42:40.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/13"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-01 00:42:27.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d89a3b45ad2a26c15d92bb729f78b0bc02e1f32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d89a3b45ad2a26c15d92bb729f78b0bc02e1f32e"}]},"branch":"refs/heads/master"},"dddfdb007faed36eceba5c0440207e48aa8a2a7a":{"kind":"REWORK","_number":14,"created":"2026-01-01 03:56:47.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/14"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-01 03:56:39.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dddfdb007faed36eceba5c0440207e48aa8a2a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dddfdb007faed36eceba5c0440207e48aa8a2a7a"}]},"branch":"refs/heads/master"},"c7623cea3a485e272ea547eb4a40b919b5cc16f3":{"kind":"REWORK","_number":15,"created":"2026-01-01 19:55:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/15"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-01 19:55:34.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c7623cea3a485e272ea547eb4a40b919b5cc16f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c7623cea3a485e272ea547eb4a40b919b5cc16f3"}]},"branch":"refs/heads/master"},"f394e5f86a078200d705ad90630c8df8631509a5":{"kind":"REWORK","_number":16,"created":"2026-01-04 21:14:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/16"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-04 21:14:14.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f394e5f86a078200d705ad90630c8df8631509a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f394e5f86a078200d705ad90630c8df8631509a5"}]},"branch":"refs/heads/master"},"d99c5131fb7e39bf40f2d3deea6170f91ba50385":{"kind":"REWORK","_number":17,"created":"2026-01-04 21:21:35.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/17"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-04 21:21:29.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d99c5131fb7e39bf40f2d3deea6170f91ba50385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d99c5131fb7e39bf40f2d3deea6170f91ba50385"}]},"branch":"refs/heads/master"},"9ae690232b3ef450f52551af824f24791b0bb14a":{"kind":"REWORK","_number":18,"created":"2026-01-05 16:17:10.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/18"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-05 16:04:24.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9ae690232b3ef450f52551af824f24791b0bb14a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9ae690232b3ef450f52551af824f24791b0bb14a"}]},"branch":"refs/heads/master"},"cc012ffe60e136d5bdb94923617eb7184b918873":{"kind":"REWORK","_number":19,"created":"2026-01-05 20:25:32.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/19"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-05 20:25:26.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cc012ffe60e136d5bdb94923617eb7184b918873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cc012ffe60e136d5bdb94923617eb7184b918873"}]},"branch":"refs/heads/master"},"9d9ba77a73a7e2844006f996a8c153e531f3e981":{"kind":"REWORK","_number":20,"created":"2026-01-05 21:01:43.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/20"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-05 21:01:37.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d9ba77a73a7e2844006f996a8c153e531f3e981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d9ba77a73a7e2844006f996a8c153e531f3e981"}]},"branch":"refs/heads/master"},"f65f21c8a165ed178b9bd29a50da139fb66e04a2":{"kind":"REWORK","_number":21,"created":"2026-01-05 22:14:23.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/21"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-05 22:14:17.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f65f21c8a165ed178b9bd29a50da139fb66e04a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f65f21c8a165ed178b9bd29a50da139fb66e04a2"}]},"branch":"refs/heads/master"},"1124dba5321d468c23529e44e2448b766ced3682":{"kind":"REWORK","_number":22,"created":"2026-01-05 23:21:50.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/22"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-05 23:21:43.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1124dba5321d468c23529e44e2448b766ced3682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1124dba5321d468c23529e44e2448b766ced3682"}]},"branch":"refs/heads/master"},"c30c187f177a3e20541b64b99b584ea8312443c7":{"kind":"REWORK","_number":23,"created":"2026-01-06 02:06:57.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/23"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-06 02:06:51.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c30c187f177a3e20541b64b99b584ea8312443c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c30c187f177a3e20541b64b99b584ea8312443c7"}]},"branch":"refs/heads/master"},"fe754cd2e92c694087fd8af37555bafc003f3b90":{"kind":"REWORK","_number":24,"created":"2026-01-06 04:40:01.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/24"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-06 04:39:54.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fe754cd2e92c694087fd8af37555bafc003f3b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fe754cd2e92c694087fd8af37555bafc003f3b90"}]},"branch":"refs/heads/master"},"ade6cdf6eb82fc07f650e4f1c6d114175e471b82":{"kind":"REWORK","_number":25,"created":"2026-01-06 06:21:17.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/25"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-06 06:21:10.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ade6cdf6eb82fc07f650e4f1c6d114175e471b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ade6cdf6eb82fc07f650e4f1c6d114175e471b82"}]},"branch":"refs/heads/master"},"5f8414ddc4f0b54e096fca6e0b9e9d515699b794":{"kind":"REWORK","_number":26,"created":"2026-01-07 21:49:09.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/26"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-07 21:48:57.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f8414ddc4f0b54e096fca6e0b9e9d515699b794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f8414ddc4f0b54e096fca6e0b9e9d515699b794"}]},"branch":"refs/heads/master"},"d84b643e6eaa68b51821b20b274c2cc484ba8057":{"kind":"REWORK","_number":27,"created":"2026-01-07 22:57:33.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/27"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-07 22:57:27.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d84b643e6eaa68b51821b20b274c2cc484ba8057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d84b643e6eaa68b51821b20b274c2cc484ba8057"}]},"branch":"refs/heads/master"},"fecadcdaf7b765cbc02aea404d0c6c4b84a72fa4":{"kind":"REWORK","_number":28,"created":"2026-01-08 00:22:10.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/28"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 00:22:03.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fecadcdaf7b765cbc02aea404d0c6c4b84a72fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fecadcdaf7b765cbc02aea404d0c6c4b84a72fa4"}]},"branch":"refs/heads/master"},"00d85b94f50920de17804a0b6dab61a4476d9dac":{"kind":"REWORK","_number":29,"created":"2026-01-08 03:56:41.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/29"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 03:56:35.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/00d85b94f50920de17804a0b6dab61a4476d9dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/00d85b94f50920de17804a0b6dab61a4476d9dac"}]},"branch":"refs/heads/master"},"4467c9d3aa39389e6069627f21b1951929b25a88":{"kind":"REWORK","_number":30,"created":"2026-01-08 04:57:13.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/30"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 04:57:05.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4467c9d3aa39389e6069627f21b1951929b25a88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4467c9d3aa39389e6069627f21b1951929b25a88"}]},"branch":"refs/heads/master"},"94e22166e3255852b870297e91d8c0e0b80b5cf4":{"kind":"REWORK","_number":31,"created":"2026-01-08 06:54:38.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/31"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 06:54:33.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94e22166e3255852b870297e91d8c0e0b80b5cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/94e22166e3255852b870297e91d8c0e0b80b5cf4"}]},"branch":"refs/heads/master"},"3b61b23a7c0a22dcaa3c6ad4ac9c20dd7cc0cd6d":{"kind":"REWORK","_number":32,"created":"2026-01-08 07:44:15.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/32"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 07:44:09.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b61b23a7c0a22dcaa3c6ad4ac9c20dd7cc0cd6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3b61b23a7c0a22dcaa3c6ad4ac9c20dd7cc0cd6d"}]},"branch":"refs/heads/master"},"33a65b5883e64c549be3564d31136b33ce3095ae":{"kind":"REWORK","_number":33,"created":"2026-01-08 20:28:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/33"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 20:28:20.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/33a65b5883e64c549be3564d31136b33ce3095ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/33a65b5883e64c549be3564d31136b33ce3095ae"}]},"branch":"refs/heads/master"},"43672523e6aa12920ef3435439d9fc6d00faea06":{"kind":"REWORK","_number":34,"created":"2026-01-08 21:39:27.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/34"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 21:39:20.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/43672523e6aa12920ef3435439d9fc6d00faea06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/43672523e6aa12920ef3435439d9fc6d00faea06"}]},"branch":"refs/heads/master"},"c71da245b13823f553c3b692334a8107ef369707":{"kind":"REWORK","_number":35,"created":"2026-01-08 23:51:30.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/35"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-08 23:51:21.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c71da245b13823f553c3b692334a8107ef369707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c71da245b13823f553c3b692334a8107ef369707"}]},"branch":"refs/heads/master"},"d2028393b65f68af42b6ad650a24e0f025933f15":{"kind":"NO_CHANGE","_number":36,"created":"2026-01-09 05:19:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/36"}}},"commit":{"parents":[{"commit":"a9f2417e0ffd227a5091c5aa251baa8678df5db8","subject":"Build images using UV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9f2417e0ffd227a5091c5aa251baa8678df5db8"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-09 05:19:18.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d2028393b65f68af42b6ad650a24e0f025933f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d2028393b65f68af42b6ad650a24e0f025933f15"}]},"branch":"refs/heads/master"},"7c7aa4652cdba40c45a93182ae72dbc77ea9579c":{"kind":"REWORK","_number":37,"created":"2026-01-12 18:01:03.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/37"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-12 18:00:56.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c7aa4652cdba40c45a93182ae72dbc77ea9579c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c7aa4652cdba40c45a93182ae72dbc77ea9579c"}]},"branch":"refs/heads/master"},"0ebe8ca7e7251b8bb0fc8a37ca0e74b2f7c9d620":{"kind":"REWORK","_number":38,"created":"2026-01-12 22:04:03.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/38"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-12 22:03:54.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0ebe8ca7e7251b8bb0fc8a37ca0e74b2f7c9d620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0ebe8ca7e7251b8bb0fc8a37ca0e74b2f7c9d620"}]},"branch":"refs/heads/master"},"061be07e212a56aaec5ac0822d23b7a9618807a1":{"kind":"REWORK","_number":39,"created":"2026-01-13 20:15:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/39"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-13 20:15:45.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/061be07e212a56aaec5ac0822d23b7a9618807a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/061be07e212a56aaec5ac0822d23b7a9618807a1"}]},"branch":"refs/heads/master"},"b5d52a7a9b791f7c5bb9bccb7e148c00d96c5f9e":{"kind":"REWORK","_number":40,"created":"2026-01-14 04:04:40.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/40"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 04:04:32.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b5d52a7a9b791f7c5bb9bccb7e148c00d96c5f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b5d52a7a9b791f7c5bb9bccb7e148c00d96c5f9e"}]},"branch":"refs/heads/master"},"1f8ab8f5ab3c3f4c75a012fd526217d0060a6ba5":{"kind":"REWORK","_number":41,"created":"2026-01-14 05:09:39.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/41"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 05:09:32.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f8ab8f5ab3c3f4c75a012fd526217d0060a6ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f8ab8f5ab3c3f4c75a012fd526217d0060a6ba5"}]},"branch":"refs/heads/master"},"1f780fb25925437e199f4f4f85501e636a4f668e":{"kind":"REWORK","_number":42,"created":"2026-01-14 05:14:14.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/42"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 05:14:09.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f780fb25925437e199f4f4f85501e636a4f668e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f780fb25925437e199f4f4f85501e636a4f668e"}]},"branch":"refs/heads/master"},"b2ca30c927fcb66af4b47beab17e58b272efaf82":{"kind":"REWORK","_number":43,"created":"2026-01-14 06:00:20.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/43"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 06:00:14.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b2ca30c927fcb66af4b47beab17e58b272efaf82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b2ca30c927fcb66af4b47beab17e58b272efaf82"}]},"branch":"refs/heads/master"},"47f56a14589ee3097007dfd9f72984ead2f2a3b2":{"kind":"REWORK","_number":44,"created":"2026-01-14 19:09:46.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/44"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 19:09:38.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/47f56a14589ee3097007dfd9f72984ead2f2a3b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/47f56a14589ee3097007dfd9f72984ead2f2a3b2"}]},"branch":"refs/heads/master"},"77b87cf500711e1ee40387cc939b65534e5173ec":{"kind":"REWORK","_number":45,"created":"2026-01-14 21:26:09.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/45"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 21:26:02.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/77b87cf500711e1ee40387cc939b65534e5173ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/77b87cf500711e1ee40387cc939b65534e5173ec"}]},"branch":"refs/heads/master"},"e3b15c3206f12185bede931291248bdd15d6b40b":{"kind":"REWORK","_number":46,"created":"2026-01-14 22:27:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/46"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-14 22:26:53.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e3b15c3206f12185bede931291248bdd15d6b40b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e3b15c3206f12185bede931291248bdd15d6b40b"}]},"branch":"refs/heads/master"},"228aa5e1c64c48df770e54c3f352f399b5569bae":{"kind":"REWORK","_number":47,"created":"2026-01-15 02:40:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/47"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-15 02:40:46.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/228aa5e1c64c48df770e54c3f352f399b5569bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/228aa5e1c64c48df770e54c3f352f399b5569bae"}]},"branch":"refs/heads/master"},"f3d027ad1669001f9019926f2be8f49c5bd49014":{"kind":"REWORK","_number":48,"created":"2026-01-15 04:57:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/48"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-15 04:57:26.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f3d027ad1669001f9019926f2be8f49c5bd49014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f3d027ad1669001f9019926f2be8f49c5bd49014"}]},"branch":"refs/heads/master"},"28454fa5c5e268a3623f67650a6c8ec52ebc5b7d":{"kind":"REWORK","_number":49,"created":"2026-01-15 06:41:27.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/49"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-15 06:41:19.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/28454fa5c5e268a3623f67650a6c8ec52ebc5b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/28454fa5c5e268a3623f67650a6c8ec52ebc5b7d"}]},"branch":"refs/heads/master"},"5c1be01b8d7e612e73fe56799683386335f98bb6":{"kind":"REWORK","_number":50,"created":"2026-01-15 20:24:07.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/50"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-15 20:23:59.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c1be01b8d7e612e73fe56799683386335f98bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c1be01b8d7e612e73fe56799683386335f98bb6"}]},"branch":"refs/heads/master"},"f24e368ad1da24f07b2536836daa2e5c953ff7ff":{"kind":"REWORK","_number":51,"created":"2026-01-15 23:04:14.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/51"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-15 23:04:07.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f24e368ad1da24f07b2536836daa2e5c953ff7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f24e368ad1da24f07b2536836daa2e5c953ff7ff"}]},"branch":"refs/heads/master"},"ffc8112f3966c82c1b153eb003b1dbe1454945fb":{"kind":"REWORK","_number":52,"created":"2026-01-15 23:49:58.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/52"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-15 23:49:51.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ffc8112f3966c82c1b153eb003b1dbe1454945fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ffc8112f3966c82c1b153eb003b1dbe1454945fb"}]},"branch":"refs/heads/master"},"148f8ab5e527ce6cb9d64c0a244de84ba9c1d0e2":{"kind":"REWORK","_number":53,"created":"2026-01-22 21:04:58.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/53"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:04:50.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/148f8ab5e527ce6cb9d64c0a244de84ba9c1d0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/148f8ab5e527ce6cb9d64c0a244de84ba9c1d0e2"}]},"branch":"refs/heads/master"},"e4a0a35d67a97ad57de091a344444113aa7509b1":{"kind":"REWORK","_number":54,"created":"2026-01-22 21:35:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/54"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:35:53.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e4a0a35d67a97ad57de091a344444113aa7509b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e4a0a35d67a97ad57de091a344444113aa7509b1"}]},"branch":"refs/heads/master"},"59bdf470db66af43c3969aa908c1db6c45d32a12":{"kind":"REWORK","_number":55,"created":"2026-01-22 22:09:13.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/55"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 22:09:07.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59bdf470db66af43c3969aa908c1db6c45d32a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59bdf470db66af43c3969aa908c1db6c45d32a12"}]},"branch":"refs/heads/master"},"e047e6ceceec11552cc6b64f75fcb2338f536b8a":{"kind":"REWORK","_number":56,"created":"2026-01-22 22:27:22.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/56"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 22:27:14.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e047e6ceceec11552cc6b64f75fcb2338f536b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e047e6ceceec11552cc6b64f75fcb2338f536b8a"}]},"branch":"refs/heads/master"},"d7424595f294e51004f010297edfbacd2635fbb8":{"kind":"REWORK","_number":57,"created":"2026-01-22 23:23:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/57"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 23:23:46.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d7424595f294e51004f010297edfbacd2635fbb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d7424595f294e51004f010297edfbacd2635fbb8"}]},"branch":"refs/heads/master"},"acada8071c058d01b858ab18e9cb2108e8b03961":{"kind":"REWORK","_number":58,"created":"2026-01-23 03:44:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/58"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 03:44:15.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/acada8071c058d01b858ab18e9cb2108e8b03961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/acada8071c058d01b858ab18e9cb2108e8b03961"}]},"branch":"refs/heads/master"},"60ec802fd4e4db3c3854e176b1833543089d3c4d":{"kind":"REWORK","_number":59,"created":"2026-01-23 05:18:23.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/59"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 05:18:16.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/60ec802fd4e4db3c3854e176b1833543089d3c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/60ec802fd4e4db3c3854e176b1833543089d3c4d"}]},"branch":"refs/heads/master"},"3da334b9159a9bd3e39d402c1d6df7587653aae2":{"kind":"REWORK","_number":60,"created":"2026-01-23 06:28:25.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/60"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 06:28:19.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3da334b9159a9bd3e39d402c1d6df7587653aae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3da334b9159a9bd3e39d402c1d6df7587653aae2"}]},"branch":"refs/heads/master"},"cbf887b210829235c89381b5ac022e2032c228d6":{"kind":"REWORK","_number":61,"created":"2026-01-23 17:27:14.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/61"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 17:27:04.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cbf887b210829235c89381b5ac022e2032c228d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cbf887b210829235c89381b5ac022e2032c228d6"}]},"branch":"refs/heads/master"},"e84462d65cbe6c9d1bbb48d00874427934157ef5":{"kind":"REWORK","_number":62,"created":"2026-01-23 17:31:42.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/62"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 17:31:36.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e84462d65cbe6c9d1bbb48d00874427934157ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e84462d65cbe6c9d1bbb48d00874427934157ef5"}]},"branch":"refs/heads/master"},"2b10e3a11fcc667a5bf07aecb3d17abab67b4f14":{"kind":"REWORK","_number":63,"created":"2026-01-23 21:06:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/63"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 21:05:52.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2b10e3a11fcc667a5bf07aecb3d17abab67b4f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2b10e3a11fcc667a5bf07aecb3d17abab67b4f14"}]},"branch":"refs/heads/master"},"c517000e8a5103bb73249a71936205618ddb971c":{"kind":"REWORK","_number":64,"created":"2026-01-23 21:58:43.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/64"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 21:58:37.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c517000e8a5103bb73249a71936205618ddb971c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c517000e8a5103bb73249a71936205618ddb971c"}]},"branch":"refs/heads/master"},"4f667031a7daf362f6c98861b54aff03875fdbfc":{"kind":"REWORK","_number":65,"created":"2026-01-24 21:32:17.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/65"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-24 21:28:37.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4f667031a7daf362f6c98861b54aff03875fdbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4f667031a7daf362f6c98861b54aff03875fdbfc"}]},"branch":"refs/heads/master"},"95dc1fca85bfd462791e6e77a5188d15490ae864":{"kind":"REWORK","_number":66,"created":"2026-01-24 21:42:39.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/66"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-24 21:42:34.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95dc1fca85bfd462791e6e77a5188d15490ae864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95dc1fca85bfd462791e6e77a5188d15490ae864"}]},"branch":"refs/heads/master"},"9338918d0e6cce90347c1831e18f618907278d25":{"kind":"REWORK","_number":67,"created":"2026-01-24 23:45:20.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/67"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-24 23:45:14.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9338918d0e6cce90347c1831e18f618907278d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9338918d0e6cce90347c1831e18f618907278d25"}]},"branch":"refs/heads/master"},"219d5504832166e2814fa6ace2340993e9367c42":{"kind":"REWORK","_number":68,"created":"2026-01-25 00:28:48.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/68"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 00:28:38.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/219d5504832166e2814fa6ace2340993e9367c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/219d5504832166e2814fa6ace2340993e9367c42"}]},"branch":"refs/heads/master"},"1f435822c996e4f0b0cd3a2b7a93d59f81ece152":{"kind":"REWORK","_number":69,"created":"2026-01-25 00:33:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/69"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 00:32:55.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f435822c996e4f0b0cd3a2b7a93d59f81ece152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1f435822c996e4f0b0cd3a2b7a93d59f81ece152"}]},"branch":"refs/heads/master"},"c31681e79788d2d5007030edea310552337e06d4":{"kind":"REWORK","_number":70,"created":"2026-01-25 00:35:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/70"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 00:35:49.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c31681e79788d2d5007030edea310552337e06d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c31681e79788d2d5007030edea310552337e06d4"}]},"branch":"refs/heads/master"},"c4c3247615841852e484c393533a59e6c5d166ba":{"kind":"NO_CHANGE","_number":71,"created":"2026-01-25 00:36:07.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/71"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 00:36:00.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c4c3247615841852e484c393533a59e6c5d166ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c4c3247615841852e484c393533a59e6c5d166ba"}]},"branch":"refs/heads/master"},"df543590499ddcbe3931ce69a64ed9ec1d4b51e3":{"kind":"REWORK","_number":72,"created":"2026-01-25 00:51:39.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/72"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 00:51:32.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/df543590499ddcbe3931ce69a64ed9ec1d4b51e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/df543590499ddcbe3931ce69a64ed9ec1d4b51e3"}]},"branch":"refs/heads/master"},"3184f056d6a6cee1d2549a1b66bd98fcca54c91c":{"kind":"REWORK","_number":73,"created":"2026-01-25 03:05:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/73"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 03:05:37.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3184f056d6a6cee1d2549a1b66bd98fcca54c91c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3184f056d6a6cee1d2549a1b66bd98fcca54c91c"}]},"branch":"refs/heads/master"},"7c0bbf19557d429875351bcdb282768a6f0cf4d5":{"kind":"REWORK","_number":74,"created":"2026-01-25 03:26:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/74"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 03:25:54.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c0bbf19557d429875351bcdb282768a6f0cf4d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c0bbf19557d429875351bcdb282768a6f0cf4d5"}]},"branch":"refs/heads/master"},"f7f94274c54dfed8ab94bb676246535117160a74":{"kind":"REWORK","_number":75,"created":"2026-01-25 08:37:46.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/75"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 08:37:39.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f7f94274c54dfed8ab94bb676246535117160a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f7f94274c54dfed8ab94bb676246535117160a74"}]},"branch":"refs/heads/master"},"4d5b08b36b2e02a762acc53a72d684d8a4ba6608":{"kind":"REWORK","_number":76,"created":"2026-01-25 08:39:01.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/76"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 08:38:50.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4d5b08b36b2e02a762acc53a72d684d8a4ba6608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4d5b08b36b2e02a762acc53a72d684d8a4ba6608"}]},"branch":"refs/heads/master"},"806cea130186be667b5effa09aef2f6c9bb8bbd8":{"kind":"REWORK","_number":77,"created":"2026-01-25 08:41:18.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/77"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 08:41:11.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/806cea130186be667b5effa09aef2f6c9bb8bbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/806cea130186be667b5effa09aef2f6c9bb8bbd8"}]},"branch":"refs/heads/master"},"7d0d3ab5939fc9128e532ab4e91b73ed0b8b6454":{"kind":"REWORK","_number":78,"created":"2026-01-25 08:52:21.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/78"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 08:52:15.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7d0d3ab5939fc9128e532ab4e91b73ed0b8b6454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7d0d3ab5939fc9128e532ab4e91b73ed0b8b6454"}]},"branch":"refs/heads/master"},"2e8e8b192a91efaa92b2f47b0dbecdc5f7552aad":{"kind":"REWORK","_number":79,"created":"2026-01-25 17:29:56.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/79"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 17:29:50.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e8e8b192a91efaa92b2f47b0dbecdc5f7552aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2e8e8b192a91efaa92b2f47b0dbecdc5f7552aad"}]},"branch":"refs/heads/master"},"fd9a0a7e1a989c5d91644cdb718e507f1769bea6":{"kind":"REWORK","_number":80,"created":"2026-01-25 18:48:19.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/80"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 18:48:12.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fd9a0a7e1a989c5d91644cdb718e507f1769bea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fd9a0a7e1a989c5d91644cdb718e507f1769bea6"}]},"branch":"refs/heads/master"},"9d99bb69f9aa119419a3953a6ba903641b135317":{"kind":"REWORK","_number":81,"created":"2026-01-25 19:42:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/81"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 19:41:52.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d99bb69f9aa119419a3953a6ba903641b135317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d99bb69f9aa119419a3953a6ba903641b135317"}]},"branch":"refs/heads/master"},"e1251caae554ed868b83ef1c44fe52a4d11319fd":{"kind":"REWORK","_number":82,"created":"2026-01-25 22:22:51.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/82"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 22:22:44.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e1251caae554ed868b83ef1c44fe52a4d11319fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e1251caae554ed868b83ef1c44fe52a4d11319fd"}]},"branch":"refs/heads/master"},"d509705d81f3f8704bfece92e7c0b13562c7e50f":{"kind":"REWORK","_number":83,"created":"2026-01-25 22:25:40.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/83"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-25 22:25:32.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d509705d81f3f8704bfece92e7c0b13562c7e50f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d509705d81f3f8704bfece92e7c0b13562c7e50f"}]},"branch":"refs/heads/master"},"076dba3f4164bcef805a0a814274188c7993ea97":{"kind":"REWORK","_number":84,"created":"2026-01-26 05:57:18.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/84"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-26 05:57:12.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/076dba3f4164bcef805a0a814274188c7993ea97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/076dba3f4164bcef805a0a814274188c7993ea97"}]},"branch":"refs/heads/master"},"05b362297f61982f97430c49c14fcfc62aac6005":{"kind":"REWORK","_number":85,"created":"2026-01-26 06:03:08.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/85"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-26 06:03:01.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/05b362297f61982f97430c49c14fcfc62aac6005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/05b362297f61982f97430c49c14fcfc62aac6005"}]},"branch":"refs/heads/master"},"9c8ed11f4bf7239e247d2f7f417d4abce2e1d27b":{"kind":"REWORK","_number":86,"created":"2026-01-26 06:49:41.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/86"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-26 06:49:35.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9c8ed11f4bf7239e247d2f7f417d4abce2e1d27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9c8ed11f4bf7239e247d2f7f417d4abce2e1d27b"}]},"branch":"refs/heads/master"},"6a000ae1bff1ef595520e15eb80182929f834569":{"kind":"REWORK","_number":87,"created":"2026-01-26 16:11:51.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/87"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-26 16:11:45.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6a000ae1bff1ef595520e15eb80182929f834569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6a000ae1bff1ef595520e15eb80182929f834569"}]},"branch":"refs/heads/master"},"95591385bc0f5404d20e67e28040d5aebc1a2a59":{"kind":"REWORK","_number":88,"created":"2026-01-26 16:52:00.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/88"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-26 16:51:54.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95591385bc0f5404d20e67e28040d5aebc1a2a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95591385bc0f5404d20e67e28040d5aebc1a2a59"}]},"branch":"refs/heads/master"},"3e1da971c48230e4621cfa08ccad0a496020f523":{"kind":"REWORK","_number":89,"created":"2026-01-26 17:39:59.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/53/971953/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/53/971953/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/53/971953/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/53/971953/89"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-12-30 06:40:49.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-26 17:39:53.000000000","tz":-360},"subject":"[WIP] Add arm64 architecture","message":"[WIP] Add arm64 architecture\n\nChange-Id: I2629d39383aa9b75fc8247169bcd3ae5e1284709\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3e1da971c48230e4621cfa08ccad0a496020f523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3e1da971c48230e4621cfa08ccad0a496020f523"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
