)]}'
{"id":"openstack%2Floci~973692","triplet_id":"openstack%2Floci~master~I5e505c763b36671af8a2a294a2497c399fd1a93e","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-01-21 18:56:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5e505c763b36671af8a2a294a2497c399fd1a93e","subject":"Add masakari and masakari-monotors images","status":"MERGED","created":"2026-01-16 19:26:36.000000000","updated":"2026-01-21 19:11:35.000000000","submitted":"2026-01-21 18:56:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973692","meta_rev_id":"28b9c0db42bf2460492fda3704369367bcf183ba","_number":973692,"virtual_id_number":973692,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 18:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2026-01-21 16:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2026-01-20 17:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2026-01-21 16:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-17 00:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 17:14:44.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-01-21 16:50:54.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"c9698ffd3e09ba0d9f195744d0f68c01c0f68ded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-16 19:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e59a2d959945766389251f8b76d864d29bf33ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-16 21:59:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8a3c8eaca84d4ce838fb8494a36297b84438748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 00:58:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954a94c9f1b94d9793f622510a516c28\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c2c5c2aed8a140b39c0efb906d393138 : SUCCESS in 2h 58m 25s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9021cf34db374b5b94ed513fc0385a26 : SUCCESS in 1h 25m 42s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f53b1aa9b2e741b583f2be3d36d83641 : SUCCESS in 43m 37s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4b47e5a845104732a699d1a65f7e4307 : SUCCESS in 41m 41s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/bf1d0d9e0f52480dbca77cc67160a611 : SUCCESS in 1h 33m 16s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e75f9bd16f9641228046e6d8981a344b : SUCCESS in 1h 16m 11s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fde312d4b61b417ba1aefa7e626489c3 : SUCCESS in 26m 31s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2dab097150f949d6b7231423eb1fc352 : SUCCESS in 1h 11m 59s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ec2626a9d1654fb09da826e56bc74b42 : SUCCESS in 1h 02m 09s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1a570c80214f40f7806d02bd3afcf2d7 : SUCCESS in 35m 36s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/34899941f0de4d0fbf4f103d859e1ee8 : SUCCESS in 52m 25s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/69944ca0bad349f8927ce9b437ee4afa : SUCCESS in 1h 23m 56s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2c4ab2405b32473c945d3cde8343ac99 : SUCCESS in 48m 16s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/db59c8d0885c476daccae09ca0b6ad29 : SUCCESS in 1h 18m 23s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/0dfa8615d30647d482b725dd443467c3 : SUCCESS in 1h 03m 47s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2d007a2fa4cb40e0a5db1386a2049eac : SUCCESS in 44m 55s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3371f9a43cf544a2a61eae21245f818f : SUCCESS in 53m 19s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/fe1705ee4e5e4c8eba2cc09189075fc0 : SUCCESS in 1h 18m 45s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d5632112cae14691bbc3165f85984dea : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3140cb5ae776f469eee47f3b8d37c6be7c916419","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-01-20 17:14:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"90ad6a45a361f6782ed7eb750ab6c1c193aa27cf","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-01-21 16:50:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6c16ba86a0275789d8e0fccff1d22104e722265c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 16:51:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2f9bb1a245106647a709daf10656c7270894212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 18:56:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13bbcad37502407b8ee17adcf73c4f05\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f6475695893d47c282ef762e88b40d23 : SUCCESS in 1h 42m 00s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9765ab5ee1a549018095264c4c582f2d : SUCCESS in 1h 10m 58s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a49806b35230488b96d934d9bcf0f400 : SUCCESS in 42m 32s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/92fa971eab8c409c8c4052a6031815b3 : SUCCESS in 1h 34m 45s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba18e6f0f58d4a2eba8529e8ffb1ac3e : SUCCESS in 41m 37s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/89819939b1914a94b7768e19730ec003 : SUCCESS in 50m 10s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6e375e89e9db4cae94b93d606bd80d5d : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5aaa39e76a883496220c6b678d67488a6bce572f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 18:56:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"28b9c0db42bf2460492fda3704369367bcf183ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 19:11:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f70851ec380c4f99a58979c39ed841d6\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b8b00b030a434c82a1b075bd55dee8bd : SUCCESS in 7m 22s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/08a772766b0b4be4aef73705cb8b206f : SUCCESS in 8m 34s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d1bbc79467d94778aec8ab70994c5f66 : SUCCESS in 8m 42s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8b44e7949dc6424582c3d3b311ee3682 : SUCCESS in 8m 09s\n- loci-promote-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6b5e902841db4976947b70d89fa38b90 : SUCCESS in 8m 08s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e63ceb69eba1433ea455542bf441c633 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e133aac4538ccf23d8ebe303b40d196d301fb16","revisions":{"cb86073c40adc663e9a6af043671a66a6027a809":{"kind":"REWORK","_number":1,"created":"2026-01-16 19:26:36.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/92/973692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/92/973692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/973692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/973692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/973692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/92/973692/1"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 19:26:14.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 19:26:14.000000000","tz":-360},"subject":"Add masakari and masakari-monotors images","message":"Add masakari and masakari-monotors images\n\nChange-Id: I5e505c763b36671af8a2a294a2497c399fd1a93e\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cb86073c40adc663e9a6af043671a66a6027a809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cb86073c40adc663e9a6af043671a66a6027a809"}]},"branch":"refs/heads/master"},"0e133aac4538ccf23d8ebe303b40d196d301fb16":{"kind":"REWORK","_number":2,"created":"2026-01-16 21:59:19.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/92/973692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/92/973692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/973692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/973692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/973692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/92/973692/2"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 19:26:14.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 21:59:11.000000000","tz":-360},"subject":"Add masakari and masakari-monotors images","message":"Add masakari and masakari-monotors images\n\nChange-Id: I5e505c763b36671af8a2a294a2497c399fd1a93e\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0e133aac4538ccf23d8ebe303b40d196d301fb16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0e133aac4538ccf23d8ebe303b40d196d301fb16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
