)]}'
{"id":"openstack%2Floci~973720","triplet_id":"openstack%2Floci~master~I699a266ac5f291771f2bc1161279a0a30b9c0d65","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-01-21 18:56:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I699a266ac5f291771f2bc1161279a0a30b9c0d65","subject":"Remove monasca image","status":"MERGED","created":"2026-01-16 21:41:51.000000000","updated":"2026-01-21 19:02:43.000000000","submitted":"2026-01-21 18:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973720","meta_rev_id":"de4a5d846310da5de7a68738e51f232fea1690c5","_number":973720,"virtual_id_number":973720,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 18:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2026-01-21 16:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2026-01-20 17:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2026-01-21 16:50:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-17 00:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 17:14:35.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-01-21 16:50:30.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"e08f31502a5332921034608f9edfbf069cbcfba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-16 21:41:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b212deec53b467735001e56956ffcb218b89044f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-16 21:50:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"45d73486c65191035921e2af1c19c060273bda3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 00:50:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01cea79fe60449f38cbda1b3590ef684\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c4524abbfafd4dc587900d99ca48081b : SUCCESS in 2h 54m 02s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ed48c52d2159419ca8cc0244257e00c4 : SUCCESS in 1h 19m 10s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2df2c7e5197f4376b07842d5f37e74ad : SUCCESS in 47m 54s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/610fe167fae745199c6b1f1ac5168f43 : SUCCESS in 32m 55s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ce571fc1e93c43b5b667012f69590e4e : SUCCESS in 55m 40s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/74164793402341b0a51c2b679cc0c663 : SUCCESS in 1h 12m 26s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4e2d2abea3bc4086a15ca79beac7ad52 : SUCCESS in 42m 17s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e64666b629ad4a518c02ddb3510af56b : SUCCESS in 54m 33s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a39f828e1f9e4c1ebc9645063bb87c36 : SUCCESS in 1h 05m 18s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0cb3cedcc50f44b99a62767b65887377 : SUCCESS in 31m 34s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/81ff2f5e57fe418ebf1f2f7f434e8466 : SUCCESS in 1h 25m 59s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/33fde6501eb749d9b69b40fd0a1ff78b : SUCCESS in 1h 15m 33s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b4f04bd40b214a3ea2f90699902d9d53 : SUCCESS in 36m 21s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/da0d4e5b31db44808ee20683c7359c97 : SUCCESS in 41m 55s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/72bca502d2614271aa59ac79a5840aef : SUCCESS in 1h 20m 55s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e2a80baeee694baa83a74d8f4b927e0d : SUCCESS in 35m 24s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9eebdcdddcee4c498a61d8db4324cdc4 : SUCCESS in 55m 09s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/fc9c5ba0ab884de2af7cbf15d761e672 : SUCCESS in 1h 18m 23s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/da58239a0a6947f7b65b648abcd46afa : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":2},{"id":"fb0443c7dfe84f5e8daf06ad7c9438e33a535dc1","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-01-20 17:14:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aab2c919971be8f1b1954e7fa4178aff44c0a44d","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-01-21 16:50:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6265b6323da5e56789e7c351071da992674c4b6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 16:50:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"764fb955d02a76f0f55700dd3dd2fac8a7df94e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 18:56:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b13976e5f5d42a0bb97d9f2a2585d7f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7a805b94f41b4890a80404e68e79c66c : SUCCESS in 1h 47m 28s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/25cd24344364446897e1922b017aa12b : SUCCESS in 58m 07s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/786450f5187e488aa4df2e4dde27eab5 : SUCCESS in 44m 15s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7e7a181900054cfc92e256dd3caf8860 : SUCCESS in 1h 38m 24s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1d67cc9c3f094958886954136a6fd568 : SUCCESS in 1h 12m 27s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ee0788344bae432ab5d252a44ecadf70 : SUCCESS in 1h 13m 44s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/60e1f4f832de403697302b4d4b8d43b4 : SUCCESS in 1h 18m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3cf1db8032dcfd9e90674f2ebe9a08078c42af0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 18:56:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"de4a5d846310da5de7a68738e51f232fea1690c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 19:02:43.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa90eeff881244248b1a666f3fecdf69\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/52cf61b5256849589ad3cb8aa56caed5 : FAILURE in 5m 51s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fb3cb7ba3f944a7694fdf524730224c2 : FAILURE in 5m 24s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f47aacae97134078b805bce706f1395e : FAILURE in 5m 43s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cab6f365334e4a30b4ea4b5d02952811 : FAILURE in 5m 23s\n- loci-promote-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/cd723294b0594db59a730606f609d45b : FAILURE in 5m 43s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/739b14af31e0464e8101ef547a9fb9e3 : FAILURE in 5m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae2fb9172e0f4625e24c576bf88b401fb603270c","revisions":{"026b0e548028fa3ac0f6bafb6d245fd469d7e77d":{"kind":"REWORK","_number":1,"created":"2026-01-16 21:41:51.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/20/973720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/973720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/973720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/973720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/973720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/973720/1"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 21:39:59.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 21:39:59.000000000","tz":-360},"subject":"Remove monasca image","message":"Remove monasca image\n\nThe monasca project has been retired for quite\na while now.\n\nSee I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\n\nChange-Id: I699a266ac5f291771f2bc1161279a0a30b9c0d65\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/026b0e548028fa3ac0f6bafb6d245fd469d7e77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/026b0e548028fa3ac0f6bafb6d245fd469d7e77d"}]},"branch":"refs/heads/master"},"ae2fb9172e0f4625e24c576bf88b401fb603270c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-16 21:50:38.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/20/973720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/973720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/973720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/973720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/973720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/973720/2"}}},"commit":{"parents":[{"commit":"5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257","subject":"Re-encrypt secrets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5ccb3b99ea7ba6dcc13963c9f9b0a2d19c499257"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 21:39:59.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-16 21:50:13.000000000","tz":-360},"subject":"Remove monasca image","message":"Remove monasca image\n\nThe monasca project has been retired quite\na while ago.\n\nSee I3cb522ce8f51424b64e93c1efaf0dfd1781cd5ac\n\nChange-Id: I699a266ac5f291771f2bc1161279a0a30b9c0d65\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ae2fb9172e0f4625e24c576bf88b401fb603270c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ae2fb9172e0f4625e24c576bf88b401fb603270c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
