)]}'
{"id":"openstack%2Floci~974372","triplet_id":"openstack%2Floci~master~I661e750ac8f7503da7efad38c9f83811c4fd71bd","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-01-27 02:01:45.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I661e750ac8f7503da7efad38c9f83811c4fd71bd","subject":"Add swift image","status":"MERGED","created":"2026-01-22 21:28:28.000000000","updated":"2026-01-27 07:13:18.000000000","submitted":"2026-01-27 07:04:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974372","meta_rev_id":"8eb76e2004e4071c7bd3cd28a44b50a3bc7ceae0","_number":974372,"virtual_id_number":974372,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 07:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2026-01-26 16:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2026-01-26 17:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":1,"date":"2026-01-26 17:15:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 01:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 16:41:53.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2026-01-26 17:15:03.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"7b0a3ca53754b84b52751bc76e2c5ea3b6254cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-22 21:28:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47b267e8a9e6a27bef45d240e070665f1f2b4831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 01:28:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a30812bcfe3b4976b8cbea635dc1c5c6\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/daeae806c3fc4d01a251b78a7f4cf1c0 : SUCCESS in 3h 58m 10s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/90dd893695ba4eb69bf62fb259d58b08 : SUCCESS in 1h 06m 29s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e3ae99f2d91b47aeaf439ef0cfd886fb : SUCCESS in 1h 18m 06s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5b28391db7124846b90e9e9fc16b5753 : SUCCESS in 42m 29s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/08ba440d20f047508975f38f8eee203d : SUCCESS in 38m 22s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e5562097c4f447e8f4b300ea5c84071 : SUCCESS in 1h 01m 53s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1c94247f8d3d435a81b8f9d1c29b1b0a : SUCCESS in 30m 35s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/12ffe336dfac477f9f75c066fa7cd926 : SUCCESS in 42m 00s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3379b5fd32004519815aa5813215c6fd : SUCCESS in 49m 02s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f53f6237fdbc4ef8bf227c01b9c461bb : SUCCESS in 40m 05s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/16c4fc2844f844b3b4dda5c763934385 : SUCCESS in 1h 14m 25s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6ca569181aec4978a803d552d20f8449 : SUCCESS in 57m 36s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6c17f315e76a48229962b5c010cc2556 : SUCCESS in 29m 32s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/89dec048143a4873b530614fbd777d8b : SUCCESS in 1h 32m 45s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/30a6fb993b3244418fe0a8d7c87db9c0 : FAILURE in 2h 13m 34s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/56ddf6cdc4fe4d9e98c58f5d9689619c : FAILURE in 46m 17s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/b8b4a4291d234f60b6838eeb552b2113 : SUCCESS in 56m 53s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d44644ad124d4a04bf394d57921b7fbd : FAILURE in 2h 47m 36s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/001df08982dc4b939e0873cf5a22dd93 : FAILURE in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ec777d9438f77c6af1a73984d5d7889100a7a710","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 03:43:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ba89cf02df4c20c86270e136ae19d07bfcacee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 06:42:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c34c143fe734888a922d4cfb7f4c1b8\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/d2b2f9d5128242f89ea32387da232172 : SUCCESS in 2h 58m 52s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/03082ee09662493081984790b6b0e1ac : SUCCESS in 39m 25s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/098de0ee2c7d477eba6854ab4164d3b8 : SUCCESS in 1h 04m 09s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0fd9768a9d7246238c0bd3e9de1ef75d : SUCCESS in 44m 28s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ce5215cb434a4156b49a68f4f7675655 : SUCCESS in 44m 51s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d3dff278581b4d6d8d6769e7f53cdcd6 : SUCCESS in 46m 50s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ba777d2dbfbf41d79841c3972a5fc1e8 : SUCCESS in 34m 54s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4a7b923b0198497c9cc44be7ff9b5994 : SUCCESS in 44m 39s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/499e08754f8341a5a020ee78703132c0 : SUCCESS in 46m 42s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c39c8e2ed3b343c8b6ca635f1aafb68a : SUCCESS in 48m 30s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dffad001a22243e09eeec49fc48e29d9 : SUCCESS in 44m 26s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/98d5bc8c746b47199165b08fc4440a04 : SUCCESS in 1h 05m 56s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cad611f96c6842a28084c3c8032dc236 : SUCCESS in 50m 18s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/b6de61ef861b4d49b8302399675fb16f : SUCCESS in 37m 02s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/239f07cb01234a799461b9c8ef24e8a0 : FAILURE in 2h 07m 12s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/474ea655c7034cb5890b383482a024ee : FAILURE in 38m 19s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3ee5ded2dd144ae0865dabe5b92c4445 : SUCCESS in 36m 52s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/0a87c30efbef465097e1ae57f9e063f3 : FAILURE in 2h 16m 44s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/75f2a8b6560b4664af482d3bf5b169bf : FAILURE in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0fa3a9a34c30a2285c84f0aff6ca5e0ed7c54889","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 07:48:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18242bc7d6ec16117e817269ac3921677f3ad2c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 10:43:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fd327f1a112439ca8d27665ee8c7627\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/c89e21a3370944a0b780ae3e305d3155 : SUCCESS in 2h 50m 11s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0722af43d9194c818459c54e0341b33e : SUCCESS in 1h 06m 45s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d3c50ee284404c84aa3dd5a7ab6bc67f : SUCCESS in 57m 17s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8b2e09e5cc0a4417a8c0f3eb0a1a3949 : SUCCESS in 48m 17s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4fb5d5f6cb304ebca17137442a54ec63 : SUCCESS in 1h 21m 48s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1be685633e7847cca3fc5471a72cc448 : SUCCESS in 48m 00s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c46da3a1fe9a48d980406656fb9a8348 : SUCCESS in 45m 02s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/994e2a13280f47c3ad30e33f45dae808 : SUCCESS in 1h 17m 54s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f8066f8eb25b482e90ff1f35a7fe003f : SUCCESS in 1h 12m 59s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c7908bc1afaa4607b51274b29be858c6 : SUCCESS in 42m 44s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8fb0365b9fd0427094f6ead285b671c2 : SUCCESS in 1h 13m 40s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e4903d6019f941f38f2be3e4f2904d11 : SUCCESS in 47m 23s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/42a85af45f2a47449321c8ee59ad3189 : SUCCESS in 29m 09s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/5737bfc7ffc64213858148f3cdf8ba74 : SUCCESS in 31m 28s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/80e11a5fd19d4e7db80f9b39d17301b8 : FAILURE in 2h 10m 45s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/815f6f32a3dd4e85b1c26ec37b11dcdc : FAILURE in 56m 43s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/02d748eca3d94142b20db27709d32086 : SUCCESS in 29m 50s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/42f05cf0847a422cba54ffead590b9f9 : FAILURE in 2h 10m 29s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a0d52a2ac7c14e7a8f566b55ada2fb77 : FAILURE in 37m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9aede199da6bc5dd4f84eff8ab0defe51fb071c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 17:13:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d138326f45cdcdf8c735d5b49dbb4026793093d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 20:52:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"59e2caf81042aee9b2eeb2eec93b329e4e2c9902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-23 23:35:16.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1b64e14e8d6325425e6ef24d2f36269997269034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 02:23:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567f114c80ef412b93badc59b1dedc43\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/b30c85d1a4dd44699887a7e103009231 : SUCCESS in 2h 45m 48s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/95648ff720a145ada5e484f736c68f74 : SUCCESS in 39m 03s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/452a8290d5e34fe9abd84209d6717090 : SUCCESS in 1h 13m 31s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f1267262116446c5984d01f2d60217de : SUCCESS in 28m 26s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a092f279cf764b4e96f652a3113b673f : SUCCESS in 42m 39s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/81b10e2a681a491f9530aa786190daed : SUCCESS in 55m 37s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c3c78713a14748c6b4bf75fd8e0ca891 : SUCCESS in 31m 13s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1ed54c91daa447e38b2ba25a1c03749c : SUCCESS in 1h 15m 26s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/491632b3e25f4b79913e945e28ba9fc3 : SUCCESS in 1h 15m 37s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/90126b26a1df4a789bd041b03c701a68 : SUCCESS in 36m 19s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/58110e1c169e4aac91d6a77ee0f9920c : SUCCESS in 42m 58s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/37ef4a0193114c3bba12a4a4e17e76b5 : SUCCESS in 1h 16m 39s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb64ff9d7bdb45b49ea81eebf00e8e0d : SUCCESS in 30m 26s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d9a35839065a4429a9507902565e11b6 : SUCCESS in 1h 06m 13s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/6724d8acd3004a72bc7db4169c7c3e2d : SUCCESS in 45m 27s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/48577c1432c64f11a100305e9202a16d : SUCCESS in 28m 59s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/40e9ccee744742e6802a5c5abf63f61d : SUCCESS in 36m 23s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9f788a6cdf2f46e1bab8e7c0e199bd08 : SUCCESS in 1h 18m 45s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/5643bb36dddb41d2a15901e03ed83f4d : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":4},{"id":"c31c1f43bc74b49e57837b868d01c6fd44e98891","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-01-26 16:41:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c3443f2a17dfb82b7c6033334aeed70bd76475a5","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-01-26 17:15:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b3e2bdfa01eb8e858a740b446161e8d42e125407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 17:15:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"62c546a9c03f2c18c0eb52433f12545a373dc65d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:48:30.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa444e003d34cdab2b2a6b89579afbe\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/70507500a5e54880a68cec7265888ebd : SUCCESS in 1h 32m 29s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/393c49dbf36b47ffacc3c36fb8f3dff4 : SUCCESS in 36m 06s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1399079082574a6eb562e0bc67feb706 : SUCCESS in 1h 15m 56s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/100c588f0df2490e9559eaadc972f443 : SUCCESS in 1h 04m 25s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9d44f41911df46c7bc5635889224a4e9 : SUCCESS in 1h 17m 43s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/0e9978b78f9c4521b471a84b3e9e9d1a : POST_FAILURE in 39m 37s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/311e52b37db34099aca82c1dd17a084f : POST_FAILURE in 38m 24s","accounts_in_message":[],"_revision_number":4},{"id":"071edb9208b0ac1a0ad5eddf7311fb4636009b24","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-26 19:01:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5fbde564d8d381c36f92a10b8910a35e5c660a96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 22:00:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fab6a7686d141f98f38831faba5f821\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/e381b1e246cb4bd498992b8c0e749af7 : SUCCESS in 2h 51m 42s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/91ee7fe2cbb44e849901dd49a3fba0f2 : SUCCESS in 1h 11m 29s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/66855f23c39d4ec8923416bfc198380b : SUCCESS in 1h 08m 39s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3d6e19b519dc425987b01f0a70b1a5ef : SUCCESS in 47m 19s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fe2638a844b64d8cae56dbaedd494f90 : SUCCESS in 1h 22m 10s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0a6ed34bba054e259b56fc9cff5af58a : SUCCESS in 59m 48s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/caacd4a2501249d682b5f86b9cb21400 : SUCCESS in 29m 07s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8e7361cbcd82429b9076c891f409ad28 : SUCCESS in 47m 47s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8c89fd1d0baf46ea91c599054472d4e3 : SUCCESS in 1h 01m 19s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7c2e8fcb00e74f15b67712bf2682142c : SUCCESS in 36m 09s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/aa10518e157845ad98830fb5bfeba577 : SUCCESS in 47m 55s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f75a3e70603746b8b7104ee88b4af7dd : SUCCESS in 1h 14m 21s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/42858c9aa0f2408b8b7645c62db40e86 : SUCCESS in 46m 17s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3f80a4e9064949f6a7e9592c2f63ae29 : SUCCESS in 1h 08m 45s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3e5d87b93bb34a8ebc1c42daaff433b1 : SUCCESS in 46m 33s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/1374016709984939864b07b2008ab6ff : SUCCESS in 1h 29m 26s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/b5999984c858464f95d02128bd845dae : SUCCESS in 1h 05m 45s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e93efa694fe9446593eaf0b4f2c1ac54 : SUCCESS in 1h 08m 05s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/c80685d2ccdd407eb7d058a7d18e07a0 : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":4},{"id":"8c4b3efc7d4f88c0917f028dc3280c2fb5157c5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 22:00:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3a8586c0b5d0c5d9f228b9c66f0114cfb1ef1b76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 23:25:20.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/842ef8c67ca643abbfe7736093980d7b\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/0d02f3376bf647b797feb8681c2d0976 : SUCCESS in 1h 19m 07s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e25e1a5e0c9f4cbc9e58b647f2ff8221 : SUCCESS in 56m 05s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cc40b50eabeb4745b1228105ab3e0075 : POST_FAILURE in 39m 09s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0f6ac49cfaa842d1abd05766d7bf2c0c : SUCCESS in 48m 28s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1ac1d5eef3e74508821fbb3a3223a508 : SUCCESS in 1h 04m 02s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e1c624c61e894991a25d0c43f5a8184f : SUCCESS in 54m 36s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/3d7492b544aa42cab7e217e4aab1cd8d : SUCCESS in 1h 10m 53s","accounts_in_message":[],"_revision_number":4},{"id":"96dad0de04a3e6542c4ea998f0687bd361492ee7","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-01-27 02:01:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca98daf9c6f9aa18904cf4412df02180e7845a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:51:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4269659a1af48e98e3ce7d5321a37b9\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/87af5d62245442be934e76f1ed184e55 : SUCCESS in 3h 04m 04s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/63007d458ae94f9eb100e846d09d2a4e : SUCCESS in 1h 10m 38s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/616d74e97ec94e3181ccbe60a91073bd : SUCCESS in 44m 42s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/697787e35e7c4ac783936506a8dc71cf : SUCCESS in 39m 47s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8ceaa3b08f204a04889a8f13175b3a89 : SUCCESS in 1h 33m 13s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2cc0c2df16bf4a0a9b46e4f135f76ded : SUCCESS in 1h 12m 02s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/9aacfc5cfa2f4e3584bd399d9108cdc8 : SUCCESS in 31m 48s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/460341ae725a4088bf7ad49b62f199e6 : SUCCESS in 46m 52s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5d6a3d95b0c443979c72e25058e96d5d : SUCCESS in 1h 17m 48s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/26a2430dc97f442d9472044d3cad7ee9 : SUCCESS in 50m 53s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b2317fd00d014ac986430f6bba4016cc : SUCCESS in 1h 05m 54s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/becb24c4a4a94943a54cbed5755e4d38 : SUCCESS in 1h 16m 48s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/49af1f1efe6b4e9aa2214f362bef8c46 : SUCCESS in 49m 17s\n- loci-build-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ae46331a23a04556bf6b1678ce72d96e : SUCCESS in 59m 34s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/983fee836d2b4af4a553ccb1cb29d0f9 : SUCCESS in 1h 17m 52s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d3b9a3f6b96e44c99596e2710df624ab : SUCCESS in 32m 56s\n- loci-build-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d1649e2be9b44b96944e4cdd1316cf5b : SUCCESS in 58m 18s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9848705b61514db39b4e49dbc66b8816 : SUCCESS in 1h 04m 47s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2690cdbc539c4413a6ff39916c00fd08 : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":4},{"id":"f74af27b933c9f96a47e886176f7bc00dd0c6475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:51:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b14fcd9f38767b44176b248652a2765b44282d26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:04:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01e9b54d2dc44b79ba60f36c587aefe4\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7740a42db1ec4418bad97e8e7e1d0aa1 : SUCCESS in 1h 12m 12s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/123401ea0437494392ace4deb7e252d1 : SUCCESS in 1h 07m 16s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/478aa7ae035a4e2fab498244e56fd4e4 : SUCCESS in 43m 34s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/924f76ecf4714c8990a6111eedecaded : SUCCESS in 45m 24s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/832a91be5a3e441091783f5f87d01399 : SUCCESS in 44m 13s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/c265ceae879c4cc5886e8354840e90fd : SUCCESS in 1h 05m 45s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e056871728aa44cba055d2bfd5080c61 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":4},{"id":"fc4b78729014c684d8be1f2cb560dba6d2fb590e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:04:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8eb76e2004e4071c7bd3cd28a44b50a3bc7ceae0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:13:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abd13db3ffa940d2aba99a57e317fc71\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/249c3a4f40e2414abfeef8df21150c1a : SUCCESS in 6m 31s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2a9e0c2c5be44a348961127052c08ed3 : SUCCESS in 7m 45s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e188e2bf0d1f40948e4b33c9b98763a1 : SUCCESS in 7m 46s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a3a4981caa83478ba5ef52327c076e73 : SUCCESS in 7m 46s\n- loci-promote-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/dc1289b290684040ab5732026a6ecfb8 : SUCCESS in 8m 49s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/77974c2d6f3d44f5a57ff20724b1a246 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bf95717ed892586caaf368512b5c682dd7a84674","revisions":{"dc635e7870930c8f227348959e217a34e996cf1f":{"kind":"REWORK","_number":1,"created":"2026-01-22 21:28:28.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/72/974372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/72/974372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/72/974372/1"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:28:15.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:28:15.000000000","tz":-360},"subject":"Add swift image","message":"Add swift image\n\nChange-Id: I661e750ac8f7503da7efad38c9f83811c4fd71bd\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dc635e7870930c8f227348959e217a34e996cf1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dc635e7870930c8f227348959e217a34e996cf1f"}]},"branch":"refs/heads/master"},"f6faad0c4de1d20613949a8d080f077deeb26e03":{"kind":"REWORK","_number":2,"created":"2026-01-23 17:13:54.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/72/974372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/72/974372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/72/974372/2"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:28:15.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 17:13:46.000000000","tz":-360},"subject":"Add swift image","message":"Add swift image\n\nChange-Id: I661e750ac8f7503da7efad38c9f83811c4fd71bd\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f6faad0c4de1d20613949a8d080f077deeb26e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f6faad0c4de1d20613949a8d080f077deeb26e03"}]},"branch":"refs/heads/master"},"74326789b5b8c134841291d254de273b7f48ecfc":{"kind":"REWORK","_number":3,"created":"2026-01-23 20:52:14.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/72/974372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/72/974372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/72/974372/3"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:28:15.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 20:52:08.000000000","tz":-360},"subject":"Add swift image","message":"Add swift image\n\nChange-Id: I661e750ac8f7503da7efad38c9f83811c4fd71bd\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/74326789b5b8c134841291d254de273b7f48ecfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/74326789b5b8c134841291d254de273b7f48ecfc"}]},"branch":"refs/heads/master"},"bf95717ed892586caaf368512b5c682dd7a84674":{"kind":"NO_CHANGE","_number":4,"created":"2026-01-23 23:35:16.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/72/974372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/72/974372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/72/974372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/72/974372/4"}}},"commit":{"parents":[{"commit":"5afdf8a312b318b05844da5e748bda2c5d315898","subject":"Merge \"Add masakari and masakari-monotors images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5afdf8a312b318b05844da5e748bda2c5d315898"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-22 21:28:15.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-01-23 23:35:07.000000000","tz":-360},"subject":"Add swift image","message":"Add swift image\n\nChange-Id: I661e750ac8f7503da7efad38c9f83811c4fd71bd\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf95717ed892586caaf368512b5c682dd7a84674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf95717ed892586caaf368512b5c682dd7a84674"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
