)]}'
{"id":"openstack%2Floci~977010","triplet_id":"openstack%2Floci~master~I16289bb335451c0734c0a52fa458e70ac8cf4837","project":"openstack/loci","branch":"master","topic":"nova","attention_set":{},"removed_from_attention_set":{"37622":{"account":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"last_update":"2026-02-24 07:56:13.000000000","reason":"Change was submitted"},"34520":{"account":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"last_update":"2026-02-24 07:56:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16289bb335451c0734c0a52fa458e70ac8cf4837","subject":"fix(nova): change default shell from rbash to bash for migration","status":"MERGED","created":"2026-02-16 17:01:33.000000000","updated":"2026-02-24 08:07:36.000000000","submitted":"2026-02-24 07:56:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977010-nova","meta_rev_id":"0aeb42ab327a6121686fd15b10accb7a8c4a1a87","_number":977010,"virtual_id_number":977010,"owner":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 07:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":2,"date":"2026-02-23 23:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-02-23 23:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2026-02-23 23:33:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 19:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 18:51:47.000000000","updated_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"reviewer":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"state":"REVIEWER"},{"updated":"2026-02-19 19:09:22.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"88b2a82218330afcd38d602a45f4d2f049946968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"date":"2026-02-16 17:01:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc0c30c7824e0c9922341eeb62d33015149665f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e118f84b862141ac9c9b84d67a307e4a\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/075e96088a5e4fcf8d1bd8ab720ffee1 : SUCCESS in 2h 44m 16s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a6dfd09bb642455fbdd2ce41d5fe29f2 : SUCCESS in 44m 09s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/930fdc4225e5483ab8f272dfbb116848 : SUCCESS in 47m 40s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/389e70d5c99e4933b5dc3bd9a7deae03 : SUCCESS in 37m 12s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c1108057196c469a811b0188ae2d7826 : SUCCESS in 1h 01m 51s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d4e86020db47457187587c970fed423e : SUCCESS in 1h 12m 43s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/adc887150c5642e68d816fc90454e960 : SUCCESS in 44m 00s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/65c1da40b5eb4fbe963fd50c1362f79c : SUCCESS in 1h 01m 57s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/32fadd4833db42b3bf1db4835df2ff73 : SUCCESS in 44m 46s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8c452ee1f0de4d5bae7ad1cfaec6af3c : SUCCESS in 48m 26s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a718b322527d4c6b9a2295ab934d11c7 : SUCCESS in 1h 05m 54s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e5f2c798b980444c86216ac46e2eca8e : SUCCESS in 1h 03m 12s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/46d45f62a152450a844399d58aaacd9a : SUCCESS in 27m 07s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e51525dc1a4a4945be08b7028479b258 : SUCCESS in 11m 58s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/d3b05978688845d08e1c882b587451bf : SUCCESS in 19m 40s\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/049472c1fb0b43c5882057f13d9c1d35 : SUCCESS in 22m 08s\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/e2769ab47edf4a848a1b8c7be0761bb5 : SUCCESS in 17m 33s\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/dde7f5df9a374d77902d0d56d37ca166 : SUCCESS in 17m 34s\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/f48600356d2a46ef9cfef88fefbc1631 : SUCCESS in 19m 37s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/75c480865d1a473f8a96671d6f6988cb : SUCCESS in 1h 18m 50s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/690bf3858cf84ecbabff318d28ea3377 : SUCCESS in 29m 23s\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/8285660da3a04666902625f2457c4549 : SUCCESS in 12m 14s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/a1213f1e72e442afa42fbef130395eea : SUCCESS in 22m 45s\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/2f7f02bff8dd41fdbfecef9bce1c040b : SUCCESS in 21m 51s\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/144cce48325c4bbea2e3a10ed933c233 : SUCCESS in 19m 29s\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/147e8b7cc1e94a1a830d80fc52763114 : SUCCESS in 16m 55s\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/ed5cc0314cba44e7800c873cdfbc9960 : SUCCESS in 17m 04s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/84721cbcd7ce457182716cd1b5995578 : SUCCESS in 1h 03m 43s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ec648ecc182f4fc09c673d5fe4d51bd7 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c912712a54ded3e3f012c50991027669c6e52668","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-19 18:51:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27d90d3bfea5c71aab6812759fcc93a564c0b5f6","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-02-19 19:09:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06a0d95aace16d5be846c3a0d456c4a140912db4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 19:09:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"963c9499eb6e98e32dfe1591c6f35b6fc0227dd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 20:35:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98f5522a0daf4dbb8c244e1b55723b6c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/529450b184494503b4db629a70be5c05 : SUCCESS in 1h 24m 48s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a8c599f40a5e4037a0d416302c018a9b : SUCCESS in 59m 40s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/c7956c4f633145ad8aae24f883ec5cfc : SUCCESS in 1h 08m 47s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/18f1aa4fcc524a5993712cef62bca74f : SUCCESS in 1h 16m 36s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ffbc61a5cbea474eb125501621d3c718 : SUCCESS in 1h 09m 10s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/90903a9bf1d8430c89c6e6f74ea31a94 : FAILURE in 7m 56s\n- loci-upload-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/08fa2ba4766f49dead09f292e0237284 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/78a5e7bf7ebc49c7a37e4fbf9b5f43e0 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/4e1c43d8a2c54ce49e9822dbbd77e588 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/043698d161ac4e76ab595ddc192e9ad8 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/f37d3421f18c4db3b08777b5f61d055d : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/f10e469848fa481ebc605c57cd54a711 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/dbbe2532b01b4d99951cfdc3d90838f1 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/da1a840c167b4a31acf3f9cce3783c59 : FAILURE in 6m 20s\n- loci-upload-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f4e695e454564aef89e1a43c666b73b4 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/bf514902eb1c4a7da8516f32c9def433 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/3d4b1f193645462991719ba72eddb9cb : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/f37e3d8c5913491aaf77c3074a6dc228 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/3f50c66568af4f45a9d72d0423d57c37 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/e877cd08e77b493aabe7fd7d2d6d4cb3 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/bb1fc4189ce54782b9f071d847c3459f : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":1},{"id":"c006d916ba98527ef994e8317a128da231a6b7d1","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-19 22:03:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"210185dd4e4f855f5fec41135b4f6c25977dbd68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 01:09:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99aad5576c34ca5b71344db7afefa58\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/15d64487860d4449abf05b6c29e155d7 : SUCCESS in 3h 04m 48s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0a816a8d5d394660a23d6349453eb2fd : SUCCESS in 30m 09s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f52448fb90b34f3085b777f47fd3de07 : SUCCESS in 1h 02m 33s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/571211fe1181413eb6591ee608894f4b : SUCCESS in 30m 21s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ed25f9216bb14480897852d105b579e8 : SUCCESS in 59m 46s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/059c49fc27aa42a180832f917667da46 : SUCCESS in 1h 14m 35s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4dcfda3c7aa04b219bf13e3448e7c8e0 : FAILURE in 59m 14s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/fb93ca18d72a4ab49fe19ee55f7c1467 : SUCCESS in 30m 24s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72a7bc06b4574b248739d93e3e0dddbb : FAILURE in 2h 29m 47s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/29bad1c82a734b27a49321f2b353d259 : SUCCESS in 46m 38s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7625fadbbe7542ea807f1f362f4a5a57 : SUCCESS in 33m 34s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1a0936e0d0c14596a058ca2c4f0fecba : SUCCESS in 50m 00s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/29d059f0f9204872bac2be7d03349ad8 : SUCCESS in 44m 48s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/94b3b52b18504ef68f65cf7c890b6e8c : FAILURE in 6m 30s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/101caeef5b1540c0bc8a4d0372ac88fe : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/f5dde04acf19456589471b9df3721aa6 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/9fe784202f6c48589a3885809faf25bb : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/106c53466938473284b4ea0ac7d8222d : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/c7957182d8564a359cfd5bcaeb867a5b : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d75cf7da81b5445ab781a729f092eb5c : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d8511485feac4238845ec5e2795cad2b : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/bff4d2759bf04fc69b1e7547c2bd266c : FAILURE in 9m 12s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/773646c0a8e24b8bb3262d3b97f4bd16 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/fc3aad5d2ece47da940b68176e76ceea : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/212b582c4a4b4bb3b40387f4e4d83926 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/54caef18801e44baa47679db7af9a94b : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/0f376d05330a42e0af295d2b9dd9843b : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/1f97a05d27e041c18078d76809272975 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/aa722ff5658c483bac00af10b00607e6 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":1},{"id":"ad64737bf005f996e59f433312b2c93a9a266c5c","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-20 03:04:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63cd1854f6a66a7c3f7287fc39da5c5c442c7678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 05:40:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/942627b5a011456fa112e7c2bbe87f99\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/11d7ff754c284125a6e388e8e17d0328 : SUCCESS in 2h 35m 35s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/0e0946f5dc7142dfbedc56a64b917f3b : SUCCESS in 52m 31s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3be032bbfd344d358c779fe62b6b2185 : SUCCESS in 1h 04m 22s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8722aae28cbf4814810803f639398cd7 : SUCCESS in 28m 53s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7cdc5fc7e3ed4e239763c04b91545fb9 : SUCCESS in 43m 49s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2be6cf0c30e94f08a46d2413878c6386 : SUCCESS in 48m 54s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/48d7dd9a2696498ebee72fd156b9e972 : SUCCESS in 45m 06s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72350a93288249cd912e35ca52c417cd : SUCCESS in 1h 09m 20s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3e0ebbe65c4d408c854f9e311d96e6a4 : SUCCESS in 1h 07m 50s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/feb6b1a4c70c4f27b155139b7e7a639c : SUCCESS in 38m 30s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/10227a0f2d3d4599bf772d51cab85fe4 : SUCCESS in 59m 29s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d1a6f16394014bdab0ecf833f6bedeb5 : SUCCESS in 1h 21m 55s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0ca8958291f2476cbe8d0fe9d7dd80fd : SUCCESS in 47m 34s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/194c90c21d8345a095fd6342803aedbe : FAILURE in 14m 12s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/068cc0b3226b4759924b15752c8daf03 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/76c3213cfe984798b9fe8826dfc5db10 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/08860bba23f84e71a78584b00ab2740a : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/e9853513e364481c9b7c31ef4c609c0d : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/7ed60d058d924e64832fa4ee761f9910 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/b930e5e677354007a75b0337c7e2f422 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f23e85b7f06b40a38e7fc025afc8c730 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/13ee1364e87f442ca68c47cb71807bea : FAILURE in 14m 29s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/aa554f8b98864e3d885ebbda288924b6 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/26269f945eeb45db936ac6d11b73ce4f : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/c916bb5e0b984e669ab618c78cfaa3ab : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/428406a5fa534942a0b2431e91babbfb : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/148930eb18a4427dba6d0f376c4aa060 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/07b9f17705f24b45ab3599f8b900f8fc : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/c6ae9b2a47b04ed5a6d85c4706429840 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":1},{"id":"692acbf432e7b2ff0cf22a31465c2d9c7edd8c97","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-20 15:50:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42984ccb12680694b7b9d7ed09f6b02f9a1639a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 18:26:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/299a812501fa496c9c65f401fdc72b3f\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/5b4cb4ade03c4788a3a17b91dc739360 : SUCCESS in 2h 34m 48s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fb4b7bae18364a499592ba13d968e336 : SUCCESS in 52m 53s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/030d3d49e6424bf1a656455102890905 : SUCCESS in 1h 16m 34s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/61c6b0825a194eb5886f7e55c4a7ea9a : SUCCESS in 40m 06s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3039983ee7a54a9fa7e5e440f3936834 : SUCCESS in 36m 13s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2223f131d9c54625a8f6afbed77d1c11 : SUCCESS in 1h 11m 26s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/91225a71bc5447cba9b77d9ecde1d7f8 : SUCCESS in 40m 06s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/684a3ce0b4994cfabcff209ba0f4577b : SUCCESS in 36m 07s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1ec9ddb7e37d4ebeb00c41955adb3166 : SUCCESS in 1h 19m 09s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9cd3eb93bf154f6fbf03b461bf0a69f4 : SUCCESS in 32m 07s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/cf0bf58bd17342c688ac4a81cea9b544 : SUCCESS in 32m 08s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/7661133bf0fc4bc091f4bd91f23f6957 : FAILURE in 1h 58m 11s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9810bd15fb52424184275af95c202bcc : SUCCESS in 29m 49s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ae940b2b166d4fbd9fd318316dfe2d4e : FAILURE in 13m 34s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/43b7c787d9e4409dbf5090fe6a8dbe9b : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/64e17fc60ecd43e7b80c3697adbdb398 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/fbd8ea9bd9da426e8eb92c93ea75e0b6 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/7bc399f494d846b392c7a90ed1a653de : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/0c472931859f4150ab2bb90e1fb5033f : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/26032f18ed484836a2bcc5979a40ad7c : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9e66ac1a72a642d6aa94dc1ce71339bc : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/697469f47d2c445297f739717241652e : FAILURE in 8m 27s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/49f5d7ee9b2d449b88bb193f59b7f40a : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/3e0623899db84b299e3cefab2df7d3f0 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/4252f47db87d4c7c8950c4ef3a623f80 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/d6cf2d80c7a74c578f733d38f5a38eb6 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/9dedd83d380043e5b851d56484cce975 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ce42b875b5f14a79bdc8493805bb961c : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d55375f6d3e44e2cb7006aea2598f19b : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":1},{"id":"3272f1ea6b121e654384287adee326149d3672b1","author":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"date":"2026-02-22 21:17:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08ff616c27070dba66510d8ba3312d91b8070795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 23:24:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85208696b7214fa8b2ae6d886501bf31\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7374afd6ad274bc98ef7bca6c59f99a2 : SUCCESS in 2h 06m 22s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/4f019bf18ff2452594d74ad43a47c6dc : SUCCESS in 26m 13s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/64e5764905bc4586b1f637ec1f4e3923 : SUCCESS in 1h 02m 40s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b383a459ca974be58fae9ba6faf6f967 : SUCCESS in 36m 07s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/a446b9d2958a4ac4aeb72eb6bab39bd4 : SUCCESS in 29m 02s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f80d9c3892a14e3a8de94ce985fe440e : SUCCESS in 1h 12m 22s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/ad3313199d3c4eb39f37803dccea4027 : SUCCESS in 39m 13s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9e425ec5ec3f4898af45fc8133f5979a : SUCCESS in 32m 30s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e124c8e66da3432182dd3d9971bb93bc : SUCCESS in 1h 18m 06s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9fe0d69431a84e8ba0946b7ac6d5f3e5 : FAILURE in 1h 10m 44s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/37c066272ba34dc4b636d4f658247ab0 : SUCCESS in 29m 11s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6784c302b5fc43108f5d1dca824e9c2d : SUCCESS in 1h 18m 09s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/63c54947d67849008fa7ea0d4102a017 : SUCCESS in 46m 00s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/43be56ae59ee496698e6be688285bcba : FAILURE in 15m 22s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/dda24609069f449b8de0127ecdd9006b : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/7d80c0129f9e407082e48517c3f8a833 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/2a39ec387237499ab6c46bafb5919613 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/668b50590c8845ca95d4056131a31100 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/ec9ea2fc40c24159b44b66c58aa26a0c : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e4a2f28efc2247bebd44111ac9c492cc : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/229f7c8f3d2b4a30899132e47c402061 : SKIPPED Skipped due to failed job loci-build-base-2025-1-ubuntu_noble_uv\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/55de82486a5c4ea7ba04c697139d06fc : FAILURE in 10m 56s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/f5d80d7cf61946bc9cd297d9c7f0c08b : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/8068f8cd9b984b8997fc8372ab4b08bb : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/8ca36d3e66354ebf84ddaeaa22469392 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/f5df8932d8604bbaaec26096fb340193 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/924eeb9addf143c4840948e91144dbe3 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/be87b9537d9042bd8a416e4f14dbefd9 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/91f99f8ef1384bbd9345ac1d2469d222 : SKIPPED Skipped due to failed job loci-build-base-2025-2-ubuntu_noble_uv","accounts_in_message":[],"_revision_number":1},{"id":"06a472ed32e9f73d3dbaf5d1bd3f8f4a71d538d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"real_author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-23 23:33:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_37622\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"}],"_revision_number":2},{"id":"1a2cfd951fa5ffc68d41d5410514a2df7e6f0f47","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-23 23:33:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"33f156dada8af64c56239698b19a7ebbf9662d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 03:14:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f39bf7a66ef42b4885bfe6f6f456dfa\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f5d77e5d3442459f9096561a7cb5f01c : SUCCESS in 3h 39m 25s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/cf63cd7205314e3ca4c676b2a45cd80e : SUCCESS in 47m 57s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/72e2fb8fa9e74208ab4aab3cf16c82bf : FAILURE in 1h 23m 52s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/93fc9cb093f64e92abfc2f4f94f164ac : SUCCESS in 28m 05s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/2e16812ce2ec4ba0b1759b7f8edd8bdc : SUCCESS in 1h 04m 39s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e7d1ef84b96949dfa70bd488ec7ea260 : FAILURE in 1h 46m 28s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/8a8f4393d7a04e52a6418e2d49a7f9f4 : SUCCESS in 36m 49s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d69771c3dbb14081b8af800505cf4d3a : SUCCESS in 1h 00m 22s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ba0a626cf34f4b68aa843611e3351855 : FAILURE in 1h 32m 22s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d9265ebd3c9747ea8fe7648a4fbbcd8d : SUCCESS in 47m 18s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/238b363a6c0547feb15d29fa89598dac : SUCCESS in 1h 05m 51s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72e551709a244eb4b509dd8a1a1cd598 : FAILURE in 2h 24m 22s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f26183ba9b7c41c5a21c244bcb4b5741 : SUCCESS in 35m 21s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ff0165f22c44413ea1edf7968ce2e905 : SUCCESS in 10m 07s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/7ac1c3b1d48f40be9f2bfd911ccc72d4 : SUCCESS in 21m 27s\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/ecfcbadc9e1a4f09a1578065ce281b8b : SUCCESS in 26m 40s\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/fdfb2c48adcc4b20b8569bf8ef45d784 : SUCCESS in 22m 38s\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/89a83e1e05db4a369140c1d0b1381f9c : SUCCESS in 21m 43s\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/afa7c779edf943f29066b9731fc0ade4 : SUCCESS in 21m 10s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f68a8669b1a144c8a34ae4312ce51cf4 : SUCCESS in 48m 42s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/58de32964a0d462e9a7b7b6f67e49b6a : SUCCESS in 35m 03s\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/1f449384d58145b683e6bf3ccafdd0c8 : SUCCESS in 12m 16s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/b2227af85fb14ad6bde319750e086190 : SUCCESS in 25m 55s\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/ac811fa91d374b418b3e949b889140e0 : SUCCESS in 21m 21s\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/c1f33c2c3b844225bff19891fd59f049 : SUCCESS in 21m 20s\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/b72da339bfbd4205881a27d42df458c4 : SUCCESS in 19m 12s\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/8b2885af9d67437bb28341ad101359ff : SUCCESS in 25m 00s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ee58126bb78c4ef7b9ff35e5dba1c569 : FAILURE in 1h 11m 32s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/8cbfdb7ffa7a41cbb591366ca12b1b38 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3e08e0befd3ceed85dd4a73a402d2403cb814773","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-02-24 03:57:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c812834db9591dc17391674efb4a5a0d8a68b247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 06:30:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb56bdc63cca4d22a55db682b103c215\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/90627f0864044c09a2e96da39021bdf5 : SUCCESS in 2h 33m 22s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/d1a3c2a0dc574c2a9c1501e2ddc14367 : SUCCESS in 58m 46s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b1ad2c202b2e4c52bb1d13b1ce6b1d68 : SUCCESS in 1h 15m 45s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/5d2cfd921105401f8f2da573472fb10e : SUCCESS in 30m 47s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e26ab758e73c4619bf26d3f63f691d8b : SUCCESS in 34m 45s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/6dfc19b0580747f98f8e22bfa5f15674 : SUCCESS in 1h 10m 44s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/78e8f61cdc494da8b3f8ab8b91f99d54 : SUCCESS in 37m 44s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8b1c1c563b594a328eb61bf10bfee18a : SUCCESS in 52m 08s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/30e2da88ad2f447e8b99f60bbdc2bfe8 : SUCCESS in 1h 18m 34s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/4349640b5bfa4327bb585fa977e57d77 : SUCCESS in 39m 43s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e2af5eafbc4548519154e9116a2c92a8 : SUCCESS in 1h 03m 45s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ebecf71d26184c348e213d3c9fffcf12 : SUCCESS in 1h 17m 16s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a867ac28f9ca4679a00dce9432fe5f8e : SUCCESS in 49m 04s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/fd476abab6eb4535960c1c6a590e3164 : SUCCESS in 11m 02s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/31e424312ed14fef88861c95cf8f7546 : SUCCESS in 22m 06s\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/bab360827f004203ae270d9c8d5f34ef : SUCCESS in 25m 31s\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/31ede2a800b942fb92d7378537587933 : SUCCESS in 22m 33s\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/63240c056b3a448792b5840360d747cf : SUCCESS in 20m 27s\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/d1fadd708e0c4902936a7d04fc4f7854 : SUCCESS in 18m 12s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/4376f5d8fc6a4d31b171c8d2808ea9b5 : SUCCESS in 47m 06s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d8c2194368fd4d5c974cc4c2b3786a7d : SUCCESS in 29m 43s\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/e9e7097b90424693833bcb91e89b71ce : SUCCESS in 10m 28s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/70ef9c79c6f04ea39f944ef72726726f : SUCCESS in 24m 07s\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/08c40614e9474361a13e5db654879b0f : SUCCESS in 19m 46s\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/12385424a972455887eb2e52d9796659 : SUCCESS in 24m 16s\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/2f43104f918f49e7b11175dea591145b : SUCCESS in 22m 29s\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/83fa8858473c4fdabefdcb80570473e9 : SUCCESS in 18m 29s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/bad6d3ff0d3b453b93a3597e00f5272c : SUCCESS in 47m 06s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/7c0155f3c4bf4b7e9e5343554c217b50 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8ad4bb81ed2091bae9472a11a78a013d10f59a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 06:31:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"66c1853a4604d02cc7c11478bfae237f124b528a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 07:56:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b291a4c1634b7eb74129024ff08e4e\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1729a7554eaa4d57abb1e896ada489b3 : SUCCESS in 1h 24m 47s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/b7c5ab21133c4752aed6effd5b2e9843 : SUCCESS in 1h 10m 26s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/96b6af69862044608c0607e60bda5f45 : SUCCESS in 1h 17m 12s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/b635e7b569b54d33985590571649f765 : SUCCESS in 1h 14m 27s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/45fef0ec5dce4eef8391b0ecb1256d43 : SUCCESS in 1h 14m 04s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/812d25660c084f11924fd4401cb23c45 : SUCCESS in 11m 37s\n- loci-upload-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/892431a205c54240abd3df5d53ca62c5 : SUCCESS in 2m 48s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/88ead7f83a1b4cea8c419a4f5b5e438e : SUCCESS in 26m 57s\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/cf4a18fb976a4a03a95f2ea5aca7bed4 : SUCCESS in 29m 26s\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/8286b5d74d2f43cea7dbf984baeb8989 : SUCCESS in 24m 44s\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/7a32fc891989497a97e3f5089bf1cbba : SUCCESS in 24m 33s\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/61954824de974c3f99e680b9fd8599c1 : SUCCESS in 23m 32s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9fb399b948c842f2806d219f7cf1d143 : SUCCESS in 10m 05s\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/1a86b30144d84e31a1c19d6818cb23b2 : SUCCESS in 10m 26s\n- loci-upload-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/325ad6d03af049bfa0b58a610ce4cb07 : SUCCESS in 3m 34s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/00f5204858374d449b2acb18bbd6ef70 : SUCCESS in 31m 02s\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/f71a29091f6147a58a74d8f71bc52352 : SUCCESS in 24m 06s\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/82918a1d319d4d0da3f6969d0e06952f : SUCCESS in 26m 27s\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/e2e49f8b2e6d4bf398ad671ba9ae68c8 : SUCCESS in 27m 08s\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/508c51707c394ceaacab247d4990dcb0 : SUCCESS in 22m 29s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/27082cbfca6b4d60a665abd5e924f4b7 : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"484b1f2f56a1a6fbc08189a2c645abd5fe2a078c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 07:56:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0aeb42ab327a6121686fd15b10accb7a8c4a1a87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:07:36.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bb67701eb72486ca05ec471b1e42093\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/27ede7922cb84e8489d5976c08ecf956 : SUCCESS in 8m 02s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/3e513a8fce1f423a93b314d0ccf236ba : SUCCESS in 9m 32s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c0ae4b130d6a48328932cb4fa043f2ce : SUCCESS in 9m 37s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8d3c5ce70f084b19a9d15ae5d1e7cac6 : SUCCESS in 10m 12s\n- loci-promote-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f5fa7ac204344009bd6705fdbba65121 : SUCCESS in 9m 04s\n- loci-promote-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a22688e56dc346feaacab683cb292e1c : FAILURE in 3m 40s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/ab7258ce0253438bbec062ac8f3834c9 : SUCCESS in 8m 06s\n- loci-promote-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/d316080793bb46bd9ad38d4d7e71623f : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6e7bdf3f38340ef2ed89ac70ff614326b4a4bb8","revisions":{"537592dcb8190525eeafb66051d8a62e48cef843":{"kind":"REWORK","_number":1,"created":"2026-02-16 17:01:33.000000000","uploader":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"ref":"refs/changes/10/977010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/977010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/977010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/977010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/977010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/977010/1"}}},"commit":{"parents":[{"commit":"f8b061f70c8bdcabb83cbfece80bd1ca32d07a98","subject":"Add necessary packages to Swift image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8b061f70c8bdcabb83cbfece80bd1ca32d07a98"}]}],"author":{"name":"Saeed Padari","email":"sam137115@gmail.com","date":"2026-02-16 16:58:34.000000000","tz":210},"committer":{"name":"Saeed Padari","email":"sam137115@gmail.com","date":"2026-02-16 16:58:34.000000000","tz":210},"subject":"fix(nova): change default shell from rbash to bash for migration","message":"fix(nova): change default shell from rbash to bash for migration\n\nThe restricted shell (rbash) prevented SCP/SSH commands needed\nduring instance migration with attached local disks using config drive.\nSwitching the Nova user\u0027s default shell to bash allows file transfers\nand migration operations to succeed without shell restrictions.\n\nChange-Id: I16289bb335451c0734c0a52fa458e70ac8cf4837\nSigned-off-by: Saeed Padari \u003csam137115@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/537592dcb8190525eeafb66051d8a62e48cef843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/537592dcb8190525eeafb66051d8a62e48cef843"}]},"branch":"refs/heads/master"},"c6e7bdf3f38340ef2ed89ac70ff614326b4a4bb8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-23 23:33:27.000000000","uploader":{"_account_id":37622,"name":"saeed padari","display_name":"Saeed Padari","email":"sam137115@gmail.com","username":"saeedpadari"},"real_uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/10/977010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/10/977010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/977010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/977010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/10/977010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/10/977010/2"}}},"commit":{"parents":[{"commit":"5a8a564338c94b3c4370bd4c3a59699ba27897b1","subject":"Use ipv4 addresses on buildkit arm worker nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5a8a564338c94b3c4370bd4c3a59699ba27897b1"}]}],"author":{"name":"Saeed Padari","email":"sam137115@gmail.com","date":"2026-02-16 16:58:34.000000000","tz":210},"committer":{"name":"saeed padari","email":"sam137115@gmail.com","date":"2026-02-23 23:33:27.000000000","tz":0},"subject":"fix(nova): change default shell from rbash to bash for migration","message":"fix(nova): change default shell from rbash to bash for migration\n\nThe restricted shell (rbash) prevented SCP/SSH commands needed\nduring instance migration with attached local disks using config drive.\nSwitching the Nova user\u0027s default shell to bash allows file transfers\nand migration operations to succeed without shell restrictions.\n\nChange-Id: I16289bb335451c0734c0a52fa458e70ac8cf4837\nSigned-off-by: Saeed Padari \u003csam137115@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c6e7bdf3f38340ef2ed89ac70ff614326b4a4bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c6e7bdf3f38340ef2ed89ac70ff614326b4a4bb8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
