)]}'
{"id":"openstack%2Floci~980304","triplet_id":"openstack%2Floci~master~I5f5658759c2a053bc2c982199d3d3baa4011dc72","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-03-16 23:54:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f5658759c2a053bc2c982199d3d3baa4011dc72","subject":"Add open-iscsi multipath-tools packages to Cinder image","status":"MERGED","created":"2026-03-12 15:11:02.000000000","updated":"2026-03-17 00:05:49.000000000","submitted":"2026-03-16 23:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980304","meta_rev_id":"a4db5afad0ad995409e2d9843d6ca09a9217d491","_number":980304,"virtual_id_number":980304,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-16 23:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2026-03-16 22:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2026-03-12 20:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2026-03-16 22:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 18:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-12 20:46:47.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-03-16 22:28:57.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"f447c1f32b6be18a5657af5ff009529ac656f5a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-03-12 15:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62a332a3d2e849a7bdead6942bbb0f43c9403b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:27:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f13934fb33cf4ab89f56709737501007\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/9f7bab300a1d4027b344d06472ebc3ce : SUCCESS in 3h 15m 30s (non-voting)\n- loci-build-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/1dea19a3e8ee484db7cdd5b3f916d9fc : SUCCESS in 52m 09s\n- loci-compute-kit-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/f0b1b5db6c134c72b9e3a0534d0dfca8 : SUCCESS in 47m 37s\n- loci-cinder-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/38a669efff2f42c1b717b03fb4c6b306 : SUCCESS in 41m 28s\n- loci-build-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/26ac6c134a294909be51965624702a3c : SUCCESS in 1h 02m 08s\n- loci-compute-kit-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/37392b0ec87e4289b2f17622659a8231 : SUCCESS in 1h 14m 40s\n- loci-cinder-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/e24dfd21e33e43d6ac19561eb9245f9a : SUCCESS in 28m 59s\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/a7ec1be028ba44ffbf4c8dd1c2093f5d : SUCCESS in 1h 28m 57s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ca4ab7bd004f409892c99c46d42486d6 : SUCCESS in 1h 21m 45s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/263a0108243e4551b0af4b1bbbdc740e : SUCCESS in 47m 13s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c53709b2a7574f68a6326710f6372663 : SUCCESS in 1h 04m 06s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c664954c4ed74cd7a15712269c9e9853 : SUCCESS in 1h 16m 47s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/e34fe3ac35ff4d4e92b92d91250ccce6 : SUCCESS in 31m 44s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/aed930edcb0143bc850fffde3b3af949 : SUCCESS in 12m 52s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/bfcce2e22a114d4596eb3eb5b49f7f60 : SUCCESS in 26m 45s\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/2774518f7831425fba9ebcebd1ac3b58 : SUCCESS in 27m 01s\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/d955ef34d5a8483aa13114746ba3f5f0 : SUCCESS in 23m 40s\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/7235af0a29524e7e9f648322f8635fe7 : SUCCESS in 23m 30s\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/059622db1b964059bda1d277f6839963 : SUCCESS in 21m 47s\n- loci-compute-kit-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/4ff25d6b64d540c0a453d5f72db1f6f6 : SUCCESS in 50m 58s\n- loci-cinder-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/07a7d301f8fc4ef0bdb7e76d1909285e : SUCCESS in 49m 25s\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/16aa0b146bf942a59a845196ed9766d9 : SUCCESS in 13m 58s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/fe3ac24ef4f846d18e300417a01e0c81 : SUCCESS in 28m 35s\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/3e6f6b369c1d4436b60de161fdf60f88 : SUCCESS in 19m 08s\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/627d0e8b43494b34a47b10648840f114 : SUCCESS in 23m 45s\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/d79feecb335041068fdcb419d93397ed : SUCCESS in 25m 01s\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/0de887349d604ae1ba4da5d97d320a90 : SUCCESS in 20m 00s\n- loci-compute-kit-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/f7523cb045314eb8aa517f640b3df95a : SUCCESS in 52m 14s\n- loci-cinder-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/2b453fccc7144db392ced15de8a031ba : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2371acb138eaefccc68d08f4cb337ccb98408665","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-03-12 20:46:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"251873642dcdbbcc065baa4fe789dffe14671617","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-03-16 22:28:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4647d7451d2b6354471ed9a608e0689cce5999f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 22:29:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76b4b102a3ceb00195ca0f5a2e305b9f8214431e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:54:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6957ce7e4dca4fadb7ee02df3df47d11\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/f54481be6fd441838098a6d3eaa19a9f : SUCCESS in 1h 20m 53s (non-voting)\n- loci-upload-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/431e17ef73f54b07801a4c959ff14ac3 : SUCCESS in 41m 09s\n- loci-upload-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/7f4ad27d1f4d494cb33752917f9a3b07 : SUCCESS in 1h 12m 47s\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6a2388375a9342f2823c431141c5e4ba : SUCCESS in 1h 08m 29s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/25625dd280804bf88856a13c5d327819 : SUCCESS in 1h 10m 31s\n- loci-build-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/c5a3aabf9b7c48b098bab092fa01cdc5 : SUCCESS in 10m 37s\n- loci-upload-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/9d03eb747fb2440a9984ea39f9a53f1e : SUCCESS in 3m 38s\n- loci-build-2025-1-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/6d476a7b5fdc40a1a8d0b3bfc59fa2b1 : SUCCESS in 27m 10s\n- loci-build-2025-1-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/7ccef1cf0e194adb9bdf5687bdb94c87 : SUCCESS in 25m 13s\n- loci-build-2025-1-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/81cc06d123f14a1e8c2a93a6d50fa40d : SUCCESS in 20m 01s\n- loci-build-2025-1-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/1b41292ef206460791daaf27c7cd0d2e : SUCCESS in 21m 23s\n- loci-build-2025-1-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/2d02c31ed14447eea697eb8dc62caafe : SUCCESS in 18m 31s\n- loci-upload-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/64c99d6cd8874d43bae7752d17d78f05 : SUCCESS in 23m 12s\n- loci-build-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/368aadc5c20342f88bfc9c156f49b15b : SUCCESS in 11m 10s\n- loci-upload-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a04fe29d2b06432f8c5f887cccd07a24 : SUCCESS in 3m 36s\n- loci-build-2025-2-ubuntu_noble_uv_0 https://zuul.opendev.org/t/openstack/build/25a2838c2abe42c0ae6f544a05c61cb5 : SUCCESS in 23m 29s\n- loci-build-2025-2-ubuntu_noble_uv_1 https://zuul.opendev.org/t/openstack/build/c036050968af40a2863278195e241724 : SUCCESS in 19m 20s\n- loci-build-2025-2-ubuntu_noble_uv_2 https://zuul.opendev.org/t/openstack/build/bb20db4098f940ff921b0effe9c3185b : SUCCESS in 29m 30s\n- loci-build-2025-2-ubuntu_noble_uv_3 https://zuul.opendev.org/t/openstack/build/79ddd172c05448ef8c61b75faa627bf6 : SUCCESS in 19m 03s\n- loci-build-2025-2-ubuntu_noble_uv_4 https://zuul.opendev.org/t/openstack/build/bbe1ba902d21426e9e5e5af35990ff11 : SUCCESS in 17m 13s\n- loci-upload-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/620caeb0e5c84bcab8d86e79ffe883d0 : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d4a04aa903cf0c1fceb6a39c9e0b5f9d3ca29aee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 23:54:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4db5afad0ad995409e2d9843d6ca09a9217d491","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 00:05:49.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11a4facfac08446ba00173e6d0b185bb\n\n- loci-promote-2024-2-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/aa9627d107364285a057fe3f34947fbc : SUCCESS in 8m 12s\n- loci-promote-2025-1-ubuntu_jammy https://zuul.opendev.org/t/openstack/build/fd6a7b09204a40b392965d94416da07e : SUCCESS in 8m 37s\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/677912b44f654e7694a3193d30401754 : SUCCESS in 9m 14s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/f0efb788e79243998015448fb7e88f17 : SUCCESS in 10m 02s\n- loci-promote-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/0c8409d72cfd43788d577fd6f61547be : SUCCESS in 10m 33s\n- loci-promote-base-2025-1-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a43b83f9e7ee40cebaefe9a724a8c748 : FAILURE in 3m 55s\n- loci-promote-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/1c435067eef248cfa4f51262a54fb78d : SUCCESS in 8m 53s\n- loci-promote-base-2025-2-ubuntu_noble_uv https://zuul.opendev.org/t/openstack/build/a7d1e96f5ebe4164bb94d89fd8dd75be : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a011153620e3c296209cdc4cad3c864ca90f95ed","revisions":{"a011153620e3c296209cdc4cad3c864ca90f95ed":{"kind":"REWORK","_number":1,"created":"2026-03-12 15:11:02.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/04/980304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/04/980304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/980304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/980304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/04/980304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/04/980304/1"}}},"commit":{"parents":[{"commit":"5a8a564338c94b3c4370bd4c3a59699ba27897b1","subject":"Use ipv4 addresses on buildkit arm worker nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5a8a564338c94b3c4370bd4c3a59699ba27897b1"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-12 15:10:35.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-12 15:10:35.000000000","tz":-300},"subject":"Add open-iscsi multipath-tools packages to Cinder image","message":"Add open-iscsi multipath-tools packages to Cinder image\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: I5f5658759c2a053bc2c982199d3d3baa4011dc72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a011153620e3c296209cdc4cad3c864ca90f95ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a011153620e3c296209cdc4cad3c864ca90f95ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
