)]}'
{"id":"openstack%2Floci~981828","triplet_id":"openstack%2Floci~master~Id1aca2646e733694835d282801eeb536e37aa6ab","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-03-27 21:48:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1aca2646e733694835d282801eeb536e37aa6ab","subject":"Use openstack-helm _loci jobs for testing images","status":"MERGED","created":"2026-03-23 19:57:26.000000000","updated":"2026-03-27 22:02:49.000000000","submitted":"2026-03-27 21:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981828","meta_rev_id":"463a969adb74376835dab00658388c69c0995545","_number":981828,"virtual_id_number":981828,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-27 21:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2026-03-27 15:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2026-03-27 20:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":1,"date":"2026-03-27 20:24:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 19:57:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 15:01:15.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"},{"updated":"2026-03-27 20:24:37.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"ac95bf85de74fc273428f5c60b4d3884a975cf13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-03-23 19:57:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3c1193f5b0e5c392324f5ddca9228db768c5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 19:57:46.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"3fc7aa8c07ebe08e0815f7899d1694745d8e3374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-03-25 22:07:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8a03dd411ad54c634bdee4ac5db4374f4cdfa2fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 00:13:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcf82fc607d246658be138a18f46f56a\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/45b218175d5344959228a2df37410c79 : SUCCESS in 2h 00m 46s (non-voting)\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5f4111aef1844471bf6ea816eeced85e : SUCCESS in 33m 58s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d869c715bfe4469cad47f01313c41d1c : SUCCESS in 1h 21m 31s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8759f4b0d00747118dde49e160214df9 : SUCCESS in 46m 13s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/2d38e9db432044abbb20f7a2ff950098 : SUCCESS in 31m 30s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d71784e988914b13a4aacb4e54b3cdf5 : SUCCESS in 1h 20m 07s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/dd1e57b6df314581871196e72cf0099e : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ee48f1bf97260c1a824e0130728b42ec6d39e2d0","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-03-27 15:01:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d260dfac6bc172f72b04f21fc4a8d6366bf5f896","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-03-27 20:24:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b35eb7c1c2f3459705f1e5d2f089eeb4e6d6e04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 20:24:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e8552e508ceef0b98959a8ac573335af08e8acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 21:48:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d72af3e19d4a74b26a06f55a0c4656\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/7bf69040e40140e38f1479f60f28afa9 : SUCCESS in 1h 19m 09s (non-voting)\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d579301b9c1b47f68aef7da7f0570b53 : SUCCESS in 1h 11m 19s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/98fd71a0c3ee468e9b407264ef6fec07 : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":2},{"id":"801f03ad537e6982ba8c5987ef43248b210df58a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 21:48:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"463a969adb74376835dab00658388c69c0995545","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 22:02:49.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0344745f531f4cdea25c5e7a824127ba\n\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d1a27e22db02402b89ad185c8306c618 : SUCCESS in 8m 54s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/31adfac425a0416486d772c84b3c5b38 : FAILURE in 8m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3069ce166cf8de406ff7bcd3039393e48d314e05","revisions":{"34cc1433082e5ece3820bd394875e32537345e89":{"kind":"REWORK","_number":1,"created":"2026-03-23 19:57:26.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/28/981828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/981828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/981828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/981828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/981828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/981828/1"}}},"commit":{"parents":[{"commit":"c071e9328b13b9c5c6aa8a9463e30402d79a4e1f","subject":"Merge \"Add open-iscsi multipath-tools packages to Cinder image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c071e9328b13b9c5c6aa8a9463e30402d79a4e1f"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 19:48:24.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 19:57:20.000000000","tz":-300},"subject":"Remove UV image build jobs","message":"Remove UV image build jobs\n\nWe are going to move building Openstack images\nto the openstack-helm-images repo and use the\ntraditional approach with per project Dockerfiles.\n\nThis repository will stay available for those\nwho would like to continue using the wheels image.\n\nThe images will be published to the registry\nwith the image tags\n\n\u003copenstack_release\u003e-ubuntu_\u003cubuntu_release\u003e_loci\n\ne.g. 2025.2-ubuntu_noble_loci\n\nIn openstack-helm we introduced the jobs and overrides\nnecessary to test images with this tag.\n\nPS also removes the jobs for 2024.2 Openstack release.\n\nDepends-On: Icef2f09afe7435fa29e6a6bc5c6837e7b0d932f2\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Id1aca2646e733694835d282801eeb536e37aa6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/34cc1433082e5ece3820bd394875e32537345e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/34cc1433082e5ece3820bd394875e32537345e89"}]},"branch":"refs/heads/master"},"3069ce166cf8de406ff7bcd3039393e48d314e05":{"kind":"REWORK","_number":2,"created":"2026-03-25 22:07:31.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/28/981828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/981828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/981828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/981828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/981828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/981828/2"}}},"commit":{"parents":[{"commit":"7c4693438d551bff719b466e89c357441f9d8ec7","subject":"Remove *_uv jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c4693438d551bff719b466e89c357441f9d8ec7"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 19:48:24.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-25 22:03:54.000000000","tz":-300},"subject":"Use openstack-helm _loci jobs for testing images","message":"Use openstack-helm _loci jobs for testing images\n\nBy default the openstack-helm project uses\nimages built in the openstack-helm-images repo.\n\nThis repository will stay available for those\nwho would like to continue using the wheels image.\n\nThe PS updates the image test jobs to use\n_loci overrides in the openstack-helm repo which\nwere introduced specifically for testing Loci images.\n\nDepends-On: Icef2f09afe7435fa29e6a6bc5c6837e7b0d932f2\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Id1aca2646e733694835d282801eeb536e37aa6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3069ce166cf8de406ff7bcd3039393e48d314e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3069ce166cf8de406ff7bcd3039393e48d314e05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
