)]}'
{"id":"openstack%2Floci~981829","triplet_id":"openstack%2Floci~master~I90a71bb043547eab52bbc6961220d32be987300f","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-03-24 20:16:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90a71bb043547eab52bbc6961220d32be987300f","subject":"Remove *_uv jobs","status":"MERGED","created":"2026-03-23 20:05:17.000000000","updated":"2026-03-24 20:34:03.000000000","submitted":"2026-03-24 20:16:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981829","meta_rev_id":"ad9be1084d568b4b1c0817a7b3c96f7a77bb4e67","_number":981829,"virtual_id_number":981829,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-24 20:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":2,"date":"2026-03-24 18:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":2,"date":"2026-03-24 17:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"all":[{"value":1,"date":"2026-03-24 18:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-23 20:05:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-24 17:27:23.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"},{"updated":"2026-03-24 18:43:52.000000000","updated_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"reviewer":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"state":"REVIEWER"}],"messages":[{"id":"6a310eafc966c699dfb44de2cd2539c6f93001fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-03-23 20:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7c19c186cfdecc4fd762d1de0c994cf8c14f33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 20:05:55.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/loci on branch master.  The\nproblem was:\n\n  Job loci-compute-kit-2025-1-ubuntu_noble_uv not defined\n\nThe problem appears in the \"opendev.org/openstack/loci\" project stanza:\n\n  project:\n      vars:\n        helm_version: \"3.18.1\"\n        container_command: docker\n      check:\n        jobs:\n          - loci-buildset-registry\n          # 2024.2 Ubuntu Jammy\n          - loci-build-2024-2-ubuntu_jammy\n  ...\n\n  in \"openstack/loci/zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":1},{"id":"8a5b07b6f506abeb388c266058989086a60a7c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-03-23 20:09:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"08a15b71d951266af9f979b05c8bf4f021de816f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-03-23 22:17:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8155b564495e6fef252e124ec2032dfb5ada67a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 00:52:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be3941d13a34020922f5eb9a9edc717\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/dbd7794dc5a144358215f5cf42a066d0 : SUCCESS in 2h 33m 41s (non-voting)\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1da21e1e51c948ff9f428f516468aadf : SUCCESS in 42m 58s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/72e0ebb0b7f24fcbbf92a576b6278ae8 : SUCCESS in 1h 19m 02s\n- loci-cinder-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/928ce107e70740be8f48b7a3e70404d1 : SUCCESS in 51m 28s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/90d15448f31046d6b51424d48261abc8 : SUCCESS in 1h 02m 12s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/be781439df6b45998e423662845f625d : SUCCESS in 1h 24m 02s\n- loci-cinder-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/5c1b4c80c44f4df3a0c1268ec057be08 : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c833c9e2494338fcb019bddb3a10b93ba4b933a0","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-03-24 17:27:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"10cd11e6cd3100ee9af2640cb82f30848477d88a","author":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"},"date":"2026-03-24 18:43:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"58d9a31c74ad029516bc7e352dc91b198b9a9ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 18:46:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e32a700014c3021d74cf2612108800ea14feac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 20:16:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff301c24ce945f590432298d86669fb\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/1878d181f1b8485db0e676ab0c50abfd : SUCCESS in 1h 24m 47s (non-voting)\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/0e9c72d5f1c24989bf9d7cef280ca0c7 : SUCCESS in 1h 14m 04s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/6f0a38a542de44d0952c6b212852f566 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":3},{"id":"db08246f16d046fe06e201d728c685492da35064","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 20:16:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ad9be1084d568b4b1c0817a7b3c96f7a77bb4e67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 20:34:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c31ffe2aa5548728b8e1b4117b03fa4\n\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3686838cc5294ac4a5c01aca54893fd3 : SUCCESS in 9m 54s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/c7d598fafcd845d2826601b7a395c400 : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c4693438d551bff719b466e89c357441f9d8ec7","revisions":{"36ee0c0ebd38b3ba07882d266e7382054d22eda8":{"kind":"REWORK","_number":1,"created":"2026-03-23 20:05:17.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/29/981829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/981829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/981829/1"}}},"commit":{"parents":[{"commit":"c071e9328b13b9c5c6aa8a9463e30402d79a4e1f","subject":"Merge \"Add open-iscsi multipath-tools packages to Cinder image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c071e9328b13b9c5c6aa8a9463e30402d79a4e1f"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 20:01:57.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 20:01:57.000000000","tz":-300},"subject":"Remove _uv deployment jobs","message":"Remove _uv deployment jobs\n\nWe are going to move image build jobs to\nthe openstack-helm-images repo and use separate\nDockerfiles for different projects.\n\nThe Loci images will be published to the registry\nwith the tags \u003copenstack_release\u003e-ubuntu_\u003cubuntu_release\u003e_loci\ne.g. 2025.2-ubuntu_noble_loci.\n\nThis PS removes dependency on _uv test jobs\nto make it possible to rename the _uv jobs to _loci\nand then use these jobs for testing Loci images.\n\nChange-Id: I90a71bb043547eab52bbc6961220d32be987300f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/36ee0c0ebd38b3ba07882d266e7382054d22eda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/36ee0c0ebd38b3ba07882d266e7382054d22eda8"}]},"branch":"refs/heads/master"},"8f37844dfd45afb9a09a53d12e4e6d517afeb940":{"kind":"REWORK","_number":2,"created":"2026-03-23 20:09:44.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/29/981829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/981829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/981829/2"}}},"commit":{"parents":[{"commit":"c071e9328b13b9c5c6aa8a9463e30402d79a4e1f","subject":"Merge \"Add open-iscsi multipath-tools packages to Cinder image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c071e9328b13b9c5c6aa8a9463e30402d79a4e1f"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 20:01:57.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 20:09:38.000000000","tz":-300},"subject":"Remove _uv deployment jobs","message":"Remove _uv deployment jobs\n\nWe are going to move image build jobs to\nthe openstack-helm-images repo and use separate\nDockerfiles for different projects.\n\nThe Loci images will be published to the registry\nwith the tags \u003copenstack_release\u003e-ubuntu_\u003cubuntu_release\u003e_loci\ne.g. 2025.2-ubuntu_noble_loci.\n\nThis PS removes dependency on _uv test jobs\nto make it possible to rename the _uv jobs to _loci\nand then use these jobs for testing Loci images.\n\nChange-Id: I90a71bb043547eab52bbc6961220d32be987300f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8f37844dfd45afb9a09a53d12e4e6d517afeb940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8f37844dfd45afb9a09a53d12e4e6d517afeb940"}]},"branch":"refs/heads/master"},"7c4693438d551bff719b466e89c357441f9d8ec7":{"kind":"REWORK","_number":3,"created":"2026-03-23 22:17:25.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/29/981829/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/981829/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/981829/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/981829/3"}}},"commit":{"parents":[{"commit":"c071e9328b13b9c5c6aa8a9463e30402d79a4e1f","subject":"Merge \"Add open-iscsi multipath-tools packages to Cinder image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c071e9328b13b9c5c6aa8a9463e30402d79a4e1f"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 20:01:57.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-03-23 22:12:58.000000000","tz":-300},"subject":"Remove *_uv jobs","message":"Remove *_uv jobs\n\nWe are going to move image build jobs to\nthe openstack-helm-images repo and use separate\nDockerfiles for different projects.\n\nThe Loci images will still be published to the registry\nwith the tags \u003copenstack_release\u003e-ubuntu_\u003cubuntu_release\u003e_loci\ne.g. 2025.2-ubuntu_noble_loci.\n\nThis PS removes all *_uv jobs. The Loci repository\nis going to stay maintainable for those users who would\nlike to continue using the wheels image.\n\nIn the subsequent PSs we will rename some jobs and overrides\nin the openstack-helm repo so we could continue testing Loci\nimages.\n\nChange-Id: I90a71bb043547eab52bbc6961220d32be987300f\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c4693438d551bff719b466e89c357441f9d8ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7c4693438d551bff719b466e89c357441f9d8ec7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29974,"name":"Stephen Taylor","email":"staylor@mirantis.com","username":"st053q"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
