)]}'
{"id":"openstack%2Fmagnum-tempest-plugin~802453","triplet_id":"openstack%2Fmagnum-tempest-plugin~master~I20bf1153fa67f06d2e82fe688b143c6b6c85740a","project":"openstack/magnum-tempest-plugin","branch":"master","hashtags":[],"change_id":"I20bf1153fa67f06d2e82fe688b143c6b6c85740a","subject":"[ci] Set GLANCE_ENABLE_QUOTAS to false","status":"ABANDONED","created":"2021-07-27 10:49:56.000000000","updated":"2021-09-17 12:35:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"69f01cf566a3a3d7487bf45af28bff14912c0521","_number":802453,"virtual_id_number":802453,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-27 14:51:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 11:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 14:55:29.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2021-07-27 14:55:29.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"}],"messages":[{"id":"74edaa55839a9e09b3ce74257d56cad7649b50af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-07-27 10:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b2d602ea7c3cd34d87f5e6871d681d0d6c4781f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-07-27 11:01:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3c89dd9a33e0166a921f53f04214b13644a37ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:32:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3577b3509e442e85fa560f764e4db0 : SUCCESS in 4m 23s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/1d795b35425240cc80bcced9819dd50c : FAILURE in 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6b1b8050f1c779efeeb9ef56808e85b65424d89f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-07-27 13:59:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"268305311b62beecdedff248976139093253bea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 14:51:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e504587ac6e47f4991b5e33f4e12ee9 : SUCCESS in 3m 39s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/b61de059534a469cbd4e9a9ca33f71fe : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":3},{"id":"1c885faf747bcf0da41a4cbeaa2a605170ed0c19","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-07-27 14:55:29.000000000","message":"Patch Set 3:\n\nWe need this to unblock CI and revert temporary change in Cinder patch","accounts_in_message":[],"_revision_number":3},{"id":"69f01cf566a3a3d7487bf45af28bff14912c0521","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-09-17 12:35:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7050c464e3578848fd20c6c0b87f68fb243b240","revisions":{"33315691637e57852113801828d8ce2dd167d23d":{"kind":"REWORK","_number":1,"created":"2021-07-27 10:49:56.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/53/802453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum-tempest-plugin","ref":"refs/changes/53/802453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/1"}}},"commit":{"parents":[{"commit":"7762c6e80102a27cfd01a22242c81533e3d04366","subject":"Merge \"Switch to new rolevar for run-temepst role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/7762c6e80102a27cfd01a22242c81533e3d04366"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2021-07-27 10:48:53.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2021-07-27 10:49:54.000000000","tz":0},"subject":"[ci] Set GLANCE_ENABLE_QUOTAS to false","message":"[ci] Set GLANCE_ENABLE_QUOTAS to false\n\nChange-Id: I20bf1153fa67f06d2e82fe688b143c6b6c85740a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/33315691637e57852113801828d8ce2dd167d23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/33315691637e57852113801828d8ce2dd167d23d"}]},"branch":"refs/heads/master"},"30030fc2fb0458e698d88d073eb015652c128b1f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-27 11:01:43.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/53/802453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum-tempest-plugin","ref":"refs/changes/53/802453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/2"}}},"commit":{"parents":[{"commit":"7762c6e80102a27cfd01a22242c81533e3d04366","subject":"Merge \"Switch to new rolevar for run-temepst role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/7762c6e80102a27cfd01a22242c81533e3d04366"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2021-07-27 10:48:53.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2021-07-27 11:00:48.000000000","tz":0},"subject":"[ci] Set GLANCE_ENABLE_QUOTAS to false","message":"[ci] Set GLANCE_ENABLE_QUOTAS to false\n\nSince devstack set this property to true [1], Magnum CI has been broken,\nlets unbreak it.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/788056\n\nChange-Id: I20bf1153fa67f06d2e82fe688b143c6b6c85740a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/30030fc2fb0458e698d88d073eb015652c128b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/30030fc2fb0458e698d88d073eb015652c128b1f"}]},"branch":"refs/heads/master"},"e7050c464e3578848fd20c6c0b87f68fb243b240":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-27 13:59:07.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/53/802453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum-tempest-plugin","ref":"refs/changes/53/802453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum-tempest-plugin refs/changes/53/802453/3"}}},"commit":{"parents":[{"commit":"7762c6e80102a27cfd01a22242c81533e3d04366","subject":"Merge \"Switch to new rolevar for run-temepst role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/7762c6e80102a27cfd01a22242c81533e3d04366"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2021-07-27 10:48:53.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2021-07-27 13:58:51.000000000","tz":0},"subject":"[ci] Set GLANCE_ENABLE_QUOTAS to false","message":"[ci] Set GLANCE_ENABLE_QUOTAS to false\n\nSince devstack set this property to true [1], Magnum CI has been broken,\nlets unbreak it.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/788056\n\nChange-Id: I20bf1153fa67f06d2e82fe688b143c6b6c85740a\nDepends-On: Ib66b754f4a0854a0d62d62047a69b04a24434634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/e7050c464e3578848fd20c6c0b87f68fb243b240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-tempest-plugin/commit/e7050c464e3578848fd20c6c0b87f68fb243b240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
