)]}'
{"id":"openstack%2Fmagnum-ui~740468","triplet_id":"openstack%2Fmagnum-ui~stable%2Ftrain~I7a03dfea2c84a423f15cfbe4194a8492f72dbd35","project":"openstack/magnum-ui","branch":"stable/train","topic":"story/2007697-39810-stable/ussuri-stable/train","hashtags":[],"change_id":"I7a03dfea2c84a423f15cfbe4194a8492f72dbd35","subject":"Fix cluster resize","status":"MERGED","created":"2020-07-10 13:45:59.000000000","updated":"2020-07-23 18:29:04.000000000","submitted":"2020-07-23 18:27:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740468-1595528876713-bd5f39d8","meta_rev_id":"6be38fd04a56c29065a302c14cf3ca67028e6336","_number":740468,"virtual_id_number":740468,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 18:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":2,"date":"2020-07-23 07:54:07.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":2,"date":"2020-07-23 18:13:34.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":1,"date":"2020-07-23 18:13:34.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 07:54:07.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2020-07-23 18:13:34.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2020-07-23 18:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"caa88ee0f806a87e6d32440b122cf78c9c7209b0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-07-10 13:45:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"26d40352833be2126755ca1a2e48f18e74014b19","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-07-10 13:46:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"354cbf6dfeb2664d229188d525a366b33793f07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:59:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4047a626b5e5494f948f5888c51f02fc : FAILURE in 7m 02s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/06cf6d4b90be4d4bab8e434de6edc82e : FAILURE in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d07f44a468c0484c9061323f35fc253d : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10cd8a1b24944f8aa6b12bf4977085c8 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/80be7f33287642d9866f8799a57b1e81 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/890e4de4be584255b3b075b0f424dd02 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78e9750001b74b3ba430d8ae5b6aeaa0 : SUCCESS in 5m 17s (non-voting)\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/01910d0c744a4be8b75242a8e690249e : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a544afff8d83b62d0c41df9267f914026b3fd159","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-07-23 06:35:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"01866e9ba0278cbbb6e237bcf8e3ec4675028863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 07:31:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/1d6dcbad054e43b2a3108993f49e6466 : SUCCESS in 5m 48s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/3ffa4703b8694187be7819c550d1e6de : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1b7aebb3e114983bc16408cb6839230 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c41a66e47b484ce3bb599bcba0a45718 : SUCCESS in 8m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c390d85507e4915a4b4fe3ab7d96453 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1da130f38b4bf78c496add75cfb236 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6d73209bbb949a2a7770a1327c5118a : SUCCESS in 7m 38s (non-voting)\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/0af9096d69c54e6391e93d968a8b78f0 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1620665780d731b8fc5c8ebf35a14cc8096186fd","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-07-23 07:54:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bb857ef1eda3b6c5754cf2589e22c5e738553a2d","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-07-23 18:13:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4c78b02dffc7a703c60f0abbb138492267c572fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:13:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da91a9eb2d71b7409d9dcc8d76a9004469d1300f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:27:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/36e9d5042bce47fba3095e91d0a3a969 : SUCCESS in 10m 08s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/0b8f7330c55c474a958c8b6664a6654e : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2eed7004c9c4e95a5f5200cc89d1a7b : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6f1aaf9646483c81754f484d889ee3 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ddcb8bfa93348b7ac9269bf90699fb6 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/361872c28e97405dab2930824992ab09 : SUCCESS in 4m 42s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/2b5f948d370d41108dd9adde84b315aa : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"02650071c03a7cfb5806929c757ebb55973c4f24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:27:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"65e0114e500aed1d03054c18a217971ef9172197","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:29:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0162c452a27443d6a949f70db1dabdbe : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7f07bd31c9924440dd9da4de51ec175d02b1ff22","revisions":{"b9a1ed24022e1cc2bcdbfacd7f1923cc1fb39349":{"kind":"REWORK","_number":1,"created":"2020-07-10 13:45:59.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/68/740468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum-ui","ref":"refs/changes/68/740468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/1"}}},"commit":{"parents":[{"commit":"2fd052efa8a5a7ec6669c6e13622213fc43f5721","subject":"Merge \"Add rolling upgrade ui\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/2fd052efa8a5a7ec6669c6e13622213fc43f5721"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-20 04:43:39.000000000","tz":720},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-10 13:45:59.000000000","tz":0},"subject":"Fix cluster resize","message":"Fix cluster resize\n\nThere are two fixes included in this patch:\n\n1. List nodes correctly. We have renamed minion to node, which\n   cause the filter doesn\u0027t work correctly.\n2. The response from magnumclient resize method is not a json,\n   so we need to convert it to dict before return. Rolling upgrade\n   has the same issue.\n\nTask: 39810\nStory: 2007697\n\nChange-Id: I7a03dfea2c84a423f15cfbe4194a8492f72dbd35\n(cherry picked from commit 96312eaf09dfec4424938a09d63a7190b2808d20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/b9a1ed24022e1cc2bcdbfacd7f1923cc1fb39349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/b9a1ed24022e1cc2bcdbfacd7f1923cc1fb39349"}]},"branch":"refs/heads/stable/train"},"71f84204d46d9e7b98b7c10e19c45b0594d655a0":{"kind":"NO_CHANGE","_number":2,"created":"2020-07-10 13:46:34.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/68/740468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum-ui","ref":"refs/changes/68/740468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/2"}}},"commit":{"parents":[{"commit":"2fd052efa8a5a7ec6669c6e13622213fc43f5721","subject":"Merge \"Add rolling upgrade ui\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/2fd052efa8a5a7ec6669c6e13622213fc43f5721"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-20 04:43:39.000000000","tz":720},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-10 13:46:34.000000000","tz":0},"subject":"Fix cluster resize","message":"Fix cluster resize\n\nThere are two fixes included in this patch:\n\n1. List nodes correctly. We have renamed minion to node, which\n   cause the filter doesn\u0027t work correctly.\n2. The response from magnumclient resize method is not a json,\n   so we need to convert it to dict before return. Rolling upgrade\n   has the same issue.\n\nTask: 39810\nStory: 2007697\n\nChange-Id: I7a03dfea2c84a423f15cfbe4194a8492f72dbd35\n(cherry picked from commit 96312eaf09dfec4424938a09d63a7190b2808d20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/71f84204d46d9e7b98b7c10e19c45b0594d655a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/71f84204d46d9e7b98b7c10e19c45b0594d655a0"}]},"branch":"refs/heads/stable/train"},"7f07bd31c9924440dd9da4de51ec175d02b1ff22":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-23 06:35:16.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/68/740468/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum-ui","ref":"refs/changes/68/740468/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum-ui refs/changes/68/740468/3"}}},"commit":{"parents":[{"commit":"304866417dd09bb9d3da238c4305d05b56a9bc56","subject":"Fix npm jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/304866417dd09bb9d3da238c4305d05b56a9bc56"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-05-20 04:43:39.000000000","tz":720},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-23 06:35:16.000000000","tz":0},"subject":"Fix cluster resize","message":"Fix cluster resize\n\nThere are two fixes included in this patch:\n\n1. List nodes correctly. We have renamed minion to node, which\n   cause the filter doesn\u0027t work correctly.\n2. The response from magnumclient resize method is not a json,\n   so we need to convert it to dict before return. Rolling upgrade\n   has the same issue.\n\nTask: 39810\nStory: 2007697\n\nChange-Id: I7a03dfea2c84a423f15cfbe4194a8492f72dbd35\n(cherry picked from commit 96312eaf09dfec4424938a09d63a7190b2808d20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/7f07bd31c9924440dd9da4de51ec175d02b1ff22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum-ui/commit/7f07bd31c9924440dd9da4de51ec175d02b1ff22"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
