)]}'
{"id":"openstack%2Fmagnum~691721","triplet_id":"openstack%2Fmagnum~master~I2ce19b3f976619853117231ae0dae24ebe68601f","project":"openstack/magnum","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I2ce19b3f976619853117231ae0dae24ebe68601f","subject":"Drop python2 tests","status":"MERGED","created":"2019-10-28 16:57:52.000000000","updated":"2019-11-21 21:18:17.000000000","submitted":"2019-11-21 21:15:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691721-1574370920189-8d7acffc","meta_rev_id":"3018eb4081f633ec5e89c2825762abc0924a0efa","_number":691721,"virtual_id_number":691721,"owner":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-21 16:29:30.000000000","_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 21:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-21 17:14:56.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":2,"date":"2019-11-21 16:40:34.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-11-21 20:34:15.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":2,"date":"2019-11-21 16:42:07.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-11-21 20:34:15.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 02:34:04.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-11-21 16:40:34.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2019-11-21 16:42:07.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2019-11-21 17:14:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-11-21 20:34:15.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2019-11-21 21:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffebda8075ede7492a625967478dc866ea86d5aa","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-10-28 16:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"668cfc8cace361e5bc210819936dcd23fb55c7bc","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-10-28 17:00:00.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"12adb3e40ac1a2b0f2ef57820b166c464fd8964a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:37:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0ba492a07f0742419e66fa961ef19e48 : SUCCESS in 3m 50s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/efdc03a41f17433abb4ad5e29a498d93 : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb0bc327ec55448cab1a30b5a40107e3 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a72bce68dc194c2b998bf461c839b75c : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e2753f1c89450e959a5fffa9ae7a02 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fd83d986e2d464ba049b288bc0cbf32 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38bb9271f87f41d49ec2e11127d9c753 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/634dccf640e7445993b5e410210444f5 : SUCCESS in 5m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca7cd46857284c39afa8b75c55b82df5 : FAILURE in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c25c68d31fa47328d8baf610c11814d : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/629f919210514524847d2878a6d011f3 : SUCCESS in 6m 48s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/4f46b7ce19b44567bdf2a8f3c502f547 : SUCCESS in 36m 44s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/639b4ac92d9241ccb651bf88944b550a : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e426ac42f93de4d42d9a6bc41852cd1baff08871","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-10-28 17:43:29.000000000","message":"Removed Code-Review-1 by Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"84d4c7efdfb3623db2e1ee2b2d7432959f907f55","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-10-28 17:43:31.000000000","message":"Removed Workflow-1 by Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"fb8e39e80e6f1ff3363dd7bc01861d4e8c1b9aa7","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-12 02:34:04.000000000","message":"Patch Set 1: Code-Review-1\n\nrequirements-check need be fixed first, more info https://zuul.opendev.org/t/openstack/build/ca7cd46857284c39afa8b75c55b82df5/log/job-output.txt#455","accounts_in_message":[],"_revision_number":1},{"id":"d8720f132a9551552b145ce6169630239ad3cd38","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 12:37:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nrequirement-check job has been adjusted not to check the py2 req\nyou can make Depends-On: https://review.opendev.org/#/c/693631/\n\nAlso add reno to drop the py2 support\n- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria","accounts_in_message":[],"_revision_number":1},{"id":"2d70293ab8232601e15c1ed4b4644720aadc0d88","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-15 12:37:49.000000000","message":"Topic changed from dropPython2 to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"692865f87f115b4551ff7e9cab5315092c28282c","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-11-18 14:18:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85e1ac969f8744020de7efd23ea3414e0e0071d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 15:12:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8dfbbfe5b0344300969d68378b69bcb7 : SUCCESS in 5m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76c206fd7c264ac2b0d23d9dd1999790 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27a236064e8044008c2376af18ae2af1 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b9d2c83ff7642f695e8c80a0c195607 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d75a96ed77447938f94e3d1610d2d0e : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c504bcf246294f1596e6844cf8d64174 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6b1b0627fd463a85c6185caaa22f22 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce9e22ac19494ca5ad22fcbc9c55dbfa : SUCCESS in 8m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4eb94aa28294e77adc8e555c0c46eee : FAILURE in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256a2110f3be4727959ee6353b3b880a : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afcd21ad62504a109d6bbd87d3b870c8 : SUCCESS in 7m 08s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/bb6cfe6197df4354891b36e4e901fdd5 : SUCCESS in 43m 02s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/94fae4279be841f6a39639f922572b07 : SUCCESS in 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e6976f359f21bcff0e025067bfa0a91ddac4552c","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-11-19 14:34:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"259b55d4be305b9f035e02558c4e4a45f820abda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 15:19:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/141929734f53443f8bcb2b7ee9b9c562 : SUCCESS in 3m 51s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6d21063bfcf45898410118e46596ea1 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c061598a60894a48a6d8c6d88fded4ea : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6789c2158c1b485087a31c95e1e6f920 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a386a869e8447482c607f4d8a50d1f : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7884df070e844a08d40e81b79ea569d : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/761bc3cd89d84ed69e3140320acc5f7e : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/931197de28bb40e18f07d150cb9f4d48 : FAILURE in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6c04272a7254ead91b9d169c462705c : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46bb75edbfd347d3a38957bb7ace534e : SUCCESS in 7m 02s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/5731f62a00d74db4ac97367d0f265f36 : SUCCESS in 44m 01s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/55bba293d9164d1d9056f2a72fa41f90 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":2},{"id":"60d8189452c6c1c7beb22546b07a74104277d98f","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-11-19 15:41:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a90fda1eb7bd428f193779b2990219c4f3abddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 16:32:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bc172c3c740f47c39a219de5294e8982 : SUCCESS in 4m 25s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed71c1e425d14f1e8ee400b9985ea1b1 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f943e56719645a8bed9cf71ac01206f : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46316ad74860435ca34703c77be8e487 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9ec16b3bf264ec6a7e36e032c1d0427 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bca6959c495c46bdba61b6d371b82e1f : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c969571c891a4e06b2c14c821f4cd35e : SUCCESS in 5m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/120359ae46344e1c94d600bb36b2ac7c : FAILURE in 2m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0996e09b64ce4001a3f3fe0d23e36f8a : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ec6f799e5984200af74fdbcbe89c2ba : SUCCESS in 7m 33s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/5cf1f2baa4764838ac182a71a0bb079d : SUCCESS in 45m 41s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/faaff040516b4e9588ab3ac847f2a2c7 : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":3},{"id":"07314472c77e033a83c2d3749064161779651cf6","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-11-20 21:35:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"97f6149ff2d2711b7bc44405cc0a93d6f886c0bf","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2019-11-21 13:13:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea68463d7b9b7a941b4d67642ce5b616bf7319a0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-11-21 13:14:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d5b4b0100764997e7f368f1470af93e14d3e9c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 14:10:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/282520beb92a4d11b2b47f1405f2f969 : SUCCESS in 4m 12s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a94d706eea7e42e1ab463c4dd7dc2c88 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b67b0d0c51e425fb0a25d3c7045c041 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50658154882b42899225f2a8a932d2f1 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5cfdfe9db343b7a69f3b613203fd97 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ce551fca9246d8988aab20fa89f1be : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a40887cd8efd42cca017373ba2a1ab46 : SUCCESS in 6m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e5bd56cee6f47c88f1f2d18d967da48 : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72325e50c56b4a26ae993fe738f425ed : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/425a989ce2194862a972f334cebec168 : SUCCESS in 7m 21s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/2848c8b98f5640edb5f76815b829c29b : SUCCESS in 45m 06s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/80975261b2ec4e35b7f19d4da17ccd04 : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":4},{"id":"05c2d76ca2f59d44a50ed6f9821594332d5553dd","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-11-21 14:20:29.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"56cb360fa17104d5bd2f45644af8bfdc67f605cd","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-11-21 14:58:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e9578fd720620e95c5696f90a9d34ac27b9fdf7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 15:59:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67945dd1d0084447b5ca797a52c19878 : SUCCESS in 4m 46s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fc7dfa2799914dc9b816189ffdc202a3 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d40ebef5ac144eb59d4656dd5cd24eea : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd1e3c4830674cb19b93fe791e03a5d1 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef7d6550e84749079f7d62f6d17e7747 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b5c44cebc0e42469545bcbce081aead : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d514938a0ffc4908b78dc2950505e1ee : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6fd32c5f54e43c3babeef9d3a1e81e3 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7eb377141b24522a134a4a55f86f83a : SUCCESS in 7m 27s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/4e67aa2687974f87b1b4b250a2c71d26 : SUCCESS in 52m 48s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/2f06766c112d4a27aa6dfe002eebd9f8 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":5},{"id":"f7e95182c37b7abffba6c4308540c38c7f9a2715","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2019-11-21 16:29:30.000000000","message":"Patch Set 5:\n\nflwang: i could not find a way to disable the openstack-tox-py27 to non-voting.\n\nMy understanding is that it comes from the templated dependencies.\nI tried to check zuul docs, doesn\u0027t have nothing about disabling templates.\nOnly thing would be to rewrite the tox.ini somehow i guess\n\nAlthough, python27 breaks on authing in the CI machine, so they disabled it already. As the other stable/ branch won\u0027t merge this, they will still validate python2\n\nI checked and nova dropped everything related to python2 anyaways","accounts_in_message":[],"_revision_number":5},{"id":"6b2ad85cd2dff9c1bc7cbdc97361d3783d3500b0","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2019-11-21 16:40:34.000000000","message":"Patch Set 5: Code-Review+2\n\nWaiting for flwang,\n\ngmann, Any input?","accounts_in_message":[],"_revision_number":5},{"id":"c01aa971a1dcd25d975ab23e0d33c2e456343379","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-11-21 16:41:47.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bb1e70d4b5e63b71509c881a1d4b18d8c1f3887f","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-11-21 16:42:07.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c57e296ba5874ca5a2b4dc8a7303edd44daa6265","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 17:14:49.000000000","message":"Patch Set 5:\n\n\u003e (1 comment)\n\nall good, let me verify the experimental jobs also as we have defaulted the Devstack to py3 now so all existing jobs should move to py3.","accounts_in_message":[],"_revision_number":5},{"id":"a0d7dda65e8db6d17235f5941f1a31402d039bc3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-11-21 17:14:56.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"3c0a83f5c67314f7cec6276c86d3ac41c977ace1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 19:08:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (experimental pipeline).\n\n- magnum-functional-dcos https://zuul.opendev.org/t/openstack/build/ef0f77d638fc404583a635f8680bcf0c : FAILURE in 33m 09s (non-voting)\n- magnum-functional-mesos https://zuul.opendev.org/t/openstack/build/e73cac3028484851a8abed701215e056 : FAILURE in 1h 05m 29s (non-voting)\n- magnum-functional-swarm https://zuul.opendev.org/t/openstack/build/0ccb7cfd95544f7e9887b8666556c6ce : FAILURE in 1h 50m 31s (non-voting)\n- magnum-functional-k8s-ironic https://zuul.opendev.org/t/openstack/build/619a3d33ad6546afb7803ad549131e56 : FAILURE in 36m 45s (non-voting)\n- magnum-functional-swarm-ironic https://zuul.opendev.org/t/openstack/build/e0be3c0971ce480dbdfa265e4cfef832 : FAILURE in 33m 06s (non-voting)\n- magnum-functional-k8s-multinode https://zuul.opendev.org/t/openstack/build/ef646593db224fd49863c84d69d65d59 : FAILURE in 11m 52s (non-voting)\n- magnum-functional-swarm-mode-multinode https://zuul.opendev.org/t/openstack/build/24d986722d4c4c9986718aee155f4506 : FAILURE in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"69cfc7e518327f05f028d487cc5c93c6c124ad43","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2019-11-21 20:34:15.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3c86de2ac2f85d804e436d9c63fd32e01e2241c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 20:34:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8b05dad7e1ef3e62a2ad63e22ff4224d25149f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:15:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5c7fe2e344bc4661a085f2bc953ae12d : SUCCESS in 4m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af176ab73c89442d90bf033e40068406 : SUCCESS in 2m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04e79cd8360d4195bceca0d354380ba1 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5395ced47a3a44bcbb74c5b3812d070f : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f2245eee3c0403aa3d16ad831f955af : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/443348cc49264147a64a25e182d3c90f : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f2d922debd4d7eb764961098abbc98 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce920182a2dc425385b26ea715bc6028 : SUCCESS in 7m 03s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/9a60dd79215b42ce86d8318feede63ef : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":5},{"id":"7fbb8068bb33f92a9d58a771c870a7a45e329d14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:15:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3018eb4081f633ec5e89c2825762abc0924a0efa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 21:18:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/de7c8b635369423585bffc7cc4692e82 : SUCCESS in 49s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3dfc231c3e48c58d74bf76608aaa2a : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e13c05008f04574a3af7dc40a4090b1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3d17c6eb4959e5ec1d07e0cc49bd042ae9a89dc0","revisions":{"a8ab195ee41d0803556b4823cf8e97dcc0c1f42e":{"kind":"REWORK","_number":1,"created":"2019-10-28 16:57:52.000000000","uploader":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"ref":"refs/changes/21/691721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/691721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/691721/1"}}},"commit":{"parents":[{"commit":"9c7cf5d6eb050dbbcdcb036bd0af2b6586070c0c","subject":"Merge \"coreos: Use heat params for heat-agent image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/9c7cf5d6eb050dbbcdcb036bd0af2b6586070c0c"}]}],"author":{"name":"Diogo Guerra","email":"dy090.guerra@gmail.com","date":"2019-10-28 16:42:27.000000000","tz":60},"committer":{"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","date":"2019-10-28 16:56:51.000000000","tz":60},"subject":"Drop python2 tests","message":"Drop python2 tests\n\nChange-Id: I2ce19b3f976619853117231ae0dae24ebe68601f\nSigned-off-by: Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a8ab195ee41d0803556b4823cf8e97dcc0c1f42e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a8ab195ee41d0803556b4823cf8e97dcc0c1f42e"}]},"branch":"refs/heads/master"},"dfe0395165ab20f6dd70470834019d329563a470":{"kind":"REWORK","_number":2,"created":"2019-11-19 14:34:31.000000000","uploader":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"ref":"refs/changes/21/691721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/691721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/691721/2"}}},"commit":{"parents":[{"commit":"b07f8ba86127df4b2db976041be55916a6d15837","subject":"Merge \"Support TimeoutStartSec for k8s systemd services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b07f8ba86127df4b2db976041be55916a6d15837"}]}],"author":{"name":"Diogo Guerra","email":"dy090.guerra@gmail.com","date":"2019-10-28 16:42:27.000000000","tz":60},"committer":{"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","date":"2019-11-19 14:34:25.000000000","tz":60},"subject":"Drop python2 tests","message":"Drop python2 tests\n\nChange-Id: I2ce19b3f976619853117231ae0dae24ebe68601f\nSigned-off-by: Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/dfe0395165ab20f6dd70470834019d329563a470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/dfe0395165ab20f6dd70470834019d329563a470"}]},"branch":"refs/heads/master"},"b4235ea61f22c1e22ae3bcda69221f5627d081c6":{"kind":"REWORK","_number":3,"created":"2019-11-19 15:41:20.000000000","uploader":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"ref":"refs/changes/21/691721/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/691721/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/691721/3"}}},"commit":{"parents":[{"commit":"b07f8ba86127df4b2db976041be55916a6d15837","subject":"Merge \"Support TimeoutStartSec for k8s systemd services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b07f8ba86127df4b2db976041be55916a6d15837"}]}],"author":{"name":"Diogo Guerra","email":"dy090.guerra@gmail.com","date":"2019-10-28 16:42:27.000000000","tz":60},"committer":{"name":"Diogo Guerra","email":"dy090.guerra@gmail.com","date":"2019-11-19 15:41:09.000000000","tz":60},"subject":"Drop python2 tests","message":"Drop python2 tests\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\nmagnum is ready with python 3 and ok to drop the\npython 2.7 support.\n\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria\n\nDepends-On: https://review.opendev.org/693631\n\nChange-Id: I2ce19b3f976619853117231ae0dae24ebe68601f\nSigned-off-by: Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b4235ea61f22c1e22ae3bcda69221f5627d081c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b4235ea61f22c1e22ae3bcda69221f5627d081c6"}]},"branch":"refs/heads/master"},"1587b630f6b1453800e8a6c7bc9d073890ee9c97":{"kind":"REWORK","_number":4,"created":"2019-11-21 13:14:54.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/691721/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/691721/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/691721/4"}}},"commit":{"parents":[{"commit":"b07f8ba86127df4b2db976041be55916a6d15837","subject":"Merge \"Support TimeoutStartSec for k8s systemd services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b07f8ba86127df4b2db976041be55916a6d15837"}]}],"author":{"name":"Diogo Guerra","email":"dy090.guerra@gmail.com","date":"2019-10-28 16:42:27.000000000","tz":60},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-11-21 13:14:50.000000000","tz":0},"subject":"Drop python2 tests","message":"Drop python2 tests\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\nmagnum is ready with python 3 and ok to drop the\npython 2.7 support.\n\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria\n\nDepends-On: https://review.opendev.org/693631\n\nChange-Id: I2ce19b3f976619853117231ae0dae24ebe68601f\nSigned-off-by: Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1587b630f6b1453800e8a6c7bc9d073890ee9c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1587b630f6b1453800e8a6c7bc9d073890ee9c97"}]},"branch":"refs/heads/master"},"3d17c6eb4959e5ec1d07e0cc49bd042ae9a89dc0":{"kind":"REWORK","_number":5,"created":"2019-11-21 14:58:46.000000000","uploader":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"ref":"refs/changes/21/691721/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/691721/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/691721/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/691721/5"}}},"commit":{"parents":[{"commit":"b07f8ba86127df4b2db976041be55916a6d15837","subject":"Merge \"Support TimeoutStartSec for k8s systemd services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b07f8ba86127df4b2db976041be55916a6d15837"}]}],"author":{"name":"Diogo Guerra","email":"dy090.guerra@gmail.com","date":"2019-10-28 16:42:27.000000000","tz":60},"committer":{"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","date":"2019-11-21 14:58:32.000000000","tz":60},"subject":"Drop python2 tests","message":"Drop python2 tests\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\nmagnum is ready with python 3 and ok to drop the\npython 2.7 support.\n\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#completion-criteria\n\nDepends-On: https://review.opendev.org/693631\n\nChange-Id: I2ce19b3f976619853117231ae0dae24ebe68601f\nSigned-off-by: Diogo Guerra \u003cdiogo.filipe.tomas.guerra@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/3d17c6eb4959e5ec1d07e0cc49bd042ae9a89dc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/3d17c6eb4959e5ec1d07e0cc49bd042ae9a89dc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
