)]}'
{"id":"openstack%2Fmagnum~706721","triplet_id":"openstack%2Fmagnum~stable%2Ftrain~I970c1b91254d2a375192420a9169f3a629c56ce7","project":"openstack/magnum","branch":"stable/train","topic":"image-digest-stable/train","hashtags":[],"change_id":"I970c1b91254d2a375192420a9169f3a629c56ce7","subject":"Support verifying the digest for hyperkube image","status":"MERGED","created":"2020-02-09 21:15:01.000000000","updated":"2020-02-10 14:03:25.000000000","submitted":"2020-02-10 14:00:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706721-1581343253817-71032318","meta_rev_id":"d11eac75042e964ce055e24ba875897a7bc610d9","_number":706721,"virtual_id_number":706721,"owner":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"date":"2020-02-09 21:16:39.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 14:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":2,"date":"2020-02-10 12:51:36.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":2,"date":"2020-02-10 11:16:10.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":1,"date":"2020-02-10 12:51:36.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 11:16:10.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2020-02-10 12:51:36.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2020-02-10 14:00:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4eb5c0ec446b466bc95bc4caf2bd972adbaeb6cd","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-02-09 21:15:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d90cb1c199b9eff73bf0850c4e96c04c058f12b5","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-02-09 21:16:39.000000000","message":"Patch Set 1:\n\nHi Spyros and Bharat, I understand this one is kind of like a feature, but I think it would be nice if we can have it in stable/train and there are some small bug fix in the upgrade-kubernetes.sh script. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"3ac6d5216b8293cab89a19bce3d9267dc471c088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-09 22:10:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8eacbe2ee354f7cbfdea9dc8a7a7b10 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24888951f7f542558c2d57f97981b549 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2704f3697a794bef8e0d8be1dd39cbe2 : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a5537516bfe450b904a671358776fdc : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51589cedfceb415484c72961e65e864f : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21bc83927f574c1dbdd362e066a3ad97 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1ed0292bb14067a0d825d2f432d271 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c24783de6c0491395797ac1a8c4e094 : SUCCESS in 6m 26s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/ab644a411ce248fb9e1bc8538d37d24a : SUCCESS in 45m 12s\n- magnum-functional-api-python2 https://zuul.opendev.org/t/openstack/build/b3747a9948b4477097fe318203aa1579 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":1},{"id":"85a891f2c58bf36b958d5b1f271fe1c90c823822","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-02-10 11:16:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8112636bedbf58842dc2f7433b5eb89948d1d19","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-02-10 12:51:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fb9fdf8eec5416738fd9bcbaa64ffd233665d92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 12:51:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4845cdaa76054c11e50b4ce136689b730c0aed46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:00:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fd80a82fd044cb09e08effff070e0a9 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7296eedf9d964bc487aa5c4f247365bf : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9be7791b6bf1448bb54be29f8d551fd2 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f549b04662da4a5780129aaa33cbeef8 : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3072d07b9af4fd39edbfaa2a99059f9 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9960cdd7ed74491abd61f376541f3a88 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c56f1d1ea5ad47078989674715a9b02f : SUCCESS in 8m 32s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/389eaf4d951c41e69d64ea7d54d0ea1c : SUCCESS in 1h 04m 15s\n- magnum-functional-api-python2 https://zuul.opendev.org/t/openstack/build/ec3efa952ddc4ebaadadb3e83a5d0a99 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bf46ee833e8ec33100fd4a9e53bce516f0b8345e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:00:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9e3edc82d328e0844cea3882ece4e1df871fa9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:03:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e202f65e3f43d787c64755d45f6294 : SUCCESS in 2m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4d56227fb54eb4954d6a9e3c7089b7 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b625257d365186cdf68f55f9a794ceb805251d7","revisions":{"5b625257d365186cdf68f55f9a794ceb805251d7":{"kind":"REWORK","_number":1,"created":"2020-02-09 21:15:01.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/706721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/706721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/706721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/706721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/706721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/706721/1"}}},"commit":{"parents":[{"commit":"84cd9ca381ff3e11fba321649a3ca4e058832e84","subject":"Fix proxy for Grafana script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/84cd9ca381ff3e11fba321649a3ca4e058832e84"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2019-12-09 03:16:50.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-02-09 21:15:01.000000000","tz":0},"subject":"Support verifying the digest for hyperkube image","message":"Support verifying the digest for hyperkube image\n\nGiven we\u0027re using public container registry as the default registry,\nso it would be nice to have a verification for the image\u0027s digest.\nKubernetes already supports that so user can just use format like\n@sha256:xxx for those addons\u0027 tags. This patch introduces the support\nfor hyperkube based on podman and fedora coreos driver.\n\nTask: 37776\nStory: 2007001\n\nChange-Id: I970c1b91254d2a375192420a9169f3a629c56ce7\n(cherry picked from commit a943756aeab2a6f54485da39be0d986660cd7db0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/5b625257d365186cdf68f55f9a794ceb805251d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/5b625257d365186cdf68f55f9a794ceb805251d7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
