)]}'
{"id":"openstack%2Fmagnum~714021","triplet_id":"openstack%2Fmagnum~master~I02f8cb77b472142f42ecc59a339555e60f5f38d0","project":"openstack/magnum","branch":"master","topic":"story/2007256-390101","hashtags":[],"change_id":"I02f8cb77b472142f42ecc59a339555e60f5f38d0","subject":"[k8s] Upgrade k8s dashboard version to v2.0.0","status":"MERGED","created":"2020-03-20 02:44:13.000000000","updated":"2020-05-01 18:57:43.000000000","submitted":"2020-05-01 14:20:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714021-1588342842821-a9cd72ef","meta_rev_id":"cc3f2da7f40d620ecc326afcd2b3195f27540e5c","_number":714021,"virtual_id_number":714021,"owner":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"date":"2020-05-01 03:16:35.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-01 14:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":2,"date":"2020-04-30 10:56:18.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":2,"date":"2020-05-01 13:27:00.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":1,"date":"2020-05-01 13:27:00.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 07:07:30.000000000","updated_by":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"reviewer":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"state":"REVIEWER"},{"updated":"2020-04-30 10:56:18.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2020-05-01 13:27:00.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2020-05-01 14:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95541b98d5c01c66052330ff4c0c1aaccf56a279","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-03-20 02:44:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e88dff025f68a9c032d61868ea22a7bc56506749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 03:30:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/306577b8991a4bea83e8a4182ceb2bc2 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ada0aac832da44578ef69a7da4ad7e06 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafc072649434b2896edee39b7f8649f : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/049a6b75d3b7407d83ae9b0b5b3d9633 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/591e129627da40399a11a8a57e17cdfb : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52f8bd9ad2e84a40bcc98a82bac9a6f2 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ceeb6d991044804a83daf9d93d80d23 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef498b35e33e46a0b405fe1066cdd57c : SUCCESS in 6m 32s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/f6e7a04fb1ed447488f11159e1c4ee41 : SUCCESS in 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8b128ca5548fbfcd1283c3962dc0af7f6888ede0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-24 11:30:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26bde3922f841b0e999ee58226330cc72ae17592","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-03-25 06:05:30.000000000","message":"Patch Set 1: Workflow-1\n\nSorry, I think we probably need more work for this one. We need to get rid of heapster by the way.","accounts_in_message":[],"_revision_number":1},{"id":"d0292400e59a2f32f0c22bb3cc1f4dab7a291b65","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2020-03-26 09:39:56.000000000","message":"Patch Set 1:\n\nI was actually waiting for the dashboard v2 that has integration with the metrics-server.\n\nThere is an rc6 release, it should be close to be launched a stable release, would you like to wait?","accounts_in_message":[],"_revision_number":1},{"id":"f598f5ccf4ccdc005da147d26175d00d59feb036","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-26 09:41:15.000000000","message":"Patch Set 1:\n\n@dioguerra makes sense","accounts_in_message":[],"_revision_number":1},{"id":"8efa2487e27e9c638aa0895cab180e0154d6f6de","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-02 05:09:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99c3e61fe6b16b217872af07c39e39be8075ff66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 05:57:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1208d5271c2a478697994cdcea2a5f1b : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df5d1d7519b643249d95747b49abf9ba : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871dc90e36a9411fbca25c938223cc70 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1157f3844eca4fc0abbb791c958e420a : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd18343bbe5c428c963cabf82f9dfe03 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d5f683d91234c7295f192bdb5825686 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665552c11fcf46d1b13abcdb0fdcf0ca : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/727cef0ca6e54cc3a944151277d37436 : SUCCESS in 6m 21s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/dd18579d22f44c2e9a2bf503e5ff4e71 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d29d66a68103b6329f9de845436227e22d3c5235","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2020-04-02 07:07:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nSome comments","accounts_in_message":[],"_revision_number":2},{"id":"0759d82b2ea58304b1c6e8c66383cdf780447095","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-02 07:30:22.000000000","message":"Patch Set 2:\n\n(6 comments)\n\n@Diogo, thank you for the review. This one is still WIP, I will address your comments in the following patch sets. Cheers.","accounts_in_message":[],"_revision_number":2},{"id":"f94c2c449c27596edbfdadef35e174766898414b","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-02 08:58:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50b0c1a374224657bcd8827a8a9a9822d593e8b5","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-02 10:32:24.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fa247113003a038f4603dc2c065025d098f98158","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2020-04-02 11:34:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d53f5d7472de87988e5682d150df1ab9faf317f","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-05 05:32:48.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3ceacf1b8acd909b6e0c8dcf4aecb2a1be722187","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-05 19:44:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e664c258c58b0250e142b25882a16cbe19f6aaaf","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-06 12:09:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0f1b66ca5e0aa4f56d2b4f3e6443915df079eae","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-07 05:17:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5696f85e471f52f5ef77a2b3842bc3585af128a1","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-08 08:08:16.000000000","message":"Patch Set 2: Code-Review-1\n\nneed to document kube_dashboard_version in the user docs.","accounts_in_message":[],"_revision_number":2},{"id":"8738cbe3e0b2557b831f1d8c571dc508cffd638b","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-20 10:51:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be6bb895a578c6dd99b3536454c2a98b56ff9c6c","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-23 02:52:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"69e6ba1a95fff7766e38762d97aedc670a8c305d","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-23 03:01:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5fabdb9f635f7c43657d6c2d95aff76ff92f70ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 04:42:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e054e5e1a6004d8db020c127a9366b14 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0332d3154b84f829c8947321a28e14d : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161c67d6478c40f8a5e6dce8267879a7 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4687b82bc234312a338238335431dcc : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba65e54733864737bb3e7de6fc95181f : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03ea5a4011584e4a92f4f1077e121a9d : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d39ddc28914420822b5231a164a48a : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70a2334b43f84f3eba79106db9fc1571 : SUCCESS in 5m 33s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/c80a52c2e2144b88a31ba72c31732550 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":4},{"id":"fbe7951e333c83f920f2998cfc5907aab20d806d","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-23 05:20:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a6ef781707d60b574e4726ca73d15b98577346ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 06:15:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7320b46488b463f9983ec6faeebd193 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c0300ade09e49e7adaaf004c8acc33d : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac0eb253df6f482eae07d4140ad10b52 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d15fe92b73984cf899dce794457f0d86 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e75c40f86fe4fbdae0f5599895eae07 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90185c06f0094c3898d6a55873528771 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78c714ac60042d78cf27e90229a16de : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b541e19c7e042e28460df20c9749315 : SUCCESS in 6m 49s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/a34530f2fa594df0ae7104f9eb618337 : SUCCESS in 52m 52s","accounts_in_message":[],"_revision_number":5},{"id":"de4595ba79e06c4bcf251d3f8772f8e0ae6f55f4","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-23 10:14:50.000000000","message":"Patch Set 5: Code-Review-1\n\n(7 comments)\n\nJust some nits","accounts_in_message":[],"_revision_number":5},{"id":"8f03b22d8f282f91c76d6d628d93825b8044598b","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-24 03:40:01.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"b6c2ebdc304665178324c7d990f3098031c291d3","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-24 04:16:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7e334044d6d63fe7a0ff0681d99b31505baa010f","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-24 04:34:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9bc277ecad4bd828174f1ee79c0a4347293fc2f3","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-24 04:36:16.000000000","message":"Patch Set 5:\n\nshu  4:34 PM\nbasically, we don\u0027t recommend to use kube-system from security point of view.\nflwang:flag-nz:  4:35 PM\nah, good point.\n\nThe k8s dashboard community is not recommend using kube-system ns for security reason.","accounts_in_message":[],"_revision_number":5},{"id":"29e79b10c3dca552dc3265a263be63fef64c011f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:19:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29838c11747845b0b6435ab31921c7cd : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a718b7df0594f649be0627aab33b037 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bacaf04e7454f42984ae57e92b17596 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5a7765f0fc54093ae79c1f2211c778b : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/607625c0f6d2424592dcd606cbaaa701 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53377c73cae4a1b92d1cd96650e98f9 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e05ccbac13a4ced8e8f6d652ae44406 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95b58915bde413c950a29470c04d922 : SUCCESS in 6m 33s\n- magnum-functional-api https://zuul.opendev.org/t/openstack/build/7c3a42efb2f1452f922f58eabba1a909 : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":6},{"id":"c97dd4f1f17a15392e3b320b3479017f85dd950d","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-28 01:45:27.000000000","message":"Patch Set 6:\n\n@Spyros, @Bharat, the dashboard 2.0.0 has released. Can you please revisit this one? Thanks.","accounts_in_message":[],"_revision_number":6},{"id":"e1df992a10312eddcc6ab000a11cfe1a65f5f5d7","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-30 08:33:51.000000000","message":"Patch Set 6:\n\n@Spyros, @Bharat, ping","accounts_in_message":[],"_revision_number":6},{"id":"6a69473425fe170aeecfb8434302c2290527f434","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-30 10:03:15.000000000","message":"Patch Set 6:\n\nIt is so much fun to switch from v3.4.7 to 3.4.7 for etcd depending on where a patch is based on. I guess the only one using devstack and checks out branches.","accounts_in_message":[],"_revision_number":6},{"id":"ab0d6f4c3201b2b1029bf05cfaf59ee284f25825","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-30 10:56:18.000000000","message":"Patch Set 6: Code-Review+2\n\nI had to rebase on master for coredns to work. After that, it is great.\n\nhttps://s3.cern.ch/strigazi-public/review-opendev-org-714021-6-kubernetes-dashboard.png","accounts_in_message":[],"_revision_number":6},{"id":"3c159a993c37961d35fcd101186e06ca2005a588","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-30 12:57:31.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nWorks but for this to be compatible with new calico, we need to open tcp port 8443 on worker node (and master too if we want this to be able to run there), otherwise this is the result of a curl:\n\nubuntu@devstack-master:/opt/stack/magnum$ curl -k https://172.24.4.204\ncurl: (35) OpenSSL SSL_connect: SSL_ERROR_SYSCALL in connection to 172.24.4.204:443","accounts_in_message":[],"_revision_number":6},{"id":"a6b4f775024880c6683dba32673baae38efc4120","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-04-30 13:08:55.000000000","message":"Patch Set 6:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e Works but for this to be compatible with new calico, we need to\n \u003e open tcp port 8443 on worker node (and master too if we want this\n \u003e to be able to run there), otherwise this is the result of a curl:\n \u003e \n \u003e ubuntu@devstack-master:/opt/stack/magnum$ curl -k https://172.24.4.204\n \u003e curl: (35) OpenSSL SSL_connect: SSL_ERROR_SYSCALL in connection to\n \u003e 172.24.4.204:443\n\nYou tried to curl port 443 of the master node from the devstack host?","accounts_in_message":[],"_revision_number":6},{"id":"4f6f37343e0b585bf87d834bd155c284af284658","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-04-30 19:58:52.000000000","message":"Patch Set 6:\n\n(1 comment)\n\n@Bharat, can you please help me understand what\u0027s the -1 for?\n\nAnd how did you test the dashboard? As a service, it shouldn\u0027t be tested via curl. I would suggest you test it by kubectl proxy or even port-forward, but I don\u0027t think curl is the right way to test k8s dashboard.","accounts_in_message":[],"_revision_number":6},{"id":"1eaf60a3576a282e63cd77b1bb7e4dbd5f69cb10","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-30 20:45:38.000000000","message":"Patch Set 6:\n\nHi Flwang, I will remove -1 after we open port 8443 tcp on security group rules on worker/master for kubernetes services to reach kube-dashboard. To reproduce:\n\n- Rebase on latest calico (this dashboard has quite an old parent)\n- Edit svc/kubernetes-dashboard and change from ClusterIP to LoadBalancer type and `curl https://\u003cexternal-ip\u003e:443` or try to access this on your browser. It didnt work for me on browser either because the service was the able to reach the container.\n- With old calico, it works for me without any issues. As we know that with new calico, we have to open security group port for various services.","accounts_in_message":[],"_revision_number":6},{"id":"74688e8ebe061b78615e48116d01087a03098605","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-30 20:48:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77893b206c15dddc1d4c6e217f5a007b0db7515a","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-05-01 03:01:33.000000000","message":"Patch Set 6:\n\n@Bharat, based on my understanding, the recommended way to access k8s dashboard is using kubectl proxy, see https://kubernetes.io/docs/tasks/access-application-cluster/web-ui-dashboard/#command-line-proxy\n\nIf a user really want to expose the dashboard to external, then they should open the security group manually following the LB service creation. We shouldn\u0027t open the 8443 port by default. I happy to document this if you guys think it\u0027s necessary. But I prefer not opening the 8443 by default anyway.\n\n@Spyros, thoughts?","accounts_in_message":[],"_revision_number":6},{"id":"2c9da7b08507ada555caca741ff2283693d8b9d2","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-01 03:13:34.000000000","message":"Patch Set 6:\n\nAre you able to reach the dashboard with new calico too? Never mind the method, external ip or local kube proxy, it appears as though any access from outside of the node is not possible. With old calico, this is not an issue as I have already said.","accounts_in_message":[],"_revision_number":6},{"id":"04afb5f168f099646708575300e8ba435f2e2bd1","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-05-01 03:16:35.000000000","message":"Patch Set 6:\n\nYes, kubectl proxy and port-forward both work for me.","accounts_in_message":[],"_revision_number":6},{"id":"ad2978b5a6ce9e3d7f13ef910bf3fa916f107a1f","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-01 13:27:00.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThanks for the clarification, I did not realise port forwarding was an option too which works for me as I cant access things being served on localhost on my local web browser.","accounts_in_message":[],"_revision_number":6},{"id":"1686e91438cfdc17c15b981598c7929263391caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 13:27:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fcc28c578bbdb9066441ae52adaab5259ac6a4a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 14:20:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58f5d7df52d0415e8cdbc9b9daef86b6 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a099849102f341329283bc33269d4768 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39127bd24a274e6c944aa3973f0d8735 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54680be2d57f46f5866b28fd5d2dcf6d : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dca539bcde34185aaeff21b22f32ea8 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a364a712ae42412b8026ba7ec2e5404f : SUCCESS in 6m 31s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/07ffb9e5a81c485a924b6da8fcbdf398 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":6},{"id":"5ca2f624e1ecb6b56601dc066646c99713c9a7d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 14:20:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"8941b19a052e7055eb6913e0806441e6528b50a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 14:22:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706fc13876724c5e8ce95b29b87fc4d4 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c92b488da36142e3b0bbc45a6f3cbdd2 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":6},{"id":"0c6834ec37a591a94dca18c2a0440f78a8ee4f71","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-05-01 18:57:43.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 45563f37aeb62a7aaee68470f901780d976b9e05","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b4965416b145c741fa344110ae27a2536458135c","revisions":{"a86b413433709a2bc16e516f0968a30774020cd2":{"kind":"REWORK","_number":1,"created":"2020-03-20 02:44:13.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/714021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714021/1"}}},"commit":{"parents":[{"commit":"ce70da25ad4bcdecb261b7e915fc423dae560d07","subject":"Merge \"Fix join of status_reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce70da25ad4bcdecb261b7e915fc423dae560d07"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:41:37.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:43:56.000000000","tz":780},"subject":"[k8s] Update k8s dashboard verstion to 1.10.1","message":"[k8s] Update k8s dashboard verstion to 1.10.1\n\nTask: 39101\nStory: 2007256\n\nChange-Id: I02f8cb77b472142f42ecc59a339555e60f5f38d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a86b413433709a2bc16e516f0968a30774020cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a86b413433709a2bc16e516f0968a30774020cd2"}]},"branch":"refs/heads/master"},"e15b751c34cff02bcb6935ea1212d80413886256":{"kind":"REWORK","_number":2,"created":"2020-04-02 05:09:04.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/714021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714021/2"}}},"commit":{"parents":[{"commit":"ce70da25ad4bcdecb261b7e915fc423dae560d07","subject":"Merge \"Fix join of status_reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce70da25ad4bcdecb261b7e915fc423dae560d07"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:41:37.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-04-02 05:08:41.000000000","tz":780},"subject":"[WIP][k8s] Update k8s dashboard verstion to v2.0.0","message":"[WIP][k8s] Update k8s dashboard verstion to v2.0.0\n\nHepster has been deprecated for a while and the new k8s dashboard\n2.0.0 version starts to support metrics-server now. So it\u0027s time\nto upgrade the default k8s dashboard to v2.0.0.\n\nTask: 39101\nStory: 2007256\n\nChange-Id: I02f8cb77b472142f42ecc59a339555e60f5f38d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/e15b751c34cff02bcb6935ea1212d80413886256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/e15b751c34cff02bcb6935ea1212d80413886256"}]},"branch":"refs/heads/master"},"02d59af389f92c01753d550148d840156c005322":{"kind":"REWORK","_number":3,"created":"2020-04-23 02:52:52.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/714021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714021/3"}}},"commit":{"parents":[{"commit":"ce70da25ad4bcdecb261b7e915fc423dae560d07","subject":"Merge \"Fix join of status_reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce70da25ad4bcdecb261b7e915fc423dae560d07"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:41:37.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-04-23 02:47:30.000000000","tz":720},"subject":"[k8s] Update k8s dashboard verstion to v2.0.0","message":"[k8s] Update k8s dashboard verstion to v2.0.0\n\nHepster has been deprecated for a while and the new k8s dashboard\n2.0.0 version starts to support metrics-server now. So it\u0027s time\nto upgrade the default k8s dashboard to v2.0.0.\n\nTask: 39101\nStory: 2007256\n\nChange-Id: I02f8cb77b472142f42ecc59a339555e60f5f38d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/02d59af389f92c01753d550148d840156c005322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/02d59af389f92c01753d550148d840156c005322"}]},"branch":"refs/heads/master"},"b55b1dab0cb4ef084dfae90e784e4219cb31278b":{"kind":"REWORK","_number":4,"created":"2020-04-23 03:01:20.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/714021/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714021/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714021/4"}}},"commit":{"parents":[{"commit":"ce70da25ad4bcdecb261b7e915fc423dae560d07","subject":"Merge \"Fix join of status_reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce70da25ad4bcdecb261b7e915fc423dae560d07"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:41:37.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-04-23 03:01:09.000000000","tz":720},"subject":"[k8s] Update k8s dashboard verstion to v2.0.0","message":"[k8s] Update k8s dashboard verstion to v2.0.0\n\nHepster has been deprecated for a while and the new k8s dashboard\n2.0.0 version starts to support metrics-server now. So it\u0027s time\nto upgrade the default k8s dashboard to v2.0.0.\n\nTask: 39101\nStory: 2007256\n\nChange-Id: I02f8cb77b472142f42ecc59a339555e60f5f38d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b55b1dab0cb4ef084dfae90e784e4219cb31278b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b55b1dab0cb4ef084dfae90e784e4219cb31278b"}]},"branch":"refs/heads/master"},"6c09d0d76c5a6bd40f2876cf3af759bfaf5eabd2":{"kind":"REWORK","_number":5,"created":"2020-04-23 05:20:36.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/714021/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714021/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714021/5"}}},"commit":{"parents":[{"commit":"ce70da25ad4bcdecb261b7e915fc423dae560d07","subject":"Merge \"Fix join of status_reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce70da25ad4bcdecb261b7e915fc423dae560d07"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:41:37.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-04-23 05:19:58.000000000","tz":720},"subject":"[k8s] Update k8s dashboard verstion to v2.0.0","message":"[k8s] Update k8s dashboard verstion to v2.0.0\n\nHepster has been deprecated for a while and the new k8s dashboard\n2.0.0 version starts to support metrics-server now. So it\u0027s time\nto upgrade the default k8s dashboard to v2.0.0.\n\nTask: 39101\nStory: 2007256\n\nChange-Id: I02f8cb77b472142f42ecc59a339555e60f5f38d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/6c09d0d76c5a6bd40f2876cf3af759bfaf5eabd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/6c09d0d76c5a6bd40f2876cf3af759bfaf5eabd2"}]},"branch":"refs/heads/master"},"b4965416b145c741fa344110ae27a2536458135c":{"kind":"REWORK","_number":6,"created":"2020-04-24 04:34:54.000000000","uploader":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"ref":"refs/changes/21/714021/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714021/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714021/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714021/6"}}},"commit":{"parents":[{"commit":"ce70da25ad4bcdecb261b7e915fc423dae560d07","subject":"Merge \"Fix join of status_reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce70da25ad4bcdecb261b7e915fc423dae560d07"}]}],"author":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-03-20 02:41:37.000000000","tz":780},"committer":{"name":"Feilong Wang","email":"flwang@catalyst.net.nz","date":"2020-04-24 04:34:36.000000000","tz":720},"subject":"[k8s] Upgrade k8s dashboard version to v2.0.0","message":"[k8s] Upgrade k8s dashboard version to v2.0.0\n\nHeapster has been deprecated for a while and the new k8s dashboard\n2.0.0 version supports metrics-server now. So it\u0027s time to upgrade\nthe default k8s dashboard to v2.0.0.\n\nTask: 39101\nStory: 2007256\n\nChange-Id: I02f8cb77b472142f42ecc59a339555e60f5f38d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b4965416b145c741fa344110ae27a2536458135c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b4965416b145c741fa344110ae27a2536458135c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
