)]}'
{"id":"openstack%2Fmagnum~714721","triplet_id":"openstack%2Fmagnum~master~Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067","project":"openstack/magnum","branch":"master","topic":"docs","hashtags":[],"change_id":"Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067","subject":"[docs] Bring user docs up to date with recent changes","status":"MERGED","created":"2020-03-24 17:24:12.000000000","updated":"2020-07-30 12:00:36.000000000","submitted":"2020-07-30 11:59:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714721-1596110351707-5080a3ed","meta_rev_id":"a1df35cbaf21437a1177b67ad4a42cb553832111","_number":714721,"virtual_id_number":714721,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 11:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":2,"date":"2020-07-30 11:45:47.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":2,"date":"2020-07-15 09:36:41.000000000","_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2020-07-15 09:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":1,"date":"2020-07-30 11:45:47.000000000","_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},{"value":0,"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 12:38:16.000000000","updated_by":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"reviewer":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"state":"REVIEWER"},{"updated":"2020-07-15 09:36:41.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2020-07-15 09:48:26.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-07-30 11:45:47.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2020-07-30 11:59:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd043667899550b7cf4317e998197a79a12a7826","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-24 17:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a413ee81713eb6e0ba82b910094892b1e7399f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 17:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92210f9383341a6aee238e2733a78ae : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cde2b9585df46f9a69900a6e312ba61 : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"da5b6f8aaa68cb260cdbaca8539a033d219d3043","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2020-03-26 09:50:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7245ede9520803ad31e78d96f0661a3d15abef73","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-27 21:49:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0d6383633604234e175bbb129c2ef499b9e6973f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 21:59:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0539d3decc47e9b543378a6630e6c8 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aba6f20627845b793dae6433a61e3f2 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5c2f5c6dd6b2405ca9970f3b3fcd113a65efe3a0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-30 09:13:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2aee9041eb5820b53173331948cb19c56df88bb0","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-30 09:13:43.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"def8335c0da21be94bb71aa753d92863ea644c63","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-30 09:15:03.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"125654a9f8498c11a330a51e18014cd83015bade","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-30 09:15:16.000000000","message":"Topic changed from calico to docs","accounts_in_message":[],"_revision_number":3},{"id":"04cd093dcafba5ca654d8c0ef7e859b94fd3ff91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 09:25:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2b29f7550e4428bae22b507fe9f422 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416ad99d40944e57bff6a092cf8f45e4 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d7dc905ca7acdf1b5a6c1ec8e7102d4146d7c432","author":{"_account_id":29425,"name":"Diogo Guerra","email":"diogo.filipe.tomas.guerra@cern.ch","username":"dioguerra"},"date":"2020-03-31 12:38:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dfc2a25b2a056f24c36ca445ec3dd7da7de19ae8","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-31 12:40:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2a38b1f4b23a43bb3230ad615376b91788eeeec","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-07-06 14:41:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5044709388060dc8b57649d2c9d4be42263f9a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 14:55:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d8bd4b3430433c886196bb497638bc : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d654ef0d0caf4502b35b2b10558a4f4a : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e24d794aee04d9861f2f92bead7061cf7df14d46","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-07-06 15:09:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8f1ff9dda600f43ee205798e9a7a7a44d6bf69ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:24:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484e7c4c78454eafa2c3b6d912f37774 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b046dc39fbc4d2689223ee4f888fb8b : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":5},{"id":"596351f5d949f3373d09eac29109ee51e5d87689","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-07-07 02:12:49.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nVery good and thank you. Just a minor comment.","accounts_in_message":[],"_revision_number":5},{"id":"834cde492f83f39c2fbb45376e9dc7d95ba31ded","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-07-15 09:01:25.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6980080ec75593270a2bd33455167d16c8e64727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 09:12:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890c82fd33d84ab5936ee1bfd91cc67b : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c250b22d637495eb23943fae9b004e4 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d32457e4c49b1d292cf328f283c9d028e417cacd","author":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"date":"2020-07-15 09:36:41.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2b2d2179ad06c995708fb6ebdd0a12997d4ee51a","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-07-15 09:48:26.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6f893d4198b04587a86cd8f069177d5d37656906","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-07-30 11:45:47.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c51292af3c423ec522834e5d426c5a6af1295c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 11:45:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5c7bc3526b2be2d573062825e30d25f17aa3ee16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 11:59:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b73a49436a4189125c50a219c52971f8559710bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 11:59:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8cadffe60b4f0490baf304d69f90a2 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e4b3fff29149e0a81abe5cdd6e4e9d : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":6},{"id":"a1df35cbaf21437a1177b67ad4a42cb553832111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 12:00:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/315d8daf6f83442181676780e2b79bf4 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b10b55df13f681e3b4bf157b93fcb9938c42da3c","revisions":{"4b3bd86e32d59c794dd9fa330bd5e3b67694bc1c":{"kind":"REWORK","_number":1,"created":"2020-03-24 17:24:12.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/714721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714721/1"}}},"commit":{"parents":[{"commit":"5f6962fe3cb8b223557367644f999b3cf406f4d6","subject":"k8s: Upgrade flannel to v0.12.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/5f6962fe3cb8b223557367644f999b3cf406f4d6"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-24 16:36:00.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-24 17:23:55.000000000","tz":0},"subject":"docs: Bring user docs up to date with recent changes","message":"docs: Bring user docs up to date with recent changes\n\nChange-Id: Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/4b3bd86e32d59c794dd9fa330bd5e3b67694bc1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/4b3bd86e32d59c794dd9fa330bd5e3b67694bc1c"}]},"branch":"refs/heads/master"},"63ec20abda51867b93c71c43de88aae31988c6a4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-27 21:49:50.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/714721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714721/2"}}},"commit":{"parents":[{"commit":"48d6159d97f7606e31873ffb325fcc2b4f8a089d","subject":"k8s: Upgrade flannel to v0.12.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/48d6159d97f7606e31873ffb325fcc2b4f8a089d"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-24 16:36:00.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-27 21:49:34.000000000","tz":0},"subject":"docs: Bring user docs up to date with recent changes","message":"docs: Bring user docs up to date with recent changes\n\nChange-Id: Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/63ec20abda51867b93c71c43de88aae31988c6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/63ec20abda51867b93c71c43de88aae31988c6a4"}]},"branch":"refs/heads/master"},"6caf95ef0d63918cd16466533dfdfce771ab8f10":{"kind":"REWORK","_number":3,"created":"2020-03-30 09:13:06.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/714721/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714721/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714721/3"}}},"commit":{"parents":[{"commit":"61336027c2da89d0cbc4c6ccd96e567a39cd6b5c","subject":"fcos: Upgrade etcd to v3.4.6, use quay.io/coreos/etcd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/61336027c2da89d0cbc4c6ccd96e567a39cd6b5c"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-24 16:36:00.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-30 09:12:25.000000000","tz":0},"subject":"docs: Bring user docs up to date with recent changes","message":"docs: Bring user docs up to date with recent changes\n\nChange-Id: Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/6caf95ef0d63918cd16466533dfdfce771ab8f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/6caf95ef0d63918cd16466533dfdfce771ab8f10"}]},"branch":"refs/heads/master"},"d0b83ef673d6fa32abf610d78961ba3daf99442e":{"kind":"REWORK","_number":4,"created":"2020-07-06 14:41:04.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/714721/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714721/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714721/4"}}},"commit":{"parents":[{"commit":"1f6df6fb4c825e6589eb43a2c0eb366b4d114628","subject":"Merge \"Remove shebang from scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1f6df6fb4c825e6589eb43a2c0eb366b4d114628"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-03-24 16:36:00.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-06 14:40:50.000000000","tz":0},"subject":"docs: Bring user docs up to date with recent changes","message":"docs: Bring user docs up to date with recent changes\n\nChange-Id: Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/d0b83ef673d6fa32abf610d78961ba3daf99442e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/d0b83ef673d6fa32abf610d78961ba3daf99442e"}]},"branch":"refs/heads/master"},"edfe8ad3f71089ef6bb26df05b3cd278ab17f397":{"kind":"REWORK","_number":5,"created":"2020-07-06 15:09:03.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/714721/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714721/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714721/5"}}},"commit":{"parents":[{"commit":"1f6df6fb4c825e6589eb43a2c0eb366b4d114628","subject":"Merge \"Remove shebang from scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1f6df6fb4c825e6589eb43a2c0eb366b4d114628"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-06 14:48:44.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-06 15:08:38.000000000","tz":0},"subject":"[docs] Bring user docs up to date with recent changes","message":"[docs] Bring user docs up to date with recent changes\n\nChange-Id: Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/edfe8ad3f71089ef6bb26df05b3cd278ab17f397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/edfe8ad3f71089ef6bb26df05b3cd278ab17f397"}]},"branch":"refs/heads/master"},"b10b55df13f681e3b4bf157b93fcb9938c42da3c":{"kind":"REWORK","_number":6,"created":"2020-07-15 09:01:25.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/21/714721/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/714721/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/714721/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/714721/6"}}},"commit":{"parents":[{"commit":"1f6df6fb4c825e6589eb43a2c0eb366b4d114628","subject":"Merge \"Remove shebang from scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1f6df6fb4c825e6589eb43a2c0eb366b4d114628"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-06 14:48:44.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-07-15 09:01:17.000000000","tz":0},"subject":"[docs] Bring user docs up to date with recent changes","message":"[docs] Bring user docs up to date with recent changes\n\nChange-Id: Ic6fef848726b7b20eaf0cd09f67d9ea21b5ac067\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b10b55df13f681e3b4bf157b93fcb9938c42da3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b10b55df13f681e3b4bf157b93fcb9938c42da3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
