)]}'
{"id":"openstack%2Fmagnum~815521","triplet_id":"openstack%2Fmagnum~master~I32b7a91640f4f65b9662de080682761a2c182055","project":"openstack/magnum","branch":"master","topic":"clusterapi-driver","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2024-04-15 09:56:49.000000000","reason":"Change was abandoned"},"6484":{"account":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"last_update":"2024-04-15 09:56:49.000000000","reason":"Change was abandoned"},"14394":{"account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"last_update":"2024-04-15 09:56:49.000000000","reason":"Change was abandoned"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2024-04-15 09:56:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I32b7a91640f4f65b9662de080682761a2c182055","subject":"Add a stub Cluster API helm driver","status":"ABANDONED","created":"2021-10-26 15:58:21.000000000","updated":"2024-04-15 09:56:49.000000000","total_comment_count":19,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0120cb6c18b7f8a9e51bcac903087fddb67502be","_number":815521,"virtual_id_number":815521,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-06 16:08:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 17:01:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-03 21:55:45.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2021-11-25 12:43:41.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2023-08-21 02:08:21.000000000","updated_by":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"reviewer":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"state":"REVIEWER"},{"updated":"2023-09-18 01:22:48.000000000","updated_by":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"reviewer":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"state":"REVIEWER"}],"messages":[{"id":"1bf372d6107c5f9550b0e95b2b255b440b862231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-10-26 15:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"379413dc56bad12f76ec730f1d709122a8b89f6d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-10-26 15:58:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d11bee3b3898cf8fc50d52fbf12659da149c3abb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:01:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c2ec8556a500465890235c9e7fb18b65 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ca6a7ab8953408babab5c897e955f48 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9591b92d6e42699413aff0f33e2979 : FAILURE in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6180c6f78946457b83be35831130360e : FAILURE in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c64f1942b9eb40f8871d93270a6eb915 : FAILURE in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40507b315af24d35bffacd2994eb9005 : SUCCESS in 9m 13s\n- magnum-tempest-plugin-tests-api-temporary https://zuul.opendev.org/t/openstack/build/a5da1c62318841528a4583729d68ef8a : SUCCESS in 56m 31s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/fe7a258978a9448dbc2be54bad2e7bd0 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c709a1dfb67fcfa51170e40000f63f1567922d41","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-11-25 12:43:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1495389b4f04a8ef53ef68486ed9478fdae74b04","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2022-07-26 17:35:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"746340286b16b3443c4cde53f9d45f4437cb9a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 18:24:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68f1b4e422074e9da402385e8191f376\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/75a256c5f293442da96fcafc04a9ea9e : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3897322b2c2f40f9a627cd4834e6ee71 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a038fed8a84217990389d0601b6e0a : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8bdcdb93e7541ba8620c4f52464da48 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5131c3c548044f7abdfb65ddaba02e9 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71af9dd0525d455d9e69c97e12e8250d : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03410b76a2442a9885a87a532961113 : SUCCESS in 7m 34s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/34977ae7edbc43c49c748d6678cab6b3 : FAILURE in 43m 47s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/ff8603315c364a5b9a4eecdef95a3d50 : SUCCESS in 23m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b66847db38014164f6dca9676fe31988d8aa15da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2022-08-15 08:50:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0333c9da45752ff06e3b077963837e03d60dbb5f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2022-08-15 08:56:18.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fd789976f969d8aa7b61316b4a90ac4d6ebe1da8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 09:34:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba9dd53a88b467f887aaab958f05098\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e4daec393bf457a8682586e06034ad3 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e757ac780db5432a80c0a63716f80469 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2900b6ac292f4fe48336fb0a58d1fb99 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d98066d2922d43d28afd2fecc3376023 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e028638845274c8695135e99de72cc20 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72a91f7d42d34424aa5f31fc3fc26841 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d797d2b1b24827aaecf3d4095bd030 : SUCCESS in 7m 29s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/130fc90754f74fabb36b156696d49183 : SUCCESS in 37m 08s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/2198dce94a184563b13aedbef018240f : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":4},{"id":"77b3f34208342140dae6cac20f448c24dec1def4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2022-11-14 16:11:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6ff3e527eab3e4fc68a7008116eac7f8e2292172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 16:44:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7126cbc5c29644dd96f2f0e293c827dd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d70009ff72194b0481d4ab649abd9da8 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6675a3ed0cbb4a5bb9f5796e482e0648 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffcee9d86093497ea9227204543497a5 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27e6e6afaaf64f01b0ab4bc4eafa21ef : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14e1d75940a742d1b16e8bdec1c3d5ff : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc6498fdabc426fb1ccf6f7c8b7411e : SUCCESS in 5m 41s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/a3fc402f8402499f9ab3a059d933d2cf : SUCCESS in 31m 39s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/04a1f3312f1a497b82e45bceea2960a9 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":5},{"id":"a24e14b525f2565d32ad2abb4628e2dac7ba3987","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-01-29 17:38:04.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"eab8dab7dae3487ca06e94be4964693e05d456f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 18:32:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3522366b016a4b9dabc5b84b0045d6f5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fd14528b3a914f8aa8386a724fa55b33 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/599cfb0d60454c60ab8cbe1f2c607233 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc71696181f4465a1b35de4d6d4487f : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89a2c1c7539c4b879214ce6a74ffc6d9 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8349754a018f41738e4cfdcafc02fcad : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa3855466bd49fdaedd246ca89b2a75 : SUCCESS in 6m 44s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/2b85c6d7e3024af098362a30796bb309 : SUCCESS in 47m 05s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/82cb4266df2a45169f9694737e0ac860 : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":6},{"id":"ad18e6b850ea208ab334dae5ad72453012fba66e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-02-10 15:18:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"f0214d0ef5f43d7f5d1fba72e51cd526d2b6b9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-10 17:58:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e47c6291b4d4a2ebf3feaaff03d0053\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b08530bb00cf422e990465f4670ed743 : SUCCESS in 2m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/210ce5a3dfed449eaa9a59dd40e4a36e : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cb984ca2f648d6a9502748aee3d65c : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f06bce09c7b345c3be71d1b27308ee2a : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8e16498582c478ca26a1c0fe49df786 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58a13e8e0a14337b39329ffe72793b2 : SUCCESS in 7m 46s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/350bf28ef34e440dba3f6c20c0837914 : SUCCESS in 1h 58m 52s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/b063fc3ba0c6427595326ae1f20c1780 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":7},{"id":"6d617dc6969d5f8d097939551305eb8eddd40af4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-03-10 16:29:49.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"aaedb73f19c315315b94d1010959da7603c79c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 17:23:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99897d27a7c94fbcb5d3a495711832d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3579bf11b9a3453b988bf3bffb187637 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06c27d14e2684bafadfbf9ea382b9b09 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd8fcd7bbb3462aa51b7cd171881269 : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4018c889878411596537d91d56da579 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9eecbaf94b34c78a3510b9bf23c9213 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794d54482418413bbcf04433d3699d1e : SUCCESS in 6m 21s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/7b18bbe9297c4c8285271ade7c052ece : SUCCESS in 47m 39s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/cb82fc665b534d1ea2cea7e2f478c84d : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":8},{"id":"38edb21e8fbd2c60e513e293716e36ff44829fb9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-03-20 22:13:24.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"38b95d72b97bfa23548c51ff8032df2a1bb23f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 23:06:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36840702555c4ab8937f89052193e22a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/192e35576f82453b90bc102a7c315645 : SUCCESS in 1m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24142be88d8740218e1a979a3b67917c : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d9d366d00242b0916570cc573d09ad : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6b48af3cc043c8a79d1d7d6a6ee321 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/006aa125373a422cb12b4966b0d031e6 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9617e64d67466ca64092169f4db3ae : SUCCESS in 6m 06s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/8a7d5c8ebf824e2888c37fa1b456a8ba : SUCCESS in 47m 17s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/db63cf2b96144921866a7036f606203f : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":9},{"id":"ccf615564e15dbc73e122165297c17d0c5718f68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-04-25 09:53:24.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"1f9d422dc6e31ebeae212153316f4e8580db253d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-04-25 10:21:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"91285e606b643a263e97acdb17707153bc74f3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 11:11:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d943fbc31a9485483f87ff9cf76592b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e913f3266fbc43a5a1a04ab3b48870fc : SUCCESS in 1m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4cead2a649a4476a9c210c7dbc1d59b : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddea7ad009e4d3f9409f1dbcc15120e : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b647e3871c594138b6403dcd85632aeb : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e5a9c94ec5549768bdf2ea5d7fd1226 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e067637714341d88eb49087a6870c3f : SUCCESS in 6m 21s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/cf630169e8694e009136155dc351d81b : SUCCESS in 45m 19s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/e19059ba37e44a9c8423202cd05a5436 : SUCCESS in 24m 49s","accounts_in_message":[],"_revision_number":11},{"id":"739a88feedd81094308f9976fde7a6d14f00d92e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-04-25 15:31:43.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"23baf7c585545374bb744862184a7471ba041c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 16:27:13.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e248034941643938b6ee5b7ad562f51\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bd3f5bb270a442d9a195e8ead0aedee7 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/415b8e8c9df04b84bdaf40bf703c0aea : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52820e0e64094d1985abef4917066832 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8330483c132f43f7832bf7764e1a9245 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b441feb2b614bab85cd9adb8e13e433 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a252889412542d5a27da058043e1de8 : SUCCESS in 8m 29s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/022a7efd46c34f45961766fd2217072d : SUCCESS in 47m 34s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/6c7b760d92a74c4989b8bc4d0c0f82ef : SUCCESS in 24m 01s","accounts_in_message":[],"_revision_number":12},{"id":"09012428880ae99c1ca2b7e6ef4924b2b4eadc42","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-12 15:59:13.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"054c8bca5e0a8fd6178be8a669d098bcb1394ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 16:29:15.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e969ede7a2b24c8581e73e18d710f8c7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dfb8bc1279f34c91ab71272ba54f2257 : SUCCESS in 1m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7c951356bfb4fdea2aa6765378c7eca : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b16005ca33f4be3801de283c7f7a730 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd653fa75b1463c8105cb5abe790ce4 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2319dd9e4af943bc856d27ca8b3312de : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/826494a43bd74eb5bcbeeefab7b1aa62 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488a3847d4344b6b8c51c9c36be6f889 : SUCCESS in 6m 05s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/63f88a4ff628449bac10614cca51e833 : SUCCESS in 27m 06s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/36e4ffd3288f48e9865faed4d40a830b : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":13},{"id":"4c4a4f5c1eed9193fdf0a687e66c0f54a58eab46","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-06-26 12:15:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":13},{"id":"9bbb993c47ab16551206941f9aed5b2f0a88e47a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-06-26 12:16:25.000000000","message":"Topic mattp-push removed","accounts_in_message":[],"_revision_number":13},{"id":"32fc6b459641d822ca5b2a78b6a2234f6953acb4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-06-26 12:16:36.000000000","message":"Topic set to clusterapi-driver","accounts_in_message":[],"_revision_number":13},{"id":"89adda6c9f5c91071d8c31eb76ac1e15cf4208e7","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2023-08-21 02:08:21.000000000","message":"Patch Set 13: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ad435cbe8c639865fbd37a0bb899b316b66947e5","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2023-08-23 11:55:47.000000000","message":"Patch Set 13: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"5cba41bec929564ab8b7cde50486aa881abe878a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2023-09-07 01:37:07.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"ea51b547832b1b0143724703f4e9a6622c1b4e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 02:10:47.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c50ed91fb44a04a8ecce110c91d413\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8d8708089ad144a0a23ef7e1b79e3148 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/748366ba67074bc8a61b614db98afc6d : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5575dade7ed417a8bbadf83ddaf91d0 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fe72d5c4cf64e03ad0b3a7edca7aa11 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac4c7975f1814787b81d60471087d1ca : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b60a91d022a14cc18c0df35beaa0c0cd : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/598b412730e34d57bcc196c418f1587d : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc0ba497a9c46488f23e70f3d56c968 : SUCCESS in 6m 54s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/6ad90666d52a4afc9bfdb38f23689379 : SUCCESS in 32m 11s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/107e12f047674512845fe347953a8d68 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":14},{"id":"391a2ba26090d012aea22e08ebf2c6a1b95d3cf9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-12 17:43:34.000000000","message":"Patch Set 14: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"946f25d47fa7b266c559e40c0cdf585694287c08","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2023-09-14 01:27:02.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"473b4425b242bf21fd6304d4ce90d65f17819956","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-16 11:04:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c25400ae8005fd6fbe6440a081160513e456caac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-16 11:24:37.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":15},{"id":"526d4b9f0a3c5ab491c113dec490aa73df016a01","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-16 11:26:20.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"652f39f4a665719ac05f4e5c4805260123e24033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-16 11:54:00.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38959e3260ba479c966eaadd672d8191\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e5805dc8f1b145a295c8eb7025476997 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e665e73b39346559cb19430a3e3e219 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5cc1a58e3c48c79538af506971b736 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38dd8ad5c74a45ea819f28391f445976 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef89a2919ef4b10aab3ed17d86ce10c : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/befd2ad676ad4004850abd6c5f6b46f5 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7e6248d0da2472581269b9616843dbc : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af55f9a5cc284474963ae64b7571a56d : SUCCESS in 6m 46s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/847fbdaf8fbe40bcbefd08962f9e8b37 : SUCCESS in 28m 12s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/6d7c0a355b3a4d67a2cf15b906506333 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":15},{"id":"c6804f360dfb861e7960e9afc477d315aec98a6c","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2023-09-17 09:19:47.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b812b72cb2d18d0cb8890fedfad020511459aca2","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-18 09:29:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"94ccbcf702de83075df84d7e0e9f38c2e43ffdca","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-18 09:29:36.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"aa15a4365b0a11329d97906465d89874ac92d360","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2023-09-18 09:54:38.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"431f39633afcf2682250c1ceb80534c9fec1043a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-18 10:23:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8740314dd416bc24c4c2ab8f8570656d332734d9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-18 10:24:36.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b5ed61b207dd3e871dec896d058e80d32dd5cef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-09-19 16:27:07.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"04a6a5b6860095ec18a1de17c7e78e90933bacf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:04:00.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe576621e804e15a4c76ec24da18799\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eebfffb2d7af45e08d56d08e622cfc0d : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1685adbe7a844b89cb62d2955b90981 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/435c52dc60114a0ca58728a767d3fed1 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14ee29ad53de4b059f8c00a6c5dabb46 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e70044d3e9144c0f8f732331b0f1014a : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/606c2c99569b4c519921e27f36b713df : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2823cfe81a844f329d482bdfa606c34c : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd90fb76f4644791bfeb61bee7c6b702 : SUCCESS in 7m 17s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/742b8d4f961b4349b2928cf4b9d8c633 : SUCCESS in 34m 47s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/2b09ef60eda042cfae577966ea4507e0 : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":16},{"id":"37678315b8c71be663c8d4d036ddd7507c800a50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-10-06 15:18:23.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"a73aaba0bb8ac2176319c35872c2d5f98a2c0602","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-10-06 15:22:11.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"35216029a4fe9a992dea59b016e7e40ed465368e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:08:07.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8535623020e4b2bb317b6d5605234cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4a6cc05f69974d4282b545ae6605a8ee : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d186a08f12ea4d3db7cd0aea6b516791 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629da5f6c40448cabb2dbd3621b14a01 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8aeef14e79e42e58e25e634718a9615 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8e32eb7238a4f298b15106ac5b724a8 : SUCCESS in 7m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad100216b0f04a66b5eac8cf9c1e171d : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a83179b960ec442db2a5a0b11acf9b36 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70bc59df612440288f98939c4288b3d9 : SUCCESS in 10m 23s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/90f347053a3a45bd849dd0eea0221f95 : SUCCESS in 44m 03s\n- magnum-container-build https://zuul.opendev.org/t/openstack/build/d8c7e0c4c3c245dea11f0a08f5bbed99 : TIMED_OUT in 30m 52s","accounts_in_message":[],"_revision_number":17},{"id":"0120cb6c18b7f8a9e51bcac903087fddb67502be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-04-15 09:56:49.000000000","message":"Abandoned\n\ndriver will now be out of tree in magnum-capi-helm","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"3993bf7c3bf053f8d6df12681fe229bec29937d1","revisions":{"5408054cc63c10e27c27b0d701ccac9ce4bbee92":{"kind":"REWORK","_number":1,"created":"2021-10-26 15:58:21.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/1"}}},"commit":{"parents":[{"commit":"6951ac4fc6b6178637fac7bd5d05096adb5c6bdc","subject":"Merge \"Add toleration to CSI nodeplugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/6951ac4fc6b6178637fac7bd5d05096adb5c6bdc"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:58:21.000000000","tz":60},"subject":"Add stub cluster_api driver","message":"Add stub cluster_api driver\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/5408054cc63c10e27c27b0d701ccac9ce4bbee92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/5408054cc63c10e27c27b0d701ccac9ce4bbee92"}]},"branch":"refs/heads/master"},"67130b0cd72cda14ccb400e1c0e6d8457aa045f1":{"kind":"REWORK","_number":2,"created":"2022-07-26 17:35:30.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/2"}}},"commit":{"parents":[{"commit":"cb40fb3685a7b4212974052c6f2244439ff8c036","subject":"Add back pep8 test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/cb40fb3685a7b4212974052c6f2244439ff8c036"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2022-07-25 16:49:35.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/67130b0cd72cda14ccb400e1c0e6d8457aa045f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/67130b0cd72cda14ccb400e1c0e6d8457aa045f1"}]},"branch":"refs/heads/master"},"7cfe210d4aa1a5cd1beb489979be50bcec947691":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-15 08:50:03.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/3"}}},"commit":{"parents":[{"commit":"60b338126601be3db42ba65d71649c4e3878b4f3","subject":"Merge \"Fix ingress-controller link in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/60b338126601be3db42ba65d71649c4e3878b4f3"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2022-08-15 08:36:01.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/7cfe210d4aa1a5cd1beb489979be50bcec947691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/7cfe210d4aa1a5cd1beb489979be50bcec947691"}]},"branch":"refs/heads/master"},"e7aa6103e8ed550dd86b5f2085b6cc9df2d7e80e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-08-15 08:56:18.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/4"}}},"commit":{"parents":[{"commit":"60b338126601be3db42ba65d71649c4e3878b4f3","subject":"Merge \"Fix ingress-controller link in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/60b338126601be3db42ba65d71649c4e3878b4f3"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2022-08-15 08:52:44.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/e7aa6103e8ed550dd86b5f2085b6cc9df2d7e80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/e7aa6103e8ed550dd86b5f2085b6cc9df2d7e80e"}]},"branch":"refs/heads/master"},"75a6364d40ef14ff117c7771d329e8d7c9760325":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-11-14 16:11:13.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/5"}}},"commit":{"parents":[{"commit":"d3d28594b3d2c4b2f62258b8f59812e6ccd55497","subject":"Drop mesos driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/d3d28594b3d2c4b2f62258b8f59812e6ccd55497"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2022-11-14 14:14:46.000000000","tz":0},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/75a6364d40ef14ff117c7771d329e8d7c9760325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/75a6364d40ef14ff117c7771d329e8d7c9760325"}]},"branch":"refs/heads/master"},"2e5a93295f546c022052a9a1b024b464aef7074b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-01-29 17:38:04.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/6"}}},"commit":{"parents":[{"commit":"a8a6a37c52b41aa3c2bfbd1628b5c9dc134e60fa","subject":"Support tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a8a6a37c52b41aa3c2bfbd1628b5c9dc134e60fa"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-01-29 17:03:18.000000000","tz":0},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/2e5a93295f546c022052a9a1b024b464aef7074b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/2e5a93295f546c022052a9a1b024b464aef7074b"}]},"branch":"refs/heads/master"},"29777c313da8b9347efbad88f97b4218a583c4ba":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-10 15:18:04.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/7"}}},"commit":{"parents":[{"commit":"78a6ff0d037c5d14fe321fb3a8822e0344b12daf","subject":"Add supported Kubernetes version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/78a6ff0d037c5d14fe321fb3a8822e0344b12daf"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2023-02-10 15:18:04.000000000","tz":0},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/29777c313da8b9347efbad88f97b4218a583c4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/29777c313da8b9347efbad88f97b4218a583c4ba"}]},"branch":"refs/heads/master","description":"Rebase"},"79d6af087dc1b5a56ce7f60fd1229bd92d640c6c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-03-10 16:29:49.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/8"}}},"commit":{"parents":[{"commit":"40d31854916b673fd3d17459c95d82d734db1d82","subject":"Update barbicanclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/40d31854916b673fd3d17459c95d82d734db1d82"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-03-10 16:02:28.000000000","tz":0},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/79d6af087dc1b5a56ce7f60fd1229bd92d640c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/79d6af087dc1b5a56ce7f60fd1229bd92d640c6c"}]},"branch":"refs/heads/master"},"af3c4321680528840023a330217f2311f49ffff8":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-03-20 22:13:24.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/9"}}},"commit":{"parents":[{"commit":"b7092d312a64f78073591dcfa72f2853ff580ca7","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b7092d312a64f78073591dcfa72f2853ff580ca7"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2023-03-20 22:13:24.000000000","tz":0},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/af3c4321680528840023a330217f2311f49ffff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/af3c4321680528840023a330217f2311f49ffff8"}]},"branch":"refs/heads/master","description":"Rebase"},"f0664efc0d0efdbbb2b90d555d4b7d6c645a437b":{"kind":"REWORK","_number":10,"created":"2023-04-25 09:53:24.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/10"}}},"commit":{"parents":[{"commit":"b7092d312a64f78073591dcfa72f2853ff580ca7","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b7092d312a64f78073591dcfa72f2853ff580ca7"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-04-25 08:31:49.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/f0664efc0d0efdbbb2b90d555d4b7d6c645a437b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/f0664efc0d0efdbbb2b90d555d4b7d6c645a437b"}]},"branch":"refs/heads/master"},"005b5a45dd2cbb460c6bda88a9c30d0355693551":{"kind":"REWORK","_number":11,"created":"2023-04-25 10:21:21.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/11"}}},"commit":{"parents":[{"commit":"b7092d312a64f78073591dcfa72f2853ff580ca7","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b7092d312a64f78073591dcfa72f2853ff580ca7"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-04-25 10:13:13.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/005b5a45dd2cbb460c6bda88a9c30d0355693551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/005b5a45dd2cbb460c6bda88a9c30d0355693551"}]},"branch":"refs/heads/master"},"b36b1a829ed3fffed0de5e537524e10b2fb92d0d":{"kind":"REWORK","_number":12,"created":"2023-04-25 15:31:43.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/12"}}},"commit":{"parents":[{"commit":"b7092d312a64f78073591dcfa72f2853ff580ca7","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b7092d312a64f78073591dcfa72f2853ff580ca7"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-04-25 15:27:27.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b36b1a829ed3fffed0de5e537524e10b2fb92d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/b36b1a829ed3fffed0de5e537524e10b2fb92d0d"}]},"branch":"refs/heads/master"},"16462e6ad9afc87c621c115005191bf70be2f594":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-05-12 15:59:13.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/13"}}},"commit":{"parents":[{"commit":"a1252ec5534d9c5f50a1af783b91645515418c63","subject":"Merge \"Add `-p` param to `mkdir` in agent startup script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a1252ec5534d9c5f50a1af783b91645515418c63"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-05-12 15:58:47.000000000","tz":60},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/16462e6ad9afc87c621c115005191bf70be2f594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/16462e6ad9afc87c621c115005191bf70be2f594"}]},"branch":"refs/heads/master"},"0e4e3298615c23c1d58fc146555d3f0b7a47bcdf":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-09-07 01:37:07.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/21/815521/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/14"}}},"commit":{"parents":[{"commit":"327861ac43826327a8682b12df2b37e067a63b28","subject":"Merge \"Remove unused policy rule for Certificate APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/327861ac43826327a8682b12df2b37e067a63b28"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2023-09-05 14:14:23.000000000","tz":600},"subject":"Add a stub cluster_api driver","message":"Add a stub cluster_api driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\ninstead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/0e4e3298615c23c1d58fc146555d3f0b7a47bcdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/0e4e3298615c23c1d58fc146555d3f0b7a47bcdf"}]},"branch":"refs/heads/master"},"c64d45c18fa76a8702154b79e6210178a26a5f19":{"kind":"REWORK","_number":15,"created":"2023-09-16 11:24:37.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/15"}}},"commit":{"parents":[{"commit":"544f59bb3f0f93dab01b4493fb5e43588972601a","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/544f59bb3f0f93dab01b4493fb5e43588972601a"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-09-16 11:20:50.000000000","tz":60},"subject":"Add a stub Cluster API helm driver","message":"Add a stub Cluster API helm driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\nvia helm, instead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nAs this is taking ages to get any reviews and feedback\non these patches, it is also now available to test here:\nhttps://github.com/stackhpc/magnum-capi-helm\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/c64d45c18fa76a8702154b79e6210178a26a5f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/c64d45c18fa76a8702154b79e6210178a26a5f19"}]},"branch":"refs/heads/master"},"723ab08648f4080e284659e8b0bc0d3eb58b887f":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-09-19 16:27:07.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/16"}}},"commit":{"parents":[{"commit":"ce3003a4ab97619d86c87ee7729850fa4e8e60f7","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/ce3003a4ab97619d86c87ee7729850fa4e8e60f7"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-09-19 16:26:56.000000000","tz":60},"subject":"Add a stub Cluster API helm driver","message":"Add a stub Cluster API helm driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\nvia helm, instead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nAs this is taking ages to get any reviews and feedback\non these patches, it is also now available to test here:\nhttps://github.com/stackhpc/magnum-capi-helm\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/723ab08648f4080e284659e8b0bc0d3eb58b887f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/723ab08648f4080e284659e8b0bc0d3eb58b887f"}]},"branch":"refs/heads/master"},"3993bf7c3bf053f8d6df12681fe229bec29937d1":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2023-10-06 15:18:23.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/21/815521/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/21/815521/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/21/815521/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/21/815521/17"}}},"commit":{"parents":[{"commit":"a3276fd53b3a018522e3159137664d9808603514","subject":"Merge \"docs: Change Storyboard links to Launchpad\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/a3276fd53b3a018522e3159137664d9808603514"}]}],"author":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2021-10-26 15:57:09.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-10-06 15:18:23.000000000","tz":0},"subject":"Add a stub Cluster API helm driver","message":"Add a stub Cluster API helm driver\n\nThe plan is to slowly build up the ability to create K8s CRDs\nvia helm, instead of heat templates in a separate driver.\nThe starting point is adding the new experimental driver.\n\nAs this is taking ages to get any reviews and feedback\non these patches, it is also now available to test here:\nhttps://github.com/stackhpc/magnum-capi-helm\n\nstory: 2009780\n\nChange-Id: I32b7a91640f4f65b9662de080682761a2c182055\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/3993bf7c3bf053f8d6df12681fe229bec29937d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/3993bf7c3bf053f8d6df12681fe229bec29937d1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
