)]}'
{"id":"openstack%2Fmagnum~982139","triplet_id":"openstack%2Fmagnum~master~I4afdf73c2c98b4fe4fbf5fd3d4a35e8bcbeb0edf","project":"openstack/magnum","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-26 12:54:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4afdf73c2c98b4fe4fbf5fd3d4a35e8bcbeb0edf","subject":"Replace deprecated object_backport","status":"MERGED","created":"2026-03-25 17:40:22.000000000","updated":"2026-03-26 12:56:11.000000000","submitted":"2026-03-26 12:54:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982139","meta_rev_id":"110d076b3dd61674b5220a93daa28c3d2c1c7c2d","_number":982139,"virtual_id_number":982139,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 12:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-26 11:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-26 12:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-26 12:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-25 18:26:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 11:52:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-03-26 12:36:35.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"state":"REVIEWER"}],"messages":[{"id":"6fcd3dd01386a5a87b189fb5d60589a70012ef24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-25 17:40:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e144a1ef7d4d86ee1f47f63e4422a5bf53ddd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 18:26:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf3b74934ee44de4acc35d2e75587e4e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6088bd735673432cb2e16bf4c0e8a7d0 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b10a47ceec7442892a99972299bea03 : FAILURE in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83eda6a28f9f4a7ca76cecfd4400cb28 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e91b386693294c7ab33c288094ad45b2 : SUCCESS in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44f4f1f7fecb4657acff4a080f5d0c41 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c724d42d7a467a88ea47b8316120f4 : SUCCESS in 9m 15s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/29a9f3c49ca8414986c6a931aa106ce1 : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2e7282e462c589912d89ab159ece25e5bd99f911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-26 00:22:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"04e6798f004b3cdbc52e9f70e2038a2538fb8b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 01:06:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/109cb2ee5c7a481f868ec5dad0351c65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a000941a31ad47ce92b92bd6890b921c : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8c1bc877b104dd493e6c4e1bf74cebe : FAILURE in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ac0e84a1c6549da88aa5aa3a3911973 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47808970a34a4173bc185b18f39fb00d : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f75fce02700c408dab2efd98c155687d : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734365f7008b4a8b82f5a519f0e170e3 : SUCCESS in 8m 20s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/1f6771583a9e42b2b2bc06ce4831afd7 : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7837770277fb47e253cdb955139a32792e754140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-26 10:04:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4592c4910b14c0afd0c75e6c78753abc62f3ecf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 10:54:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dea4b895e3754359bed2c3f1cd3accba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf2f9033a64040d9b3608d39da78e67b : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec67c2197954d45a7bf16f1acb46673 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9446d3f2f9a4811b1c19f1971898ddf : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c66a7169e71548228a6e29b2d2af3b6b : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1bdf5186a2744c63b360fea07562d666 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e301ba9fb074495680fd7b54ff28946b : SUCCESS in 8m 50s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/b37fe35c87df4910b04688567cdd2ec2 : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":3},{"id":"d1bf6131899f4b5bc69262726812b8797bf6fccf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-26 11:52:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5be90e43a49b6f8d074c47f2fbcb08d94d12b672","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2026-03-26 12:36:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e71d881254a91388360daa4bd052ff861100ed17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:36:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"53e41bd2c99d54aab031e35472e2bd3214c7c50c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:54:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdf86cc9990405e8fa651a0eedd8cf7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2648c83a7e94ac0a18a4ea1b9f95702 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99520739099544f49aaf061a491ed96c : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa73b00a34514a6bb85e0dd9034f78b7 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9c8c8fe2d8411b97da6630a745890b : SUCCESS in 4m 08s\n- magnum-tempest-plugin-tests-api https://zuul.opendev.org/t/openstack/build/f64630dc983d42eca9e715fc251d2c8e : SUCCESS in 17m 01s","accounts_in_message":[],"_revision_number":3},{"id":"8d19b4fb12f02fc59a0c9154a1ab56a0db5e789b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:54:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"110d076b3dd61674b5220a93daa28c3d2c1c7c2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:56:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875f19769cd943ca8a55649a44bf43c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c4ea647b9e499d9faf32ffb1c74a68 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d1a9057dada47639515196f30e55b615b0efbc48","revisions":{"7b1f9b1737575c7f293d31d0efc99defa5daf9f7":{"kind":"REWORK","_number":1,"created":"2026-03-25 17:40:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/982139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/39/982139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/39/982139/1"}}},"commit":{"parents":[{"commit":"75ca7d532cd6894ac79a4633bfa19842547d372c","subject":"Merge \"fix TypeError when boot_volume_size label is passed as string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/75ca7d532cd6894ac79a4633bfa19842547d372c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-25 17:37:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-25 17:37:08.000000000","tz":540},"subject":"Replace deprecated object_backport","message":"Replace deprecated object_backport\n\nIt was deprecated in o.vo 0.10.1[1], in favor of the manifest-based\nobject_backport_versions method. This is similar to what was done\nearlier about object_class_action(). [2]\n\n[1] https://review.opendev.org/c/openstack/oslo.versionedobjects/+/224863\n[2] f58adeb2cf42bb61179fa5f084b42ae251649081\n\nChange-Id: I4afdf73c2c98b4fe4fbf5fd3d4a35e8bcbeb0edf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/7b1f9b1737575c7f293d31d0efc99defa5daf9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/7b1f9b1737575c7f293d31d0efc99defa5daf9f7"}]},"branch":"refs/heads/master"},"1fa2d97c35eff2c09b955a079caeadbd4d38320f":{"kind":"REWORK","_number":2,"created":"2026-03-26 00:22:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/982139/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/39/982139/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/39/982139/2"}}},"commit":{"parents":[{"commit":"75ca7d532cd6894ac79a4633bfa19842547d372c","subject":"Merge \"fix TypeError when boot_volume_size label is passed as string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/75ca7d532cd6894ac79a4633bfa19842547d372c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-25 17:37:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-26 00:21:55.000000000","tz":540},"subject":"Replace deprecated object_backport","message":"Replace deprecated object_backport\n\nIt was deprecated in o.vo 0.10.1[1], in favor of the manifest-based\nobject_backport_versions method. This is similar to what was done\nearlier about object_class_action(). [2]\n\n[1] https://review.opendev.org/c/openstack/oslo.versionedobjects/+/224863\n[2] f58adeb2cf42bb61179fa5f084b42ae251649081\n\nChange-Id: I4afdf73c2c98b4fe4fbf5fd3d4a35e8bcbeb0edf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1fa2d97c35eff2c09b955a079caeadbd4d38320f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/1fa2d97c35eff2c09b955a079caeadbd4d38320f"}]},"branch":"refs/heads/master"},"d1a9057dada47639515196f30e55b615b0efbc48":{"kind":"REWORK","_number":3,"created":"2026-03-26 10:04:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/982139/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/magnum","ref":"refs/changes/39/982139/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/magnum refs/changes/39/982139/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/magnum refs/changes/39/982139/3"}}},"commit":{"parents":[{"commit":"75ca7d532cd6894ac79a4633bfa19842547d372c","subject":"Merge \"fix TypeError when boot_volume_size label is passed as string\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/75ca7d532cd6894ac79a4633bfa19842547d372c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-25 17:37:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-26 10:04:47.000000000","tz":540},"subject":"Replace deprecated object_backport","message":"Replace deprecated object_backport\n\nIt was deprecated in o.vo 0.10.1[1], in favor of the manifest-based\nobject_backport_versions method. This is similar to what was done\nearlier about object_class_action(). [2]\n\n[1] https://review.opendev.org/c/openstack/oslo.versionedobjects/+/224863\n[2] f58adeb2cf42bb61179fa5f084b42ae251649081\n\nChange-Id: I4afdf73c2c98b4fe4fbf5fd3d4a35e8bcbeb0edf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/d1a9057dada47639515196f30e55b615b0efbc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/magnum/commit/d1a9057dada47639515196f30e55b615b0efbc48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
