)]}'
{"id":"openstack%2Fmanila-tempest-plugin~1003099","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If4a4a360b2725749292b2bba13b6d936cc931125","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-09-01 12:10:03.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If4a4a360b2725749292b2bba13b6d936cc931125","subject":"Enable share server replication tests in the Container driver","status":"NEW","created":"2026-08-31 18:10:48.000000000","updated":"2026-09-01 12:10:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fdd0eb4348a38a525fb15c85b559d11f16876c93","_number":1003099,"virtual_id_number":1003099,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-01 12:10:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-31 20:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 21:51:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"886eb25aef999adf5154c3bcf03e0e0b4eda5653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-08-31 18:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64deede93dd0780ec9c5627dd39c69c7de7545e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 20:24:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa2404d35b804592bed55a4f8ed196d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f7b0f2887949e0812f094ce5b2a8f8 : SUCCESS in 3m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/646538fa9cd44951a6b8d03a74f5821e : SUCCESS in 1h 13m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bb41faf6905c40f9907e61d0a16c1f30 : SUCCESS in 2h 04m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e25d9363725f4670a57e6454f8b73475 : SUCCESS in 49m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/448f752e6b4b4d9ab235e1f77ee46ee7 : FAILURE in 42m 33s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4a1d92620c4f4b5fa7c55c05cff4b947 : FAILURE in 2h 08m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2ed74574d0424e70b6483838e7a3054e : FAILURE in 1h 02m 36s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/e724c6a4172b456d81ec7530c6e12188 : SUCCESS in 40m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2156109968e9438f80130d89e718048d : SUCCESS in 35m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/02e8c830356e493ba305391aa6543dff : SUCCESS in 38m 17s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/0454d973efc74e6797a80a10052b9d27 : SUCCESS in 56m 45s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/62f189365f1f4ef8b0475654a1666b65 : SUCCESS in 42m 46s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/ae1a7ae30a744a80a8a76dd365a80d66 : SUCCESS in 1h 03m 12s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/36fe5896ed6f4c70aabf3f249e40d4ad : SUCCESS in 19m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5222b79e1cba4c56967a4a505963016c : SUCCESS in 59m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc658800f89de24324ffea9000a8f74f1146fda1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-31 21:51:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/99/1003099/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/dea8588/ : SUCCESS in 1h 25m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"31d3053db98b6f99eef7708876eabbcf36f2e4de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-09-01 10:53:27.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bbd1cd6659d448ae212da5cc5ef7850d8169c310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-09-01 10:54:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fdd0eb4348a38a525fb15c85b559d11f16876c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 12:10:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b888121b8bc443989f6b1d055defb80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f80608fff2e41b88e0c112787535f83 : SUCCESS in 4m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/031347a1589f4ea3ad8e5272b59c3337 : FAILURE in 32m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1b31586fefab43399b16e551b75c1899 : FAILURE in 17m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ae1016e7f27f4679a39ab284d02ca2bb : FAILURE in 34m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9fc25be94a464b45bdee9edeadcac5ec : FAILURE in 20m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4bbafd2834be44cabb28bc328aae3fe0 : FAILURE in 27m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3b8b9c5ccae14945a63fd3246611a724 : FAILURE in 33m 42s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/ceb9681a96d2490299a8e3b38a991ea2 : FAILURE in 41m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2f663a9db1e74a2ea2264816b0b67166 : FAILURE in 23m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0cd72c04a8f14decb50981ee7e15b5ed : FAILURE in 22m 51s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/4846246ea4d949e1928f3fb2f28f4c3c : SUCCESS in 59m 18s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5b806e17f34544d49b5ffb29af0c678a : SUCCESS in 1h 05m 15s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/626e55ca492b4415b906be1ef9231451 : SUCCESS in 1h 10m 07s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/4e27b2cbb5ef40a083444c0012605454 : FAILURE in 26m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/34d78c81d0a149da9db697b4b8d8e02b : FAILURE in 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c0534881db9ecb76479bb375f9205cfcc97cb49","revisions":{"62a0c316cf25c172c0063a2293df09f4fc2bee1d":{"kind":"REWORK","_number":1,"created":"2026-08-31 18:10:48.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/1003099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/99/1003099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/1"}}},"commit":{"parents":[{"commit":"0c6941b1376e15579c46ecda7bed4a63053aa77a","subject":"Recognize WEKAFS access rules as carrying an access_key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c6941b1376e15579c46ecda7bed4a63053aa77a"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-08-31 18:09:15.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-08-31 18:10:39.000000000","tz":-180},"subject":"Enable share server replication tests in the Container driver","message":"Enable share server replication tests in the Container driver\n\nThis change enables the share server replication tests in the\ncontainer driver starting from 2026.2 Hibiscus release.\n\nChange-Id: If4a4a360b2725749292b2bba13b6d936cc931125\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62a0c316cf25c172c0063a2293df09f4fc2bee1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/62a0c316cf25c172c0063a2293df09f4fc2bee1d"}]},"branch":"refs/heads/master"},"10b70920760a50ceab65451071bbc931514b2e12":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-01 10:53:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/1003099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/99/1003099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/2"}}},"commit":{"parents":[{"commit":"0c6941b1376e15579c46ecda7bed4a63053aa77a","subject":"Recognize WEKAFS access rules as carrying an access_key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c6941b1376e15579c46ecda7bed4a63053aa77a"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-08-31 18:09:15.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2026-09-01 10:53:27.000000000","tz":0},"subject":"Enable share server replication tests in the Container driver","message":"Enable share server replication tests in the Container driver\n\nThis change enables the share server replication tests in the\ncontainer driver starting from 2026.2 Hibiscus release.\n\nChange-Id: If4a4a360b2725749292b2bba13b6d936cc931125\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/1003099\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10b70920760a50ceab65451071bbc931514b2e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10b70920760a50ceab65451071bbc931514b2e12"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4c0534881db9ecb76479bb375f9205cfcc97cb49":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-01 10:54:26.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/99/1003099/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/99/1003099/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/99/1003099/3"}}},"commit":{"parents":[{"commit":"0c6941b1376e15579c46ecda7bed4a63053aa77a","subject":"Recognize WEKAFS access rules as carrying an access_key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c6941b1376e15579c46ecda7bed4a63053aa77a"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-08-31 18:09:15.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2026-09-01 10:54:26.000000000","tz":0},"subject":"Enable share server replication tests in the Container driver","message":"Enable share server replication tests in the Container driver\n\nThis change enables the share server replication tests in the\ncontainer driver starting from 2026.2 Hibiscus release.\n\nChange-Id: If4a4a360b2725749292b2bba13b6d936cc931125\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/999135\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c0534881db9ecb76479bb375f9205cfcc97cb49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c0534881db9ecb76479bb375f9205cfcc97cb49"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
