)]}'
{"id":"openstack%2Fmanila-tempest-plugin~1003217","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If48880d73c64e9a895e0bc0912b4680c5e6f1492","project":"openstack/manila-tempest-plugin","branch":"master","topic":"manila-test-image","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-01 09:42:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If48880d73c64e9a895e0bc0912b4680c5e6f1492","subject":"Fix image user/password for manila-test-image","status":"MERGED","created":"2026-09-01 06:08:26.000000000","updated":"2026-09-01 09:42:01.000000000","submitted":"2026-09-01 09:42:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003217-manila-test-image","meta_rev_id":"50d5b433c2a5de00e93cda8184653826f85a5228","_number":1003217,"virtual_id_number":1003217,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 09:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-09-01 08:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-09-01 08:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 07:41:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 08:56:05.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"7152c7e6d82a1bf695050b49bd297e3c7e8302c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 06:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abfecae8fdbd0b05fd7cde81c470fde047c1092a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 07:41:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52bee29eed5a42e2bf0458852a8de655\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa0d55b03cb43739ae168fc8b078de5 : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/01e3382e880641a1849e108ad3a05596 : SUCCESS in 1h 26m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/93f5c0d0a0214ab4a2c0fd1760a774cf : SUCCESS in 1h 31m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5247996e1192492484c578e8cd5930cd : SUCCESS in 1h 09m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a9b18a4e15944a2fbce3912064f55297 : FAILURE in 42m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/197f21dac1e5498aa24b617e5700fd03 : FAILURE in 1h 29m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4e060fc6d52148b3b2c7e6173c1e8058 : SUCCESS in 1h 11m 53s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/77c633ae487c4e0194835cc327d6c2e7 : SUCCESS in 58m 49s (non-voting)\n- manila-tempest-plugin-lustre-scenario https://zuul.opendev.org/t/openstack/build/fce83cab590c4b418eeca15e4aaf32aa : FAILURE in 1h 01m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7b4dfab773734a40b26c4e736bed8bc3 : SUCCESS in 33m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5537ff2df28147cf9b5fb83cd44cc9b9 : SUCCESS in 43m 26s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/459182fa1c43465c9864d64659b11526 : SUCCESS in 1h 08m 25s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5b3e7fe50f42443ba0614b8c89df0287 : SUCCESS in 48m 41s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/3106d9d36b6d4433a3117fb40705f31f : SUCCESS in 1h 08m 41s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/a0b73c7239774f40b53e21260035e982 : SUCCESS in 30m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ce75b34a0fb34dacaa4f952f6411f4a9 : SUCCESS in 48m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5065781745cebfe0cd77f74025d21957a7cdcc1f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 08:09:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58766d3b54493cd8b663d6e71a20232b9de8b6dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 08:09:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67aa85f4471301db7ee607a23f61b73fe83c32cf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-01 08:56:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/1003217/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/f3d17a6/ : SUCCESS in 1h 12m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0bfb8f429e3765216f2faa4698c365f659d6a557","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:42:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9f1443e696443678be3c3a4c63cef2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3407379c21ec4a8db9b923daa9ecda60 : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e5df40c3fa8c4804803c3ba43be58bac : SUCCESS in 1h 30m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d5e72b21e6d949afa39302801aa1a978 : SUCCESS in 42m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7c994a521a3448fc87c1b72a3449bf2f : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":1},{"id":"50d5b433c2a5de00e93cda8184653826f85a5228","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:42:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef22907162128015cdb0c7f88cdb13f6e176806d","revisions":{"ef22907162128015cdb0c7f88cdb13f6e176806d":{"kind":"REWORK","_number":1,"created":"2026-09-01 06:08:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/17/1003217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/1003217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/1003217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/1003217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/1003217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/1003217/1"}}},"commit":{"parents":[{"commit":"86c4bcfb1e9eaa5cee03ab23adafe5c0f9039e92","subject":"Merge \"Add tests for new snapshot_inherit_share_access share type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/86c4bcfb1e9eaa5cee03ab23adafe5c0f9039e92"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 06:07:25.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 06:07:25.000000000","tz":-420},"subject":"Fix image user/password for manila-test-image","message":"Fix image user/password for manila-test-image\n\nThese were recently updated [1]\n\n[1] https://review.opendev.org/c/openstack/manila-test-image/+/1001911\n\nChange-Id: If48880d73c64e9a895e0bc0912b4680c5e6f1492\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef22907162128015cdb0c7f88cdb13f6e176806d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef22907162128015cdb0c7f88cdb13f6e176806d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
