)]}'
{"id":"openstack%2Fmanila-tempest-plugin~1004313","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7fb5a5c173dc12f8a1e08c4d3d6dfed920a78eca","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-09 03:51:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fb5a5c173dc12f8a1e08c4d3d6dfed920a78eca","subject":"Skip tempest jobs for git config update","status":"MERGED","created":"2026-09-06 14:32:35.000000000","updated":"2026-09-09 03:51:28.000000000","submitted":"2026-09-09 03:51:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004313","meta_rev_id":"1ec42438dd75b6b8a687db4d47154b3d8483b676","_number":1004313,"virtual_id_number":1004313,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 03:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-09-08 22:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-09-08 22:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 16:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-06 17:33:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-09-08 22:32:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"da41c6ba420919b48f284dd8013c05f02cfad448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 14:32:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55863d83d5807f80a26a6af95ebe15a4ead4ea3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 16:17:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b257111dc1046129e502bf695cacf0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcda83481cb4f1f865b42564c3215e2 : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/aecdbdcc8c3f4d3ba961dccb67d77387 : SUCCESS in 54m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a0afe31a239b4c5b99dae4b6e01ab42a : SUCCESS in 1h 40m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d7251dcc3867478fac099aacf47c117c : SUCCESS in 1h 19m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/997f36f460a5450ab26ce7936f17e628 : FAILURE in 24m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0d443a7cc0484ffd99212203b7e43245 : FAILURE in 1h 17m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/89e371676ca640cfaec2a4e960d65630 : FAILURE in 56m 19s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/d9f27ca0326342d4921edf7ec283cf92 : SUCCESS in 45m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a58f7a95a5d47ae92dad1505af506b9 : SUCCESS in 20m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/13c444a76704428590999c904462c65b : SUCCESS in 51m 06s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/00d29973ef104c348e79205f5f6bbac1 : SUCCESS in 57m 05s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/11adb3371fad47acbc0c37411bfdb811 : SUCCESS in 41m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/2cdd2faeb1f445e3b11be5fcd26ea474 : SUCCESS in 55m 04s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/2f56c36de14c4ed7b4943abe64d91e1f : SUCCESS in 33m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b33e50a33dc44edb88f1a1056ffac1c7 : FAILURE in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c66999948b0762a67b0c17524b34dc24d67507b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-06 17:33:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/1004313/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/06e3be0/ : SUCCESS in 1h 09m 03s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/1004313/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/dc74365/ : FAILURE in 1h 15m 01s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2b9d416c2db48e31480f1dee278aa0cfcdb5c8f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-08 22:32:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3f141c3dca829dfa4783e6752e2809cc6780fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 22:33:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"427459921e25ecdd110c4f551e952ffe8ab12a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 23:22:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b21f6ccd60ac4f1cadaaeeb6e8badd3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92149c018ced4e3f93c4d48f622e252c : SUCCESS in 3m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/60c15ff6fab044599162c11fb074ebc7 : FAILURE in 13m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a323a1bb5be8465e9afd66d28e36be9c : SUCCESS in 41m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/afd26e3e1dde49ff8e047a7882445963 : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a3a7e713f897f09e4d5185d7fb5fe290a4144562","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 00:25:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6e30df574d62562e8d4c92534736789c4d74738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 02:13:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f41222c4d9e846dab055882c6689ccea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f65d34496e45079096251848ee4ec0 : SUCCESS in 3m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3ff305c985ae4d2bbeeda9fb57a94151 : SUCCESS in 58m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/48160fc68050485f815a65eaf47e9295 : SUCCESS in 47m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb699178ad9f40f093019f3a7c4d1df0 : SUCCESS in 1h 15m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ffc2d75e7cc042258f64cc6195bf8c72 : FAILURE in 20m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/872d4939fadf495ca70715b916ac9af8 : FAILURE in 1h 44m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/92ff527ff476488c90f3a93a76da95e9 : SUCCESS in 1h 00m 57s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/b68153d49c2940839538dac9f62e957d : SUCCESS in 1h 05m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d6ba69f45d7d4af8ba4cf0bdbaa60e0a : SUCCESS in 26m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6c357a80a2c341fdabdbcf7bec64c30a : SUCCESS in 41m 02s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/ce83427e09d040a08cf3b2373df18dfd : SUCCESS in 1h 05m 38s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/e879ffe96a444ce297c546bd211a13ed : SUCCESS in 1h 06m 02s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/34731840b2e34fa4ab353d6ca70a9579 : SUCCESS in 59m 03s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/26b055dd02204fdfb82278e87a289983 : SUCCESS in 28m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f2e5eeecc7e443d5b9078051d483ef48 : SUCCESS in 1h 06m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cc21da13ff2e75e82f829346673a48071a3ef44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 02:37:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f8a185128c7f24a725fec664c364212d5968f83","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-09 03:37:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/1004313/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/566bf13/ : SUCCESS in 1h 21m 02s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/1004313/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/1ddbadd/ : FAILURE in 1h 18m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4feb32982e15536d67e87c037b4729419c843dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 03:51:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0295d7f2e103496b8e318df498813a14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73269ee0c7f475d95df91aab6bd6192 : SUCCESS in 3m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/59327bd8ee094ac68109927fc82cc9f5 : SUCCESS in 1h 12m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97c64928626e4a449b09642b2674950f : SUCCESS in 38m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7f796acb5c824968919d038335760412 : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1ec42438dd75b6b8a687db4d47154b3d8483b676","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 03:51:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecfadcf2eec56d13a81ee0a21435700708160ade","revisions":{"ecfadcf2eec56d13a81ee0a21435700708160ade":{"kind":"REWORK","_number":1,"created":"2026-09-06 14:32:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/1004313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/13/1004313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/1004313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/1004313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/1004313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/13/1004313/1"}}},"commit":{"parents":[{"commit":"0c6941b1376e15579c46ecda7bed4a63053aa77a","subject":"Recognize WEKAFS access rules as carrying an access_key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c6941b1376e15579c46ecda7bed4a63053aa77a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:31:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 14:31:50.000000000","tz":540},"subject":"Skip tempest jobs for git config update","message":"Skip tempest jobs for git config update\n\n... because these files do not affect behaviors directly.\n\nAlso run these tests for requirements.txt update because the file\naffects dependencies installed.\n\nChange-Id: I7fb5a5c173dc12f8a1e08c4d3d6dfed920a78eca\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ecfadcf2eec56d13a81ee0a21435700708160ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ecfadcf2eec56d13a81ee0a21435700708160ade"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
