)]}'
{"zuul.d/project.yaml":[{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"24969cfa6c86f242c1c7ad5e4d13cd60e0c91c39","unresolved":false,"context_lines":[{"line_number":36,"context_line":"        \u0027SYSTEMD_ENV_VARS[\"m-dat\"]\u0027: OS_MANILA_DISABLE_EVENTLET_PATCHING\u003dtrue"},{"line_number":37,"context_line":"    required-projects:"},{"line_number":38,"context_line":"      - openstack/rally-openstack"},{"line_number":39,"context_line":"- job:"},{"line_number":40,"context_line":"    name: manila-tox-genconfig"},{"line_number":41,"context_line":"    parent: openstack-tox"},{"line_number":42,"context_line":"    description: |"},{"line_number":43,"context_line":"      Run tests for manila project."},{"line_number":44,"context_line":""},{"line_number":45,"context_line":"      Uses tox with the ``genconfig`` environment."},{"line_number":46,"context_line":"    post-run: playbooks/manila-tox-genconfig/post.yaml"},{"line_number":47,"context_line":"    vars:"},{"line_number":48,"context_line":"      tox_envlist: genconfig"},{"line_number":49,"context_line":""},{"line_number":50,"context_line":"- job:"},{"line_number":51,"context_line":"    name: manila-rally-no-ss"},{"line_number":52,"context_line":"    parent: rally-task-manila-no-ss"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"0c288d0b_68657302","line":49,"range":{"start_line":39,"start_character":2,"end_line":49,"end_character":0},"updated":"2026-09-15 14:42:00.000000000","message":"Job manila-tox-genconfig in openstack/manila-tempest-plugin is not permitted to shadow job manila-tox-genconfig in openstack/manila","commit_id":"9ad9b82575a5e5f6a624b2c816424367909b7b1e"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"24969cfa6c86f242c1c7ad5e4d13cd60e0c91c39","unresolved":false,"context_lines":[{"line_number":47,"context_line":"    vars:"},{"line_number":48,"context_line":"      tox_envlist: genconfig"},{"line_number":49,"context_line":""},{"line_number":50,"context_line":"- job:"},{"line_number":51,"context_line":"    name: manila-rally-no-ss"},{"line_number":52,"context_line":"    parent: rally-task-manila-no-ss"},{"line_number":53,"context_line":"    irrelevant-files: \u0026irrelevant-files"},{"line_number":54,"context_line":"      - ^(test-|)requirements.txt$"},{"line_number":55,"context_line":"      - ^.*\\.rst$"},{"line_number":56,"context_line":"      - ^api-ref/.*$"},{"line_number":57,"context_line":"      - ^doc/.*$"},{"line_number":58,"context_line":"      - ^manila/tests/.*$"},{"line_number":59,"context_line":"      - ^\\.pre-commit-config\\.yaml$"},{"line_number":60,"context_line":"      - ^releasenotes/.*$"},{"line_number":61,"context_line":"      - ^setup.cfg$"},{"line_number":62,"context_line":"      - ^tools/.*$"},{"line_number":63,"context_line":"      - ^tox.ini$"},{"line_number":64,"context_line":"    vars:"},{"line_number":65,"context_line":"      rally_task: rally-jobs/rally-manila-no-ss.yaml"},{"line_number":66,"context_line":"      devstack_plugins:"},{"line_number":67,"context_line":"        rally-openstack: https://opendev.org/openstack/rally-openstack"},{"line_number":68,"context_line":"    required-projects:"},{"line_number":69,"context_line":"      - openstack/rally-openstack"},{"line_number":70,"context_line":""},{"line_number":71,"context_line":"- job:"},{"line_number":72,"context_line":"    name: manila-rally-ss"},{"line_number":73,"context_line":"    parent: rally-task-manila-ss"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"6111310f_ac8a09f7","line":70,"range":{"start_line":50,"start_character":2,"end_line":70,"end_character":0},"updated":"2026-09-15 14:42:00.000000000","message":"Job manila-rally-no-ss in openstack/manila-tempest-plugin is not permitted to shadow job manila-rally-no-ss in openstack/manila","commit_id":"9ad9b82575a5e5f6a624b2c816424367909b7b1e"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"24969cfa6c86f242c1c7ad5e4d13cd60e0c91c39","unresolved":false,"context_lines":[{"line_number":68,"context_line":"    required-projects:"},{"line_number":69,"context_line":"      - openstack/rally-openstack"},{"line_number":70,"context_line":""},{"line_number":71,"context_line":"- job:"},{"line_number":72,"context_line":"    name: manila-rally-ss"},{"line_number":73,"context_line":"    parent: rally-task-manila-ss"},{"line_number":74,"context_line":"    irrelevant-files: *irrelevant-files"},{"line_number":75,"context_line":"    vars:"},{"line_number":76,"context_line":"      rally_task: rally-jobs/rally-manila.yaml"},{"line_number":77,"context_line":"      devstack_plugins:"},{"line_number":78,"context_line":"        rally-openstack: https://opendev.org/openstack/rally-openstack"},{"line_number":79,"context_line":"    required-projects:"},{"line_number":80,"context_line":"      - openstack/rally-openstack"},{"line_number":81,"context_line":""},{"line_number":82,"context_line":"- job:"},{"line_number":83,"context_line":"    name: manila-rally-ss-compare-branches"},{"line_number":84,"context_line":"    parent: rally-task-manila-ss-compare-branches"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"c6684fbc_6dc858b8","line":81,"range":{"start_line":71,"start_character":2,"end_line":81,"end_character":0},"updated":"2026-09-15 14:42:00.000000000","message":"Job manila-rally-ss in openstack/manila-tempest-plugin is not permitted to shadow job manila-rally-ss in openstack/manila","commit_id":"9ad9b82575a5e5f6a624b2c816424367909b7b1e"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"24969cfa6c86f242c1c7ad5e4d13cd60e0c91c39","unresolved":false,"context_lines":[{"line_number":97,"context_line":"    required-projects:"},{"line_number":98,"context_line":"      - openstack/rally-openstack"},{"line_number":99,"context_line":""},{"line_number":100,"context_line":"- job:"},{"line_number":101,"context_line":"  # Security testing for known issues"},{"line_number":102,"context_line":"    name: manila-tox-bandit"},{"line_number":103,"context_line":"    parent: openstack-tox"},{"line_number":104,"context_line":"    timeout: 2400"},{"line_number":105,"context_line":"    vars:"},{"line_number":106,"context_line":"      tox_envlist: bandit"},{"line_number":107,"context_line":"    irrelevant-files: *irrelevant-files"},{"line_number":108,"context_line":""},{"line_number":109,"context_line":"- job:"},{"line_number":110,"context_line":"    name: manila-tox-py3-threading"},{"line_number":111,"context_line":"    parent: openstack-tox-py313"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"426c6753_73695313","line":108,"range":{"start_line":100,"start_character":2,"end_line":108,"end_character":0},"updated":"2026-09-15 14:42:00.000000000","message":"Job manila-tox-bandit in openstack/manila-tempest-plugin is not permitted to shadow job manila-tox-bandit in openstack/manila","commit_id":"9ad9b82575a5e5f6a624b2c816424367909b7b1e"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"24969cfa6c86f242c1c7ad5e4d13cd60e0c91c39","unresolved":false,"context_lines":[{"line_number":106,"context_line":"      tox_envlist: bandit"},{"line_number":107,"context_line":"    irrelevant-files: *irrelevant-files"},{"line_number":108,"context_line":""},{"line_number":109,"context_line":"- job:"},{"line_number":110,"context_line":"    name: manila-tox-py3-threading"},{"line_number":111,"context_line":"    parent: openstack-tox-py313"},{"line_number":112,"context_line":"    description: |"},{"line_number":113,"context_line":"      Run unit tests with the native threading backend"},{"line_number":114,"context_line":"      (OS_MANILA_DISABLE_EVENTLET_PATCHING\u003dTrue)."},{"line_number":115,"context_line":"    vars:"},{"line_number":116,"context_line":"      tox_envlist: py3-threading"}],"source_content_type":"text/x-yaml","patch_set":1,"id":"8dd2c867_a6f16042","line":116,"range":{"start_line":109,"start_character":2,"end_line":116,"end_character":0},"updated":"2026-09-15 14:42:00.000000000","message":"Job manila-tox-py3-threading in openstack/manila-tempest-plugin is not permitted to shadow job manila-tox-py3-threading in openstack/manila","commit_id":"9ad9b82575a5e5f6a624b2c816424367909b7b1e"}]}
