)]}'
{"id":"openstack%2Fmanila-tempest-plugin~1006716","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I66d938628401efdc770598a9253814282e528fe9","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-09-22 16:47:42.000000000","reason":"Someone else replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-22 16:27:23.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2026-09-22 16:27:23.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I66d938628401efdc770598a9253814282e528fe9","subject":"Add scenario test for ensuring RO/RW grants","status":"NEW","created":"2026-09-22 12:56:11.000000000","updated":"2026-09-22 16:47:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f372c605246a0ecde32cc953c1e984b33b3cc509","_number":1006716,"virtual_id_number":1006716,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-22 15:26:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 15:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-22 16:27:23.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2026-09-22 16:27:23.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-09-22 16:47:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"918b4a240cd8caed840e615fc09ae7fb5b065255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-09-22 12:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"874149a8fb7c7535beabb5bc00b0a1d40429bf08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 15:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ceb3cc0cc4b423f8652046f996211b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0cc2d15bea4f3cac999e8c3f80cfb0 : SUCCESS in 2m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b96ba25f288646dc935f3582c035f965 : SUCCESS in 54m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/499003a7256e469ea679cd7318be32b7 : FAILURE in 2h 24m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ff2ca0f399074c58803666baf7cd7c1f : SUCCESS in 1h 08m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/094606253e844db09d1c06d1d28f12b0 : FAILURE in 26m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/deddcb82e98c4ced8206eca037bd2577 : FAILURE in 1h 46m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/588b54693cb94136a7c9d47149027a6e : SUCCESS in 58m 40s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/5f8aef65e79a43519560f8fec46834a7 : SUCCESS in 49m 32s (non-voting)\n- manila-tempest-plugin-lustre-scenario https://zuul.opendev.org/t/openstack/build/e32e201a86b34c0ab0341b955e04cdfc : FAILURE in 38m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9d2a05f279c34abc9094f9931f528471 : SUCCESS in 25m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/506e451de2ed4973a7fe8b5452ae688c : SUCCESS in 39m 03s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/637b1789bf0c41e0ae5a3a27fa491549 : SUCCESS in 1h 10m 13s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/ce0f458c746249078796779b823afd9c : SUCCESS in 1h 07m 28s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/57d89133e0b1451b96ca2ad911f23475 : SUCCESS in 1h 03m 23s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/4eaaeba4a5314fe48fae3b247b2b0b32 : SUCCESS in 34m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bcd5d9862f46439eb88f2aacdf75be45 : SUCCESS in 57m 24s (non-voting)\n- manila-tempest-plugin-multinode-virtiofs-lvm https://zuul.opendev.org/t/openstack/build/884afeb67dcd4a7cad5b9297b4ac153d : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e68b8a866ebc28d1895d723d946999b07c57d416","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-09-22 16:27:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f372c605246a0ecde32cc953c1e984b33b3cc509","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-22 16:47:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/1006716/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/b50afea/ : SUCCESS in 1h 18m 49s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/1006716/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/e7984ae/ : POST_FAILURE in 46m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4255e391557ce534e7e723e121a5abe8af7b63f","revisions":{"e4255e391557ce534e7e723e121a5abe8af7b63f":{"kind":"REWORK","_number":1,"created":"2026-09-22 12:56:11.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/16/1006716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/1006716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/1006716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/1006716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/1006716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/1006716/1"}}},"commit":{"parents":[{"commit":"f8ef0319a9c072b7a56c90d7e9a2a8a74d5cc86e","subject":"Merge \"Drop MANILA_INSTALL_TEMPEST_PLUGIN_SYSTEMWIDE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f8ef0319a9c072b7a56c90d7e9a2a8a74d5cc86e"}]}],"author":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-09-22 12:53:51.000000000","tz":-180},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2026-09-22 12:56:05.000000000","tz":-180},"subject":"Add scenario test for ensuring RO/RW grants","message":"Add scenario test for ensuring RO/RW grants\n\nWe have complex scenario tests that cover a gamut of scenarios and\nhelp us ensuring functionailities are unchanged.\n\nHowever, we lacked a scenario test to ensure that the storage\nbackends are enforcing the deny access by default, as well as\nensuring that after the access rule is removed, we are no longer\nable to read/write data.\n\nThis new test fixes covers that gap.\n\nChange-Id: I66d938628401efdc770598a9253814282e528fe9\nSigned-off-by: Carlos da Silva \u003cces.eduardo98@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4255e391557ce534e7e723e121a5abe8af7b63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4255e391557ce534e7e723e121a5abe8af7b63f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
