)]}'
{"id":"openstack%2Fmanila-tempest-plugin~1007573","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I74f530f5a405628192889b2f8a6bb8af5230d4cf","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-26 05:20:39.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I74f530f5a405628192889b2f8a6bb8af5230d4cf","subject":"Document choices using native interface from oslo.config","status":"NEW","created":"2026-09-26 02:49:02.000000000","updated":"2026-09-26 05:20:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c8d4337efd8faa4699963a69ffaebe8a5c67f412","_number":1007573,"virtual_id_number":1007573,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-26 05:20:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-26 05:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e5942d8121fcc7ed898d5f011783ac8f8daa479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-26 02:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8d4337efd8faa4699963a69ffaebe8a5c67f412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-26 05:20:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab2cb0c472c849bf8897a1ce57d890ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c079895b7fb48ae8d78932b54950653 : FAILURE in 3m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/061b53ad5f204a04b7eacf658dd7599b : SUCCESS in 1h 03m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ce41665fcd3b40f6bbadb633c4f925a4 : FAILURE in 1h 59m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3c8f071bbbe34583a05f8ec3a2bfd8ef : SUCCESS in 33m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/637b8cdd43a5478b8192d5814d2c6721 : FAILURE in 23m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b6288651c5244b50981c1eeb52865bb2 : FAILURE in 1h 22m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/969c3eec3efd4522a96245ab1dd48e73 : SUCCESS in 49m 18s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/05eaff28978a4bed9a6687bf410d3c59 : SUCCESS in 57m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2f7b01579e534c9fa48b355e50699e69 : SUCCESS in 20m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6a01a2b7a17b4dc78fc304374055d43c : SUCCESS in 35m 46s\n- manila-tempest-plugin-lvm-2026.2 https://zuul.opendev.org/t/openstack/build/e3544d5431b94fa08c4a635acddcb546 : SUCCESS in 1h 12m 21s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/f6b957501d154b9bab2cc4d3e9fc41d4 : SUCCESS in 1h 03m 33s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/56ef993dc7684aec93aa1c290c8a7216 : SUCCESS in 1h 03m 29s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/66603d2853e245bcb4b1b5c392309227 : SUCCESS in 1h 06m 15s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/142b36767f27448caf8bb05862409789 : SUCCESS in 23m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/11f73715afc640b39a1e175d396c48ab : SUCCESS in 54m 30s (non-voting)\n- manila-tempest-plugin-multinode-virtiofs-lvm https://zuul.opendev.org/t/openstack/build/a7af3ccb82c449e39a1c679fd37097ad : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78634fa65ea66498663b79036524f3b2f6c6ab68","revisions":{"78634fa65ea66498663b79036524f3b2f6c6ab68":{"kind":"REWORK","_number":1,"created":"2026-09-26 02:49:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/1007573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/73/1007573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/1007573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/1007573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/1007573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/73/1007573/1"}}},"commit":{"parents":[{"commit":"775ee2630411519a3c4e5bed2ea8b7e81b02e511","subject":"Merge \"Add cold migration tests for virtiofs shares\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/775ee2630411519a3c4e5bed2ea8b7e81b02e511"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-26 02:47:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-26 02:48:46.000000000","tz":540},"subject":"Document choices using native interface from oslo.config","message":"Document choices using native interface from oslo.config\n\nChange-Id: I74f530f5a405628192889b2f8a6bb8af5230d4cf\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78634fa65ea66498663b79036524f3b2f6c6ab68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78634fa65ea66498663b79036524f3b2f6c6ab68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
