)]}'
{"id":"openstack%2Fmanila-tempest-plugin~531568","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4998bcac7a1d902c17d2add00cd42560bf6bbf75","project":"openstack/manila-tempest-plugin","branch":"master","topic":"create-shrink","hashtags":[],"change_id":"I4998bcac7a1d902c17d2add00cd42560bf6bbf75","subject":"Scenario test: Create/extend share and write data","status":"MERGED","created":"2018-01-06 18:49:26.000000000","updated":"2019-06-22 13:26:05.000000000","submitted":"2019-06-20 11:28:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"531568-1561030136859-1a2baadc","meta_rev_id":"298d2952e36eb777ebdb6e3aad8e14f2b4c70652","_number":531568,"virtual_id_number":531568,"owner":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"value":0,"date":"2019-06-18 13:49:43.000000000","_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},{"value":0,"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},{"value":0,"date":"2019-06-20 03:10:01.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-06-20 11:28:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"value":0,"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2019-06-20 00:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"value":0,"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},{"value":0,"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 10:24:21.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"value":0,"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"value":0,"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},{"value":0,"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 10:24:21.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"value":0,"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},{"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-14 07:59:53.000000000","updated_by":{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},"reviewer":{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},"state":"REVIEWER"},{"updated":"2018-02-07 07:57:05.000000000","updated_by":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"reviewer":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"state":"REVIEWER"},{"updated":"2019-04-25 18:48:24.000000000","updated_by":{"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"},"reviewer":{"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"},"state":"REVIEWER"},{"updated":"2019-04-25 20:10:50.000000000","updated_by":{"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},"reviewer":{"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},"state":"REVIEWER"},{"updated":"2019-04-29 06:54:41.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2019-06-18 13:49:43.000000000","updated_by":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"reviewer":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"state":"REVIEWER"},{"updated":"2019-06-20 00:02:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-06-20 03:10:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-20 10:24:21.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-06-20 11:28:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a2fdda221d9019b5f14ef9b798828a9df2768e3","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-06 18:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89e313411aa5a289ebc2295a11043c2c751b626c","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-06 18:49:43.000000000","message":"Topic set to scenario-tests","accounts_in_message":[],"_revision_number":1},{"id":"5aa73e2438d0c017637f4db1eab5a5e1b4f18997","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-06 20:03:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/1/check/openstack-tox-pep8/8948497/ : FAILURE in 3m 45s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-mysql-generic/a06e80e/ : FAILURE in 42m 09s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-postgres-container/5f08bf7/ : SUCCESS in 30m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-postgres-zfsonlinux/f6ddc22/ : FAILURE in 1h 04m 30s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/eaaf626/ : FAILURE in 35m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-generic-no-share-servers/b8a0f20/ : FAILURE in 40m 45s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-scenario/0d259ea/ : FAILURE in 1h 11m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/1/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/f2c981b/ : SUCCESS in 1h 08m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/1/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/1f0410c/ : SUCCESS in 1h 02m 12s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/1/check/manila-tempest-dsvm-glusterfs-nfs/4f235bc/ : SUCCESS in 57m 20s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/1/check/manila-tempest-minimal-dsvm-dummy/d20de16/ : SUCCESS in 32m 09s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/1/check/manila-tempest-minimal-dsvm-lvm-centos-7/d4264cd/ : FAILURE in 55m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1b9b790bb2d1a97532525b578c0c972a6f8efdf6","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-06 20:46:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"564437f7e2af61c8ca069199cffa78e8dd1cb69a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-06 22:04:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/2/check/openstack-tox-pep8/e834b0e/ : FAILURE in 4m 45s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-mysql-generic/e71281e/ : FAILURE in 40m 39s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-postgres-container/034b946/ : SUCCESS in 35m 55s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-postgres-zfsonlinux/d2c3d30/ : FAILURE in 53m 05s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/ff42162/ : FAILURE in 40m 06s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-generic-no-share-servers/01b32e7/ : FAILURE in 43m 13s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-scenario/ddaff35/ : FAILURE in 1h 17m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/2/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/511eb4f/ : SUCCESS in 1h 06m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/2/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/4b3b305/ : SUCCESS in 1h 02m 25s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/2/check/manila-tempest-dsvm-glusterfs-nfs/4a756fa/ : SUCCESS in 1h 07m 53s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/2/check/manila-tempest-minimal-dsvm-dummy/1e57d58/ : SUCCESS in 30m 58s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/2/check/manila-tempest-minimal-dsvm-lvm-centos-7/1fb617f/ : FAILURE in 55m 41s","accounts_in_message":[],"_revision_number":2},{"id":"8f6c3b0d6dad45788b29a19585a45a60d39c1ded","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-06 22:25:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d4fa51d3d97f5b8a34022401c5d90b6b07290b9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-06 23:47:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/3/check/openstack-tox-pep8/8ea8c13/ : FAILURE in 3m 31s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-mysql-generic/ea1ee29/ : FAILURE in 55m 33s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-postgres-container/a997129/ : SUCCESS in 41m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-postgres-zfsonlinux/7d79dbd/ : FAILURE in 54m 31s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-postgres-generic-singlebackend/c3aff0a/ : FAILURE in 37m 16s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-generic-no-share-servers/f2ca5b7/ : FAILURE in 38m 44s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-scenario/9ad4745/ : FAILURE in 1h 21m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/3/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/df7e080/ : SUCCESS in 1h 00m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/3/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/ba69c28/ : SUCCESS in 1h 15m 33s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/3/check/manila-tempest-dsvm-glusterfs-nfs/839fef2/ : SUCCESS in 55m 34s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/3/check/manila-tempest-minimal-dsvm-dummy/d5df26c/ : SUCCESS in 33m 15s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/3/check/manila-tempest-minimal-dsvm-lvm-centos-7/2d7ec47/ : FAILURE in 1h 01m 14s","accounts_in_message":[],"_revision_number":3},{"id":"858ddc779e2ee1b7152b90cb9345d34732066bba","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-07 08:47:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3fb1bb5e3faa4e24b632f7e03d3a4ac6b2d5d605","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-07 10:52:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/4/check/openstack-tox-pep8/f3736b4/ : SUCCESS in 4m 48s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-mysql-generic/807fda1/ : SUCCESS in 52m 06s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-postgres-container/4fdb5fd/ : SUCCESS in 34m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-postgres-zfsonlinux/523abba/ : FAILURE in 55m 32s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-postgres-generic-singlebackend/dab36ec/ : SUCCESS in 41m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-generic-no-share-servers/7b5735f/ : FAILURE in 33m 05s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-scenario/e4382c7/ : FAILURE in 1h 00m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/4/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/3234b1d/ : SUCCESS in 55m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/4/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/7b61f28/ : SUCCESS in 1h 05m 45s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/4/check/manila-tempest-dsvm-glusterfs-nfs/e28c816/ : SUCCESS in 57m 28s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/4/check/manila-tempest-minimal-dsvm-dummy/3c5acce/ : POST_FAILURE in 2h 02m 32s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/4/check/manila-tempest-minimal-dsvm-lvm-centos-7/9b415b1/ : FAILURE in 54m 38s","accounts_in_message":[],"_revision_number":4},{"id":"713b3734409aa2c8b196f077767dd28579de1f22","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-07 12:47:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"58c2d7681b1b894736dc1f04536d7f18b92864d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-07 13:50:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/5/check/openstack-tox-pep8/4988093/ : SUCCESS in 3m 12s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-mysql-generic/139195e/ : SUCCESS in 54m 01s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-postgres-container/babeda8/ : SUCCESS in 33m 10s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-postgres-zfsonlinux/887c370/ : SUCCESS in 52m 24s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-postgres-generic-singlebackend/e34dab9/ : SUCCESS in 47m 38s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-generic-no-share-servers/3818ba2/ : SUCCESS in 48m 58s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-scenario/e78222e/ : FAILURE in 1h 01m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/5/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/b4fab92/ : SUCCESS in 57m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/5/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/c45d5a7/ : FAILURE in 27m 42s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/5/check/manila-tempest-dsvm-glusterfs-nfs/923f2f3/ : SUCCESS in 56m 26s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/5/check/manila-tempest-minimal-dsvm-dummy/447aa8d/ : SUCCESS in 34m 14s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/5/check/manila-tempest-minimal-dsvm-lvm-centos-7/e9fe553/ : SUCCESS in 53m 57s","accounts_in_message":[],"_revision_number":5},{"id":"30f1c162e953c8cab957b00dda7823df68c2639e","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-07 18:49:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c0a8548d7b392e56dda76b30f3c117a5e8692676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-07 22:33:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/6/check/openstack-tox-pep8/f44f924/ : FAILURE in 4m 08s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/6/check/manila-tempest-dsvm-mysql-generic/3fece58/ : SUCCESS in 58m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/6/check/manila-tempest-dsvm-postgres-container/869f996/ : SUCCESS in 33m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/6/check/manila-tempest-dsvm-postgres-zfsonlinux/109bad5/ : FAILURE in 56m 32s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend finger://ze01.openstack.org/f22fbcc595b34900876bf270e35cac86 : RETRY_LIMIT in 1m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/6/check/manila-tempest-dsvm-generic-no-share-servers/84a116c/ : SUCCESS in 48m 20s (non-voting)\n- manila-tempest-dsvm-scenario finger://ze09/53b6845f38074c0b8e830a99995a4267 : RETRY_LIMIT in 8s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/6/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/4b20fe4/ : SUCCESS in 1h 04m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/6/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/19bfe63/ : SUCCESS in 1h 10m 48s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs finger://ze01.openstack.org/59cf7a31bbc8484aa3cd4a6d415937ff : RETRY_LIMIT in 1m 45s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/6/check/manila-tempest-minimal-dsvm-dummy/7edb84e/ : SUCCESS in 34m 06s\n- manila-tempest-minimal-dsvm-lvm-centos-7 manila-tempest-minimal-dsvm-lvm-centos-7 : RETRY_LIMIT","accounts_in_message":[],"_revision_number":6},{"id":"6b69a506bda59f3db3161ca864d197106d8f90f9","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-08 07:20:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f82d6271f136bec1138f2b8aa0e5c55ce765a77d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-08 08:41:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/7/check/openstack-tox-pep8/9a1059d/ : SUCCESS in 4m 52s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-mysql-generic/5bc4745/ : SUCCESS in 56m 01s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-postgres-container/698bc5b/ : SUCCESS in 32m 23s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-postgres-zfsonlinux/cce2204/ : FAILURE in 52m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-postgres-generic-singlebackend/8417e64/ : SUCCESS in 39m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-generic-no-share-servers/6ac1ceb/ : FAILURE in 58m 55s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-scenario/74b47c4/ : FAILURE in 1h 09m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/7/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/0243006/ : SUCCESS in 1h 01m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/7/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/6485912/ : SUCCESS in 1h 14m 36s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/7/check/manila-tempest-dsvm-glusterfs-nfs/e11de8c/ : SUCCESS in 55m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/7/check/manila-tempest-minimal-dsvm-dummy/536b70d/ : SUCCESS in 33m 48s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/7/check/manila-tempest-minimal-dsvm-lvm-centos-7/942b571/ : FAILURE in 55m 30s","accounts_in_message":[],"_revision_number":7},{"id":"a0ed1f358c702a39d460360ee36a728dcd30b7eb","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-09 08:45:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8d368b5148aa9f3813285676f2d4be81ea8d4fcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-09 13:24:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/8/check/openstack-tox-pep8/7acc681/ : FAILURE in 4m 30s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-mysql-generic/ee460d3/ : TIMED_OUT in 4h 34m 59s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-postgres-container/6b9b4e6/ : SUCCESS in 39m 21s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-postgres-zfsonlinux/8d52dbb/ : FAILURE in 54m 23s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-postgres-generic-singlebackend/d56e51c/ : SUCCESS in 36m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-generic-no-share-servers/5e596de/ : FAILURE in 1h 06m 59s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-scenario/90514f9/ : FAILURE in 1h 03m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/8/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/5e1ce65/ : SUCCESS in 1h 00m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/8/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/e02c879/ : SUCCESS in 58m 16s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/8/check/manila-tempest-dsvm-glusterfs-nfs/13430c2/ : TIMED_OUT in 1h 25m 00s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/8/check/manila-tempest-minimal-dsvm-dummy/0cb93e2/ : SUCCESS in 35m 07s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/8/check/manila-tempest-minimal-dsvm-lvm-centos-7/3aecf9d/ : FAILURE in 1h 06m 09s","accounts_in_message":[],"_revision_number":8},{"id":"93f92eefd70ded77a67796f913ce784c03003642","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-09 18:24:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6f8137a864958186c79a475bd6cc7ff6af6c0665","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-09 20:50:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/9/check/openstack-tox-pep8/c44db8d/ : SUCCESS in 4m 05s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-mysql-generic/824c7e5/ : SUCCESS in 59m 25s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-postgres-container/dd1e3c0/ : SUCCESS in 31m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-postgres-zfsonlinux/4df05cb/ : SUCCESS in 55m 27s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-postgres-generic-singlebackend/08dc87a/ : TIMED_OUT in 1h 39m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-generic-no-share-servers/b387602/ : SUCCESS in 54m 30s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-scenario/e428996/ : FAILURE in 1h 08m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/9/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/582db9c/ : SUCCESS in 53m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/9/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/5bf800f/ : SUCCESS in 56m 46s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/9/check/manila-tempest-dsvm-glusterfs-nfs/922c0f5/ : SUCCESS in 58m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/9/check/manila-tempest-minimal-dsvm-dummy/e0f970d/ : SUCCESS in 39m 02s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/9/check/manila-tempest-minimal-dsvm-lvm-centos-7/9708862/ : SUCCESS in 1h 08m 38s","accounts_in_message":[],"_revision_number":9},{"id":"48313c4431e2fc556ca5bd7993a46109ac4d3400","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-01-10 14:30:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0e049503bab00e94d53e473a5b861f7fcd2a7407","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-10 16:54:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/10/check/openstack-tox-pep8/5eae4c8/ : SUCCESS in 5m 12s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-mysql-generic/5eca01b/ : SUCCESS in 47m 26s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-postgres-container/3bc2faf/ : SUCCESS in 42m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-postgres-zfsonlinux/b0fcc42/ : SUCCESS in 1h 08m 52s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-postgres-generic-singlebackend/5faf7ab/ : SUCCESS in 41m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-generic-no-share-servers/3e6141b/ : SUCCESS in 51m 07s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-scenario/e3988c9/ : FAILURE in 1h 09m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/10/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/e6b5941/ : SUCCESS in 1h 00m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/10/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/e0ecee4/ : SUCCESS in 1h 09m 55s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/10/check/manila-tempest-dsvm-glusterfs-nfs/3719a5c/ : SUCCESS in 1h 00m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/10/check/manila-tempest-minimal-dsvm-dummy/83d1074/ : SUCCESS in 38m 25s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/10/check/manila-tempest-minimal-dsvm-lvm-centos-7/7daa642/ : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":10},{"id":"66d23b20350d833f07e2da984be7591ee8c0e026","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2018-01-14 07:59:19.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bd31a1158bed8f6ce1670d18f72a7b8a43bb6e38","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2018-01-30 21:07:42.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"5608923837c33c778a2c27208beeeeb88dd1d8b4","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2018-01-30 21:10:07.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"2d2f326f6b1b453402d5848ed077a642907a7f41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-31 01:54:41.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/12/check/openstack-tox-pep8/20e6cb7/ : SUCCESS in 5m 10s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-mysql-generic/bb44f32/ : FAILURE in 1h 10m 12s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-postgres-container/10a5e76/ : TIMED_OUT in 1h 39m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-postgres-zfsonlinux/bc7c413/ : SUCCESS in 51m 47s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-postgres-generic-singlebackend/cb56b63/ : SUCCESS in 40m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-generic-no-share-servers/6ea7937/ : SUCCESS in 46m 15s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-scenario/58f7db1/ : FAILURE in 1h 02m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/12/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/2b08909/ : TIMED_OUT in 4h 37m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/12/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/2a5a463/ : SUCCESS in 1h 09m 36s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/12/check/manila-tempest-dsvm-glusterfs-nfs/121a0c6/ : FAILURE in 14m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/12/check/manila-tempest-minimal-dsvm-dummy/c0a6652/ : SUCCESS in 33m 04s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/12/check/manila-tempest-minimal-dsvm-lvm-centos-7/2a466a1/ : FAILURE in 58m 43s","accounts_in_message":[],"_revision_number":12},{"id":"755a04fbfc1ac9b02c26aa1d7e26798b80e0d6da","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-02-03 18:28:48.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"acb8c1ef1d1a6a06fa1524ee112a4a60e986b788","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-02-03 18:30:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"90b320919ace368eebf049823f37cbdb5c02acc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-03 20:04:37.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/14/check/openstack-tox-pep8/a45d420/ : SUCCESS in 4m 37s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-mysql-generic/014eee5/ : SUCCESS in 51m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-postgres-container/4e05023/ : SUCCESS in 38m 01s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-postgres-zfsonlinux/5041785/ : FAILURE in 49m 29s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-postgres-generic-singlebackend/248233b/ : SUCCESS in 43m 16s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-generic-no-share-servers/3e9f297/ : FAILURE in 34m 00s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-scenario/3b394ff/ : FAILURE in 1h 02m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/14/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/6500934/ : SUCCESS in 1h 16m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/14/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/b7d9b33/ : FAILURE in 1h 24m 44s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/14/check/manila-tempest-dsvm-glusterfs-nfs/08d9347/ : FAILURE in 17m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/14/check/manila-tempest-minimal-dsvm-dummy/fb20c4e/ : SUCCESS in 35m 15s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/14/check/manila-tempest-minimal-dsvm-lvm-centos-7/1c6e8f5/ : FAILURE in 1h 05m 57s","accounts_in_message":[],"_revision_number":14},{"id":"b9b201962509bdff8c679cb0166db7631f5371be","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-02-03 20:43:08.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"579e51d9bbabd6c979790862bd45b4c4a19a172f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-03 22:03:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/15/check/openstack-tox-pep8/edc08c0/ : SUCCESS in 4m 12s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-mysql-generic/369f04a/ : SUCCESS in 48m 44s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-postgres-container/8891ed0/ : SUCCESS in 46m 31s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-postgres-zfsonlinux/54de0e4/ : SUCCESS in 50m 41s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-postgres-generic-singlebackend/939b66d/ : SUCCESS in 44m 16s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-generic-no-share-servers/334d5ef/ : SUCCESS in 1h 01m 43s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-scenario/cd96c99/ : FAILURE in 1h 14m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/15/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/7cf9bc8/ : SUCCESS in 1h 00m 55s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/15/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/c55c0e6/ : SUCCESS in 1h 01m 11s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/15/check/manila-tempest-dsvm-glusterfs-nfs/d36b2ca/ : FAILURE in 13m 53s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/15/check/manila-tempest-minimal-dsvm-dummy/19a3b51/ : SUCCESS in 34m 04s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/15/check/manila-tempest-minimal-dsvm-lvm-centos-7/ae244a6/ : FAILURE in 54m 58s","accounts_in_message":[],"_revision_number":15},{"id":"0d66579585d79d034ab9d6de0ddf0773047b92be","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-02-04 17:00:07.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6502774d0af0797c90d20da730ba537c72f22aad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-04 18:12:03.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/16/check/openstack-tox-pep8/0a02907/ : SUCCESS in 3m 40s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-mysql-generic/a065e69/ : SUCCESS in 49m 31s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-postgres-container/794146e/ : SUCCESS in 39m 23s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-postgres-zfsonlinux/288e75f/ : SUCCESS in 45m 30s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-postgres-generic-singlebackend/02acc70/ : SUCCESS in 41m 41s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-generic-no-share-servers/4353167/ : FAILURE in 56m 41s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-scenario/6020118/ : FAILURE in 55m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/16/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/6f38e9f/ : SUCCESS in 58m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/16/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/2a8f562/ : SUCCESS in 1h 06m 23s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/16/check/manila-tempest-dsvm-glusterfs-nfs/95ea3d3/ : FAILURE in 12m 48s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/16/check/manila-tempest-minimal-dsvm-dummy/6284f40/ : SUCCESS in 31m 52s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/16/check/manila-tempest-minimal-dsvm-lvm-centos-7/26ce2e3/ : FAILURE in 48m 02s","accounts_in_message":[],"_revision_number":16},{"id":"22d1bd045b8d73fb70bdcf34a7973b8267fce26e","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-02-04 20:38:45.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7b404ebbf89de044672041683c66b0b07ea41a08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-04 21:43:01.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/17/check/openstack-tox-pep8/125cbd6/ : SUCCESS in 4m 25s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-mysql-generic/e015581/ : SUCCESS in 55m 34s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-postgres-container/45b34dc/ : SUCCESS in 33m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-postgres-zfsonlinux/cd0c2ad/ : SUCCESS in 51m 13s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-postgres-generic-singlebackend/92dca67/ : SUCCESS in 40m 59s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-generic-no-share-servers/3e4657e/ : FAILURE in 33m 03s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-scenario/d3fc010/ : FAILURE in 57m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/7339c82/ : SUCCESS in 1h 01m 48s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/2010371/ : SUCCESS in 1h 01m 01s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-glusterfs-nfs/1718c35/ : FAILURE in 13m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-dummy/d99126e/ : SUCCESS in 29m 56s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-lvm-centos-7/3452f45/ : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":17},{"id":"dd4cc6d1499b7cbf6a5826bb3f26bcc9cedd7072","author":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"date":"2018-02-04 21:59:01.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"e612b7a95cc6eea9f94c5a86691ce20c113d01c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-04 23:27:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/17/check/openstack-tox-pep8/a489548/ : SUCCESS in 4m 04s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-mysql-generic/6cbfe31/ : SUCCESS in 55m 41s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-postgres-container/8cac57f/ : SUCCESS in 36m 24s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-postgres-zfsonlinux/eb2de01/ : SUCCESS in 51m 54s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-postgres-generic-singlebackend/c412f6a/ : FAILURE in 1h 27m 29s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-generic-no-share-servers/4631acd/ : SUCCESS in 49m 03s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-scenario/bb315ad/ : FAILURE in 59m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/f30d275/ : SUCCESS in 1h 07m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/0295005/ : SUCCESS in 1h 07m 12s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/17/check/manila-tempest-dsvm-glusterfs-nfs/e8aa294/ : FAILURE in 15m 09s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-dummy/f18c417/ : SUCCESS in 36m 02s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/17/check/manila-tempest-minimal-dsvm-lvm-centos-7/e95733a/ : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":17},{"id":"513712c08ad5f3e28de5ea006aa4da5050669a8e","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2018-02-07 07:57:05.000000000","message":"Patch Set 17:\n\nThanks! Now, let\u0027s fix the https://review.openstack.org/#/c/536059/\nand combine them. \n\nI guess that  fixing 536059 (code dedup) and having this patch dependant on is the best.","accounts_in_message":[],"_revision_number":17},{"id":"f8d060d7a01b210503fbd8190f254c421b78a644","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2018-08-02 12:29:19.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"1d7ef37f4175a3b879eba5df829670dc54659c5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-02 13:57:17.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/18/check/openstack-tox-pep8/b06c8c6/ : SUCCESS in 4m 29s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-mysql-generic/2205f6f/ : FAILURE in 52m 55s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-postgres-container/15083a4/ : SUCCESS in 36m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-postgres-zfsonlinux/fc77a15/ : FAILURE in 1h 15m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-postgres-generic-singlebackend/d2f9fe3/ : FAILURE in 1h 03m 35s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-generic-no-share-servers/aa7938f/ : SUCCESS in 1h 06m 06s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-scenario/5b790da/ : FAILURE in 49m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/18/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/8eabe00/ : SUCCESS in 1h 11m 28s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/18/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/bcf09c7/ : SUCCESS in 1h 10m 36s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/18/check/manila-tempest-dsvm-glusterfs-nfs/3062718/ : FAILURE in 17m 01s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/18/check/manila-tempest-minimal-dsvm-dummy/4ce83f3/ : SUCCESS in 32m 25s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/18/check/manila-tempest-minimal-dsvm-lvm-centos-7/790a13c/ : FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":18},{"id":"85da715a0663c599c39e397a749f0d7eb1be3d71","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2018-08-02 15:35:11.000000000","message":"Patch Set 18:\n\nTatyana, Tom: I can take this over as well if it\u0027s alright with you","accounts_in_message":[],"_revision_number":18},{"id":"a2377edae691d552515af560f155888edbcaf946","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2018-08-02 16:11:12.000000000","message":"Patch Set 18:\n\nFine with me if it\u0027s OK with Tatiana.  Sorry it was neglected from review standpoint so long.","accounts_in_message":[],"_revision_number":18},{"id":"f7dcca24489deaa2d96ca5a885522575758a4296","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2018-08-06 15:27:46.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a464af50e25e0bc6ed5ee8674aaf48daa259fe3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-06 17:25:55.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/19/check/openstack-tox-pep8/13f078f/ : SUCCESS in 4m 25s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-mysql-generic/2a4aa5d/ : FAILURE in 53m 02s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-postgres-container/4807cc7/ : SUCCESS in 35m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-postgres-zfsonlinux/bc65c6a/ : FAILURE in 1h 00m 01s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-postgres-generic-singlebackend/1d10906/ : FAILURE in 59m 09s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-generic-no-share-servers/100cdda/ : FAILURE in 1h 05m 17s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-scenario/1c92edd/ : FAILURE in 50m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/19/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/0783715/ : SUCCESS in 1h 03m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/19/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/0a56133/ : SUCCESS in 1h 03m 35s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/19/check/manila-tempest-dsvm-glusterfs-nfs/e697ae9/ : FAILURE in 16m 53s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/19/check/manila-tempest-minimal-dsvm-dummy/31e931d/ : FAILURE in 39m 24s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/19/check/manila-tempest-minimal-dsvm-lvm-centos-7/1afc668/ : FAILURE in 1h 13m 35s","accounts_in_message":[],"_revision_number":19},{"id":"eab0bacc5ab194377576e0980ea36112ce92d6a2","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2018-08-08 20:20:08.000000000","message":"Uploaded patch set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"776c75fe0ec9d696942dbd4de1b33eb18da0876e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-08 22:54:07.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/20/check/openstack-tox-pep8/6739140/ : SUCCESS in 4m 32s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-mysql-generic/cd32b55/ : FAILURE in 50m 54s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-postgres-container/a765d8b/ : SUCCESS in 43m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-postgres-zfsonlinux/ba1d024/ : FAILURE in 58m 09s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-postgres-generic-singlebackend/e414d90/ : FAILURE in 54m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-generic-no-share-servers/b587ac9/ : FAILURE in 49m 39s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-scenario/e2bf9e9/ : FAILURE in 53m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/68/531568/20/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/103406a/ : SUCCESS in 1h 14m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/68/531568/20/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/b07d80c/ : SUCCESS in 1h 13m 26s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs http://logs.openstack.org/68/531568/20/check/manila-tempest-dsvm-glusterfs-nfs/e8b2df0/ : FAILURE in 20m 35s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/20/check/manila-tempest-minimal-dsvm-dummy/8a500c4/ : SUCCESS in 38m 39s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/68/531568/20/check/manila-tempest-minimal-dsvm-lvm-centos-7/e245760/ : FAILURE in 1h 13m 06s","accounts_in_message":[],"_revision_number":20},{"id":"2f29be90b1e5fc51956bf0abb2ce3c8f7a3343c8","author":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"date":"2018-09-14 16:00:31.000000000","message":"Patch Set 20: Code-Review-1\n\n(4 comments)\n\nPlease see inline comments, thanks.","accounts_in_message":[],"_revision_number":20},{"id":"a59620d5ab1538c32cf84b55462e341677f8536d","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2018-10-14 19:39:14.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"950f07fbcf470f44f67efab8c058a97a824fd367","author":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"date":"2018-10-16 19:39:59.000000000","message":"Patch Set 20:\n\n(1 comment)\n\nHey Nir, thanks for reviewing my previous comments. Please see my replies, thanks!","accounts_in_message":[],"_revision_number":20},{"id":"72aa94f485b1c9c7f274037f19d8d67c4f05e861","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-03-05 07:26:37.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"4e551793c7c6539e27860b48ce889cbaade16d5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 09:06:25.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/21/check/openstack-tox-pep8/7be492b/ : SUCCESS in 4m 34s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/21/check/manila-tempest-dsvm-mysql-generic/2ba1150/ : FAILURE in 1h 36m 28s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/21/check/manila-tempest-dsvm-postgres-container/0d13a25/ : SUCCESS in 46m 44s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/21/check/manila-tempest-dsvm-postgres-zfsonlinux/dac64b5/ : FAILURE in 52m 56s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/21/check/manila-tempest-dsvm-postgres-generic-singlebackend/eda88eb/ : SUCCESS in 54m 55s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/21/check/manila-tempest-dsvm-generic-no-share-servers/24d9ff3/ : FAILURE in 1h 12m 41s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/21/check/manila-tempest-dsvm-scenario/d9ad8b5/ : FAILURE in 1h 24m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/21/check/manila-tempest-minimal-dsvm-cephfs-native/7abb169/ : SUCCESS in 22m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/21/check/manila-tempest-minimal-dsvm-cephfs-nfs/f3c5ae6/ : SUCCESS in 29m 50s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/21/check/manila-tempest-minimal-dsvm-dummy/5ee78d1/ : SUCCESS in 36m 29s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/21/check/manila-tempest-minimal-dsvm-lvm/cbec500/ : FAILURE in 58m 15s","accounts_in_message":[],"_revision_number":21},{"id":"722cd8373cab7cfed11668c738ba42cbde4f481c","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-03-06 09:36:36.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"79c8a592f4564e7517aa3c7e5b78a9104963b407","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 11:08:07.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/22/check/openstack-tox-pep8/24dce66/ : SUCCESS in 6m 07s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/22/check/manila-tempest-dsvm-mysql-generic/88f00cd/ : FAILURE in 1h 24m 06s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/22/check/manila-tempest-dsvm-postgres-container/d6e9354/ : SUCCESS in 47m 38s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/22/check/manila-tempest-dsvm-postgres-zfsonlinux/58b0de6/ : SUCCESS in 55m 10s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/22/check/manila-tempest-dsvm-postgres-generic-singlebackend/305f1ac/ : SUCCESS in 1h 06m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/22/check/manila-tempest-dsvm-generic-no-share-servers/7d1b53a/ : FAILURE in 54m 47s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/22/check/manila-tempest-dsvm-scenario/0701c36/ : FAILURE in 1h 22m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/22/check/manila-tempest-minimal-dsvm-cephfs-native/1b02624/ : SUCCESS in 22m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/22/check/manila-tempest-minimal-dsvm-cephfs-nfs/80af532/ : SUCCESS in 22m 07s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/22/check/manila-tempest-minimal-dsvm-dummy/f703ced/ : SUCCESS in 44m 48s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/22/check/manila-tempest-minimal-dsvm-lvm/786328d/ : SUCCESS in 1h 01m 37s","accounts_in_message":[],"_revision_number":22},{"id":"18842584eac2cba69639ae4f8de9397d7aa7bd27","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-03-06 13:30:32.000000000","message":"Patch Set 22:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/22/upstream-check/manila-cDOT-manila-tempest-plugin-ss/675be57/ : SUCCESS in 1h 17m 20s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/22/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ba5da7f/ : POST_FAILURE in 17m 17s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":22},{"id":"9ba20b48e5b191d49bba535148b64907c2a7302e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-03-06 20:52:38.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)\n\nThanks for the pursuit/effort on this test Nir... please see some comments inline for enhancement","accounts_in_message":[],"_revision_number":22},{"id":"d0077a5220768b5b24f0366b95c643a036fd7822","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-03-06 20:55:51.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"b79dc4ea03d1f129740406e996c069f3d17b73b8","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-03-16 13:44:34.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"616e207b4d1e1b830749f70dbd6f9bc2c4d193fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-16 15:27:27.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/23/check/openstack-tox-pep8/3292896/ : SUCCESS in 5m 56s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/23/check/manila-tempest-dsvm-mysql-generic/cc4dda9/ : SUCCESS in 1h 25m 31s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/23/check/manila-tempest-dsvm-postgres-container/7084223/ : SUCCESS in 47m 15s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/23/check/manila-tempest-dsvm-postgres-zfsonlinux/f878dab/ : SUCCESS in 1h 14m 14s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/23/check/manila-tempest-dsvm-postgres-generic-singlebackend/c008541/ : SUCCESS in 1h 04m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/23/check/manila-tempest-dsvm-generic-no-share-servers/909c1bf/ : SUCCESS in 1h 15m 07s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/23/check/manila-tempest-dsvm-scenario/b994ef2/ : FAILURE in 1h 39m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/23/check/manila-tempest-minimal-dsvm-cephfs-native/86bca83/ : SUCCESS in 24m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/23/check/manila-tempest-minimal-dsvm-cephfs-nfs/37abe50/ : FAILURE in 32m 43s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/23/check/manila-tempest-minimal-dsvm-dummy/13e5cca/ : SUCCESS in 40m 47s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/23/check/manila-tempest-minimal-dsvm-lvm/d1e45d5/ : FAILURE in 57m 43s","accounts_in_message":[],"_revision_number":23},{"id":"66e5a507091fda62b1f0453eb49575bca92d6d3e","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-04-08 08:40:45.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"708a4cc357a0f2eaa5ccfb35159673539d6eb3a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 10:07:05.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/24/check/openstack-tox-pep8/5d1a29d/ : SUCCESS in 5m 00s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/24/check/manila-tempest-dsvm-mysql-generic/823c846/ : FAILURE in 1h 23m 54s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/24/check/manila-tempest-dsvm-postgres-container/fb05131/ : SUCCESS in 45m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/24/check/manila-tempest-dsvm-postgres-zfsonlinux/7c3f838/ : FAILURE in 58m 57s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/24/check/manila-tempest-dsvm-postgres-generic-singlebackend/67ef131/ : FAILURE in 50m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/24/check/manila-tempest-dsvm-generic-no-share-servers/20f30b4/ : SUCCESS in 57m 08s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/24/check/manila-tempest-dsvm-scenario/cad8375/ : FAILURE in 1h 22m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/24/check/manila-tempest-minimal-dsvm-cephfs-native/19aa00e/ : SUCCESS in 21m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/24/check/manila-tempest-minimal-dsvm-cephfs-nfs/e33df02/ : SUCCESS in 22m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/24/check/manila-tempest-minimal-dsvm-dummy/517f933/ : SUCCESS in 37m 29s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/24/check/manila-tempest-minimal-dsvm-lvm/495fa78/ : FAILURE in 58m 33s","accounts_in_message":[],"_revision_number":24},{"id":"188c246847491bece395465e897031c469737824","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-04-08 19:38:34.000000000","message":"Patch Set 24:\n\n(2 comments)\n\nCan you rebase this test on top of share shrink test: https://review.openstack.org/#/c/531615/\nThere are common changes that the shrink test patch already covers.","accounts_in_message":[],"_revision_number":24},{"id":"dd0a59acef04632b4193ea198c1286d0abc61dfa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-04-08 19:56:35.000000000","message":"Patch Set 24:\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"99cd5a87fbc3b090c47810c2806e5cfab1b98c74","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-04-20 19:15:18.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"618a12a8a8234d0ef45ce0d64a86b17c3be87591","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 20:32:43.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/25/check/openstack-tox-pep8/fd4cdb1/ : SUCCESS in 4m 09s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/25/check/manila-tempest-dsvm-mysql-generic/d28909a/ : SUCCESS in 1h 16m 09s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/25/check/manila-tempest-dsvm-postgres-container/7b7b134/ : SUCCESS in 41m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/25/check/manila-tempest-dsvm-postgres-zfsonlinux/33b6c29/ : SUCCESS in 57m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/25/check/manila-tempest-dsvm-postgres-generic-singlebackend/b4cb596/ : SUCCESS in 43m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/25/check/manila-tempest-dsvm-generic-no-share-servers/60605de/ : SUCCESS in 1h 05m 23s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/25/check/manila-tempest-dsvm-scenario/a6ab854/ : FAILURE in 1h 14m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/25/check/manila-tempest-minimal-dsvm-cephfs-native/ceb1621/ : SUCCESS in 22m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/25/check/manila-tempest-minimal-dsvm-cephfs-nfs/ff1ae50/ : FAILURE in 34m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/25/check/manila-tempest-minimal-dsvm-dummy/3e26165/ : SUCCESS in 36m 57s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/25/check/manila-tempest-minimal-dsvm-lvm/dd41a11/ : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":25},{"id":"d5bd15b3d710600b80deae4da62dc62ba900c8d2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-20 20:33:38.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/25/upstream-check/manila-cDOT-manila-tempest-plugin-ss/84ad577/ : FAILURE in 36s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/25/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2f0975b/ : FAILURE in 36s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"ff9cce8aa00a720509104922dc1f27585cd5776e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-04-22 17:29:50.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"4e80ee694b4777c2a5fa22eb8813498f0e8ecf82","author":{"_account_id":29631,"name":"Arthur Nascimento Santos","email":"arthurs@netapp.com","username":"throne82"},"date":"2019-04-25 18:48:24.000000000","message":"Patch Set 25:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":25},{"id":"116cc961f540e23edc9ca89c4ed1086f6d7e953a","author":{"_account_id":29288,"name":"Guilherme J.  Arantes","email":"guilherme.jannotti@gmail.com","username":"guilherme.arantes"},"date":"2019-04-25 20:10:50.000000000","message":"Patch Set 25:\n\nrun-netapp-ci","accounts_in_message":[],"_revision_number":25},{"id":"061483bc717f31e4163d1d8092000c922a6bd724","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-04-26 02:26:28.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/25/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b309b47/ : FAILURE in 3h 48m 07s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/25/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0eea684/ : FAILURE in 3h 48m 05s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"d86636031e65871fc3b6de3215e16cf12c5114a7","author":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"date":"2019-04-29 06:54:41.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"9dda6412746834400db993ef1ec71216195acfc6","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-06 14:16:14.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"ca24cbdaf4a499758c5060fadc6895d7d1a02f49","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-06 14:20:27.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"b4ad1aa83b2f16cbf483a6d36f17fcd2c1e5f6ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 15:50:08.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/26/check/openstack-tox-pep8/840c314/ : SUCCESS in 4m 34s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/26/check/manila-tempest-dsvm-mysql-generic/1053fc8/ : FAILURE in 1h 18m 45s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/26/check/manila-tempest-dsvm-postgres-container/b066977/ : SUCCESS in 48m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/26/check/manila-tempest-dsvm-postgres-zfsonlinux/57235ed/ : SUCCESS in 1h 05m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/26/check/manila-tempest-dsvm-postgres-generic-singlebackend/8d1b5a3/ : SUCCESS in 1h 01m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/26/check/manila-tempest-dsvm-generic-no-share-servers/791ba5a/ : SUCCESS in 59m 57s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/26/check/manila-tempest-dsvm-scenario/1ea651d/ : FAILURE in 1h 26m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/26/check/manila-tempest-minimal-dsvm-cephfs-native/289cb76/ : SUCCESS in 20m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/26/check/manila-tempest-minimal-dsvm-cephfs-nfs/6d71b07/ : FAILURE in 43m 59s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/26/check/manila-tempest-minimal-dsvm-dummy/0a97043/ : SUCCESS in 40m 34s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/26/check/manila-tempest-minimal-dsvm-lvm/130046f/ : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":26},{"id":"5697e9d40c5e3081c5fc933274073d6936a37886","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-06 19:38:04.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/26/upstream-check/manila-cDOT-manila-tempest-plugin-ss/9c27d25/ : FAILURE in 3h 47m 35s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/26/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/23f87e0/ : FAILURE in 2h 46m 28s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"e2c4f772b7e69d0226ddd9de2f34ab3bc1588ebc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-05-07 20:29:02.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"060bef79a29e9fcae03b187f3a608197893db34e","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-16 14:42:35.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"50dcc1513edda6ad7f9d81236b8f0c54c18918fc","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-16 14:44:17.000000000","message":"Patch Set 27:\n\n\u003e (1 comment)\n\nThank you!\nI committed a patchset to resolve this issue by remounting as you suggested.","accounts_in_message":[],"_revision_number":27},{"id":"d55590a23986e3d38ad2b1fbb35c1ef8ab0eb7f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 18:27:23.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/27/check/openstack-tox-pep8/f6be57f/ : SUCCESS in 7m 12s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/27/check/manila-tempest-dsvm-mysql-generic/8b49017/ : FAILURE in 1h 17m 53s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/27/check/manila-tempest-dsvm-postgres-container/7c588dd/ : SUCCESS in 58m 53s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/27/check/manila-tempest-dsvm-postgres-zfsonlinux/ba9aafe/ : FAILURE in 1h 15m 57s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/27/check/manila-tempest-dsvm-postgres-generic-singlebackend/4ace32d/ : SUCCESS in 58m 59s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/27/check/manila-tempest-dsvm-generic-no-share-servers/c55c2cf/ : FAILURE in 1h 11m 58s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/27/check/manila-tempest-dsvm-scenario/7275ad8/ : FAILURE in 1h 47m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/27/check/manila-tempest-minimal-dsvm-cephfs-native/76d1118/ : SUCCESS in 28m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/27/check/manila-tempest-minimal-dsvm-cephfs-nfs/3e97ef5/ : SUCCESS in 28m 09s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/27/check/manila-tempest-minimal-dsvm-dummy/ea6c071/ : SUCCESS in 40m 53s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/27/check/manila-tempest-minimal-dsvm-lvm/201cf0b/ : FAILURE in 1h 05m 27s","accounts_in_message":[],"_revision_number":27},{"id":"5bb204d09d42b8f5dd8326757ab92cecb2d145b4","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-16 19:51:11.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"a078846ade1aa7f215f82ba78d571f8e0b6b29d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:59:51.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/28/check/openstack-tox-pep8/45232e6/ : SUCCESS in 4m 20s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-mysql-generic/e3c009a/ : FAILURE in 1h 18m 41s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-postgres-container/04c7284/ : SUCCESS in 57m 07s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-postgres-zfsonlinux/b6e0ea2/ : FAILURE in 1h 52m 09s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-postgres-generic-singlebackend/d930dfa/ : SUCCESS in 1h 03m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-generic-no-share-servers/0f7fb01/ : SUCCESS in 56m 25s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-scenario/3f943ec/ : FAILURE in 1h 46m 08s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-cephfs-native/5e226cd/ : SUCCESS in 18m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-cephfs-nfs/f180532/ : SUCCESS in 26m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-dummy/f67652a/ : FAILURE in 41m 06s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-lvm/290459b/ : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":28},{"id":"b35d7292d474448847700fd239b2f4ffb80fca87","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-17 11:34:16.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"323ca4ae2fcb4803962807425e79d46063ab721c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 13:06:45.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/28/check/openstack-tox-pep8/d58acd6/ : SUCCESS in 3m 42s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-mysql-generic/923938f/ : SUCCESS in 1h 25m 00s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-postgres-container/da706df/ : SUCCESS in 50m 09s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-postgres-zfsonlinux/b6cd169/ : SUCCESS in 54m 34s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-postgres-generic-singlebackend/0b1d3de/ : SUCCESS in 53m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-generic-no-share-servers/a49b67b/ : FAILURE in 59m 43s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/28/check/manila-tempest-dsvm-scenario/5d4d28c/ : FAILURE in 1h 31m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-cephfs-native/dfc36a7/ : SUCCESS in 24m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-cephfs-nfs/ecf92c8/ : SUCCESS in 21m 31s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-dummy/0336bd4/ : SUCCESS in 42m 45s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/28/check/manila-tempest-minimal-dsvm-lvm/6c2f2fe/ : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":28},{"id":"0e725e30dbd11fd91a9a118715897b246853e20e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-18 00:36:31.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/28/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3220fd9/ : FAILURE in 3h 47m 45s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/28/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b148037/ : SUCCESS in 1h 28m 40s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":28},{"id":"ccab2d28e93b5d2ac94ef2af03c8c956e33f8a80","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-05-18 00:43:39.000000000","message":"Patch Set 28:\n\n(5 comments)","accounts_in_message":[],"_revision_number":28},{"id":"85beff76a7f49570f32b0bf051e7c7cb8564e4af","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-05-19 07:13:31.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"0f4603c7bde4904bc9e17612d4a1ed8934e06a0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 08:57:00.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/29/check/openstack-tox-pep8/a25bd62/ : SUCCESS in 4m 14s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-mysql-generic/e7163e2/ : SUCCESS in 1h 25m 42s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-container/df7212b/ : SUCCESS in 45m 15s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-zfsonlinux/f027beb/ : FAILURE in 58m 26s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-generic-singlebackend/80031d3/ : SUCCESS in 43m 27s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-generic-no-share-servers/a42dd7f/ : FAILURE in 21m 02s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-scenario/67be06c/ : FAILURE in 1h 42m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-native/3dfd354/ : SUCCESS in 20m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-nfs/6f004f2/ : SUCCESS in 24m 49s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-dummy/3e62423/ : SUCCESS in 34m 10s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-lvm/8a35299/ : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":29},{"id":"2a323277d6acf6f68119b02dd5589650067ba793","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-05-20 21:32:02.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-ss/dd73f80/ : FAILURE in 3h 47m 46s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5dbbcbe/ : SUCCESS in 1h 25m 45s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"b07bf6233d415b77583c72cabff21066aefd49f8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-05-21 22:48:33.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"bc0284e0909b98d585d2c798da7f39c61b8600b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 00:24:13.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/29/check/openstack-tox-pep8/c4fabfd/ : SUCCESS in 8m 12s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-mysql-generic/62a873b/ : FAILURE in 1h 19m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-container/2e0cfe5/ : SUCCESS in 49m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-zfsonlinux/552943e/ : SUCCESS in 1h 14m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-generic-singlebackend/dca928b/ : SUCCESS in 1h 05m 22s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-generic-no-share-servers/ac98458/ : FAILURE in 53m 47s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-scenario/20d7a52/ : FAILURE in 1h 28m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-native/645b4d2/ : SUCCESS in 23m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-nfs/82761fd/ : SUCCESS in 20m 16s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-dummy/15b5a47/ : SUCCESS in 38m 45s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-lvm/447c348/ : FAILURE in 25m 34s","accounts_in_message":[],"_revision_number":29},{"id":"d369a5079f0bce667aa6c437cab6ff79988816aa","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-06-01 12:24:16.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"2f887a9f96f0ca2df7e32c8d98f07ee0334deeb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 14:23:18.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/29/check/openstack-tox-pep8/22a39a6/ : SUCCESS in 8m 20s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-mysql-generic/1a10cbf/ : FAILURE in 1h 28m 49s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-container/43e7896/ : SUCCESS in 1h 06m 49s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-zfsonlinux/87d6526/ : SUCCESS in 1h 34m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-generic-singlebackend/91f23a5/ : SUCCESS in 1h 23m 18s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-generic-no-share-servers/b43d7f1/ : FAILURE in 1h 41m 14s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-scenario/2105795/ : FAILURE in 1h 54m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-native/844d976/ : SUCCESS in 37m 28s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-nfs/d35695d/ : SUCCESS in 23m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-dummy/e37d8ee/ : SUCCESS in 52m 16s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-lvm/0af8f57/ : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":29},{"id":"13fe962cd489ae2219c3c3fdc42e7a994691a1cc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-01 18:11:22.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-ss/068ebd8/ : FAILURE in 3h 47m 47s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/406e51c/ : SUCCESS in 1h 23m 20s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"fb227b785fd92f7084515d667f31470d399d23fe","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-06-18 13:49:43.000000000","message":"Patch Set 29:\n\nIs there anything additional I should do on this in order to merge?","accounts_in_message":[],"_revision_number":29},{"id":"c2ba518f6fefa30a76a9259f36b8f9c629d008d8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-18 20:19:41.000000000","message":"Patch Set 29:\n\nrecheck\n\nWe had a timeout exception on the last run. Code LGTM at the moment, let\u0027s wait to see the test pass..","accounts_in_message":[],"_revision_number":29},{"id":"908df0aadfc2d0bba8b87566353d869aaba4f0f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 22:24:03.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/29/check/openstack-tox-pep8/ad50ea2/ : SUCCESS in 4m 40s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-mysql-generic/71ccc58/ : SUCCESS in 1h 48m 33s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-container/fa05102/ : SUCCESS in 50m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-zfsonlinux/bf01bc9/ : SUCCESS in 1h 09m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-generic-singlebackend/282ec77/ : SUCCESS in 57m 40s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-generic-no-share-servers/e9fce07/ : SUCCESS in 1h 09m 52s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-scenario/83cefb8/ : FAILURE in 1h 30m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-native/fe6f508/ : SUCCESS in 23m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-nfs/efa45e7/ : SUCCESS in 22m 52s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-dummy/befb417/ : SUCCESS in 40m 08s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-lvm/adc0c45/ : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":29},{"id":"125741569b33de8ed336bb97e6eae2d14655854b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-18 23:34:20.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-ss/92a1690/ : SUCCESS in 1h 09m 59s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8fb19a6/ : NOT_REGISTERED\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"379663f22431fde4e92729a8bca88dcb621dec79","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-20 00:02:28.000000000","message":"Patch Set 29: Code-Review+2\n\nrecheck\n\nThis test looks good to me. We have, however a failure with Timeout for the other test we added recently (TestShareShrinkNFS.test_create_shrink_and_write). Let\u0027s see if that\u0027s an issue we need to address, \n\nThank you for your continued patience and efforts with this Nir! Hope to see you contribute more!","accounts_in_message":[],"_revision_number":29},{"id":"52b5ab3891370a297b0eb55bd1951902194d8d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 01:44:00.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/29/check/openstack-tox-pep8/e99d631/ : SUCCESS in 6m 48s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-mysql-generic/25cc436/ : SUCCESS in 1h 27m 29s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-container/4cd428c/ : SUCCESS in 50m 43s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-zfsonlinux/4c3857c/ : SUCCESS in 1h 08m 45s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-postgres-generic-singlebackend/133eb9b/ : SUCCESS in 59m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-generic-no-share-servers/066a1bf/ : FAILURE in 1h 09m 22s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/68/531568/29/check/manila-tempest-dsvm-scenario/96b99ba/ : SUCCESS in 1h 28m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-native/2d73791/ : SUCCESS in 28m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-cephfs-nfs/acab6b3/ : FAILURE in 1h 22m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-dummy/cb54272/ : SUCCESS in 45m 35s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/29/check/manila-tempest-minimal-dsvm-lvm/859109d/ : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":29},{"id":"10766d75a2b03a8165d65154d9fb4f02836cc244","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-20 03:10:01.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-ss/984dcf7/ : SUCCESS in 1h 11m 08s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/68/531568/29/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a65087b/ : SUCCESS in 1h 25m 43s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"a1b6a7f108e3170b6b802ef75dcbaa8585f6f860","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-20 10:24:21.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1\n\nThanks for the great work, Nir!","accounts_in_message":[],"_revision_number":29},{"id":"82e92ed625b46f971097407606f91fdae2a874bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 10:24:33.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"dac2d377630656f672c99dbc966afcc917ac5ef3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:28:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":29},{"id":"ac9e6c623dd869f0b8f4734b562ddf33ec41b45e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:28:56.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/68/531568/29/gate/openstack-tox-pep8/47221ab/ : SUCCESS in 4m 13s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/68/531568/29/gate/manila-tempest-minimal-dsvm-dummy/d757fd7/ : SUCCESS in 43m 57s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/68/531568/29/gate/manila-tempest-minimal-dsvm-lvm/a74d045/ : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":29},{"id":"66e5638fc125f981d9ed6e48f5da0fcbf1f2664e","author":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"date":"2019-06-22 13:26:05.000000000","message":"Patch Set 29:\n\nI should be the one to thank you, Goutham :)\nThank you for your help on this - I will certainly take a look at resolving the timeout issue on the shrink test.","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"2cb85a6edd76ae94b428cc504c9e25f419618c9a","revisions":{"f89543cc746a37314d771c76c82392d5eaf734c1":{"kind":"REWORK","_number":1,"created":"2018-01-06 18:49:26.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/1"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:24.000000000","tz":-480},"subject":"[do not merge] scenario test create/srink","message":"[do not merge] scenario test create/srink\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f89543cc746a37314d771c76c82392d5eaf734c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f89543cc746a37314d771c76c82392d5eaf734c1"}]},"branch":"refs/heads/master"},"de5236f31f2b8414df79c065c2c3e4bc0f93d9da":{"kind":"REWORK","_number":2,"created":"2018-01-06 20:46:22.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/2"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 20:46:13.000000000","tz":-480},"subject":"[do not merge] scenario test create/srink","message":"[do not merge] scenario test create/srink\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/de5236f31f2b8414df79c065c2c3e4bc0f93d9da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/de5236f31f2b8414df79c065c2c3e4bc0f93d9da"}]},"branch":"refs/heads/master"},"74d66ca61c7d7d4a60b0333362d2c7e5bf54de6d":{"kind":"REWORK","_number":3,"created":"2018-01-06 22:25:10.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/3"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 22:25:01.000000000","tz":-480},"subject":"[do not merge] scenario test create/srink","message":"[do not merge] scenario test create/srink\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74d66ca61c7d7d4a60b0333362d2c7e5bf54de6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74d66ca61c7d7d4a60b0333362d2c7e5bf54de6d"}]},"branch":"refs/heads/master"},"e3a185dea265795f5074a9f9f3e9390c79376f94":{"kind":"REWORK","_number":4,"created":"2018-01-07 08:47:33.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/4"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-07 08:47:24.000000000","tz":-480},"subject":"[do not merge] scenario test create/srink","message":"[do not merge] scenario test create/srink\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3a185dea265795f5074a9f9f3e9390c79376f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3a185dea265795f5074a9f9f3e9390c79376f94"}]},"branch":"refs/heads/master"},"db1fbe574ba3c0371ed2d89382d90639a228d49f":{"kind":"REWORK","_number":5,"created":"2018-01-07 12:47:38.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/5"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-07 12:47:29.000000000","tz":-480},"subject":"[do not merge] scenario test create/srink","message":"[do not merge] scenario test create/srink\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db1fbe574ba3c0371ed2d89382d90639a228d49f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db1fbe574ba3c0371ed2d89382d90639a228d49f"}]},"branch":"refs/heads/master"},"c0373d14a0a72aa216d23332e1a4d80553868360":{"kind":"REWORK","_number":6,"created":"2018-01-07 18:49:24.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/6"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-07 18:48:58.000000000","tz":-480},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0373d14a0a72aa216d23332e1a4d80553868360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0373d14a0a72aa216d23332e1a4d80553868360"}]},"branch":"refs/heads/master"},"64ac47b316f8e866952a88a3da7d07ec921ea599":{"kind":"REWORK","_number":7,"created":"2018-01-08 07:20:21.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/7"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-08 07:20:12.000000000","tz":-480},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64ac47b316f8e866952a88a3da7d07ec921ea599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64ac47b316f8e866952a88a3da7d07ec921ea599"}]},"branch":"refs/heads/master"},"0a3694d288ac7bc36f9aab07d14e42f548d9e289":{"kind":"REWORK","_number":8,"created":"2018-01-09 08:45:26.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/8"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-09 08:45:18.000000000","tz":-480},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a3694d288ac7bc36f9aab07d14e42f548d9e289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a3694d288ac7bc36f9aab07d14e42f548d9e289"}]},"branch":"refs/heads/master"},"4c8d435a3aef7a0dcb1717cd1738c35e4377d4d1":{"kind":"REWORK","_number":9,"created":"2018-01-09 18:24:30.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/9"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-09 18:24:22.000000000","tz":-480},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c8d435a3aef7a0dcb1717cd1738c35e4377d4d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c8d435a3aef7a0dcb1717cd1738c35e4377d4d1"}]},"branch":"refs/heads/master"},"c6b18600272aab296d4d6d8dc6b83204b7ceb2a3":{"kind":"REWORK","_number":10,"created":"2018-01-10 14:30:26.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/10"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-10 14:29:49.000000000","tz":-480},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6b18600272aab296d4d6d8dc6b83204b7ceb2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c6b18600272aab296d4d6d8dc6b83204b7ceb2a3"}]},"branch":"refs/heads/master"},"ead249ede9b4db2ae371efc3ed92962b5468cc77":{"kind":"NO_CHANGE","_number":11,"created":"2018-01-30 21:07:42.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/11"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-30 21:06:56.000000000","tz":120},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ead249ede9b4db2ae371efc3ed92962b5468cc77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ead249ede9b4db2ae371efc3ed92962b5468cc77"}]},"branch":"refs/heads/master"},"0ba2048bb194f58dbb88ef8cde50165c5d805bc4":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2018-01-30 21:10:07.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/12"}}},"commit":{"parents":[{"commit":"797257e3acdfd7875d7e3d069f568e4c2be11393","subject":"Add IPv6 support to scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/797257e3acdfd7875d7e3d069f568e4c2be11393"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-30 21:09:20.000000000","tz":120},"subject":"[do not merge] scenario test create/extend","message":"[do not merge] scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ba2048bb194f58dbb88ef8cde50165c5d805bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ba2048bb194f58dbb88ef8cde50165c5d805bc4"}]},"branch":"refs/heads/master"},"663ba06624c85654f9977a25644edf21be7bf042":{"kind":"REWORK","_number":13,"created":"2018-02-03 18:28:48.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/13"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-02-03 18:28:24.000000000","tz":-480},"subject":"Scenario test create/extend","message":"Scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/663ba06624c85654f9977a25644edf21be7bf042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/663ba06624c85654f9977a25644edf21be7bf042"}]},"branch":"refs/heads/master"},"6848ce69c7b4595dea022cfa453b6822b80d6aee":{"kind":"REWORK","_number":14,"created":"2018-02-03 18:30:28.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/14"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-02-03 18:30:15.000000000","tz":-480},"subject":"Scenario test create/extend","message":"Scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6848ce69c7b4595dea022cfa453b6822b80d6aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6848ce69c7b4595dea022cfa453b6822b80d6aee"}]},"branch":"refs/heads/master"},"c59d980f78a6530b65cd828c1b24885103e4e57d":{"kind":"REWORK","_number":15,"created":"2018-02-03 20:43:08.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/15"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-02-03 20:42:59.000000000","tz":-480},"subject":"Scenario test create/extend","message":"Scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c59d980f78a6530b65cd828c1b24885103e4e57d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c59d980f78a6530b65cd828c1b24885103e4e57d"}]},"branch":"refs/heads/master"},"4746be51e4246d4de472e73bd39d2a7dd4d87b70":{"kind":"REWORK","_number":16,"created":"2018-02-04 17:00:07.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/16"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-02-04 16:59:59.000000000","tz":-480},"subject":"Scenario test create/extend","message":"Scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4746be51e4246d4de472e73bd39d2a7dd4d87b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4746be51e4246d4de472e73bd39d2a7dd4d87b70"}]},"branch":"refs/heads/master"},"a7f9c69618dd0c241e98c719fb7800a893c9526a":{"kind":"REWORK","_number":17,"created":"2018-02-04 20:38:45.000000000","uploader":{"_account_id":26503,"name":"Tatyana","email":"tdyshka@gmail.com","username":"tdyshka"},"ref":"refs/changes/68/531568/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/17"}}},"commit":{"parents":[{"commit":"be1134c700f6e37b8611bd67c1c2a6efb26c39d9","subject":"Remove the max api microversion assertion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/be1134c700f6e37b8611bd67c1c2a6efb26c39d9"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-02-04 20:38:35.000000000","tz":-480},"subject":"Scenario test create/extend","message":"Scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a7f9c69618dd0c241e98c719fb7800a893c9526a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a7f9c69618dd0c241e98c719fb7800a893c9526a"}]},"branch":"refs/heads/master"},"e35288880e3f4341b789a579c6db3e513580f808":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2018-08-02 12:29:19.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/68/531568/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/18"}}},"commit":{"parents":[{"commit":"2e153ac29fd64ad70244fd68b4f20de33c46c19e","subject":"Merge \"s/share_net/share_network in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e153ac29fd64ad70244fd68b4f20de33c46c19e"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2018-08-02 12:29:19.000000000","tz":0},"subject":"Scenario test create/extend","message":"Scenario test create/extend\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e35288880e3f4341b789a579c6db3e513580f808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e35288880e3f4341b789a579c6db3e513580f808"}]},"branch":"refs/heads/master"},"e97b32c8533c56f24ad629af68a9c9922444fa6a":{"kind":"REWORK","_number":19,"created":"2018-08-06 15:27:46.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/19"}}},"commit":{"parents":[{"commit":"2e153ac29fd64ad70244fd68b4f20de33c46c19e","subject":"Merge \"s/share_net/share_network in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e153ac29fd64ad70244fd68b4f20de33c46c19e"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-08-06 15:27:10.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e97b32c8533c56f24ad629af68a9c9922444fa6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e97b32c8533c56f24ad629af68a9c9922444fa6a"}]},"branch":"refs/heads/master"},"5c100f61a4316e1abaf65915914d9f9622101932":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2018-08-08 20:20:08.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/20"}}},"commit":{"parents":[{"commit":"2e153ac29fd64ad70244fd68b4f20de33c46c19e","subject":"Merge \"s/share_net/share_network in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e153ac29fd64ad70244fd68b4f20de33c46c19e"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-08-08 20:19:45.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c100f61a4316e1abaf65915914d9f9622101932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5c100f61a4316e1abaf65915914d9f9622101932"}]},"branch":"refs/heads/master"},"818bda6aea747957363a19a6d1e36b26be022e1c":{"kind":"REWORK","_number":21,"created":"2019-03-05 07:26:37.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/21"}}},"commit":{"parents":[{"commit":"13a5bc1060f4731114d5a9d572f4e2c4fe12a419","subject":"Merge \"Add tests for share type availability_zones extra-spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/13a5bc1060f4731114d5a9d572f4e2c4fe12a419"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-03-05 07:25:48.000000000","tz":120},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/818bda6aea747957363a19a6d1e36b26be022e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/818bda6aea747957363a19a6d1e36b26be022e1c"}]},"branch":"refs/heads/master"},"867e94afc025cd28a6d3a31af7ca24d423e0e2ae":{"kind":"REWORK","_number":22,"created":"2019-03-06 09:36:36.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/22"}}},"commit":{"parents":[{"commit":"13a5bc1060f4731114d5a9d572f4e2c4fe12a419","subject":"Merge \"Add tests for share type availability_zones extra-spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/13a5bc1060f4731114d5a9d572f4e2c4fe12a419"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-03-06 09:35:55.000000000","tz":120},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/867e94afc025cd28a6d3a31af7ca24d423e0e2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/867e94afc025cd28a6d3a31af7ca24d423e0e2ae"}]},"branch":"refs/heads/master"},"4bb4fea27e849109438a86e539d9209ddf029266":{"kind":"REWORK","_number":23,"created":"2019-03-16 13:44:34.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/23"}}},"commit":{"parents":[{"commit":"13a5bc1060f4731114d5a9d572f4e2c4fe12a419","subject":"Merge \"Add tests for share type availability_zones extra-spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/13a5bc1060f4731114d5a9d572f4e2c4fe12a419"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-03-16 13:44:23.000000000","tz":120},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4bb4fea27e849109438a86e539d9209ddf029266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4bb4fea27e849109438a86e539d9209ddf029266"}]},"branch":"refs/heads/master"},"844fc0129834d1b05c89830c24271bcaa908090b":{"kind":"REWORK","_number":24,"created":"2019-04-08 08:40:45.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/24"}}},"commit":{"parents":[{"commit":"13a5bc1060f4731114d5a9d572f4e2c4fe12a419","subject":"Merge \"Add tests for share type availability_zones extra-spec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/13a5bc1060f4731114d5a9d572f4e2c4fe12a419"}]}],"author":{"name":"Tatyana","email":"tdyshka@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-04-08 08:38:17.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/844fc0129834d1b05c89830c24271bcaa908090b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/844fc0129834d1b05c89830c24271bcaa908090b"}]},"branch":"refs/heads/master"},"813e6ba87bc5305e6a1087c4d3d3af3fac909859":{"kind":"REWORK","_number":25,"created":"2019-04-20 19:15:18.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/25"}}},"commit":{"parents":[{"commit":"c07415d725c5f04d46e06cb72ed441250bad73a0","subject":"Scenario test: Create/shrink share and write data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c07415d725c5f04d46e06cb72ed441250bad73a0"}]}],"author":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-04-20 19:14:20.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/813e6ba87bc5305e6a1087c4d3d3af3fac909859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/813e6ba87bc5305e6a1087c4d3d3af3fac909859"}]},"branch":"refs/heads/master"},"14e07cc5f9d3bc816f8bdd0b54cdfaa6ea9404f9":{"kind":"REWORK","_number":26,"created":"2019-05-06 14:16:14.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/26"}}},"commit":{"parents":[{"commit":"c07415d725c5f04d46e06cb72ed441250bad73a0","subject":"Scenario test: Create/shrink share and write data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c07415d725c5f04d46e06cb72ed441250bad73a0"}]}],"author":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-05-06 14:12:55.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14e07cc5f9d3bc816f8bdd0b54cdfaa6ea9404f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14e07cc5f9d3bc816f8bdd0b54cdfaa6ea9404f9"}]},"branch":"refs/heads/master"},"39e59e4e09302be767d335831670eec3cf542dbb":{"kind":"REWORK","_number":27,"created":"2019-05-16 14:42:35.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/27"}}},"commit":{"parents":[{"commit":"c07415d725c5f04d46e06cb72ed441250bad73a0","subject":"Scenario test: Create/shrink share and write data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c07415d725c5f04d46e06cb72ed441250bad73a0"}]}],"author":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-05-16 14:42:18.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39e59e4e09302be767d335831670eec3cf542dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39e59e4e09302be767d335831670eec3cf542dbb"}]},"branch":"refs/heads/master"},"c86f7ffdebdb27ac680cb71c2bf9d27318fada30":{"kind":"REWORK","_number":28,"created":"2019-05-16 19:51:11.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/28"}}},"commit":{"parents":[{"commit":"c07415d725c5f04d46e06cb72ed441250bad73a0","subject":"Scenario test: Create/shrink share and write data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c07415d725c5f04d46e06cb72ed441250bad73a0"}]}],"author":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-05-16 19:47:22.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c86f7ffdebdb27ac680cb71c2bf9d27318fada30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c86f7ffdebdb27ac680cb71c2bf9d27318fada30"}]},"branch":"refs/heads/master"},"2cb85a6edd76ae94b428cc504c9e25f419618c9a":{"kind":"REWORK","_number":29,"created":"2019-05-19 07:13:31.000000000","uploader":{"_account_id":26535,"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","username":"nirgilbo"},"ref":"refs/changes/68/531568/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/531568/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/531568/29"}}},"commit":{"parents":[{"commit":"c07415d725c5f04d46e06cb72ed441250bad73a0","subject":"Scenario test: Create/shrink share and write data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c07415d725c5f04d46e06cb72ed441250bad73a0"}]}],"author":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2018-01-06 18:49:09.000000000","tz":-480},"committer":{"name":"Nir Gilboa","email":"gilboa.nir@gmail.com","date":"2019-05-19 07:11:58.000000000","tz":180},"subject":"Scenario test: Create/extend share and write data","message":"Scenario test: Create/extend share and write data\n\nImplements #4 from: http://specs.openstack.org/openstack/manila-specs/specs/release_independent/scenario-tests.html\n\nChange-Id: I4998bcac7a1d902c17d2add00cd42560bf6bbf75\nDepends-On: https://review.openstack.org/#/c/545695/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2cb85a6edd76ae94b428cc504c9e25f419618c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2cb85a6edd76ae94b428cc504c9e25f419618c9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
