)]}'
{"id":"openstack%2Fmanila-tempest-plugin~629190","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie61fa513d8cbf08a997054ccfb7b09b918c78e22","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1810988","hashtags":[],"change_id":"Ie61fa513d8cbf08a997054ccfb7b09b918c78e22","subject":"DNM - investigate ssh timeouts in tempest","status":"NEW","created":"2019-01-08 13:51:11.000000000","updated":"2019-09-23 03:25:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d7b50ae479b47ff5c6b5a25ccf45b4ba0cb0294","_number":629190,"virtual_id_number":629190,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"date":"2019-09-23 03:25:19.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":1,"date":"2019-09-22 22:26:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-09 08:41:46.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2019-09-22 22:26:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-23 03:25:19.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b84cde195faa0dd2cabe9d34e7ab46dc65dacad7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-01-08 13:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3793bae25c3d5d777e42fd34fcb2edfb74021ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 15:17:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/90/629190/1/check/openstack-tox-pep8/06da2e8/ : FAILURE in 5m 23s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/90/629190/1/check/manila-tempest-dsvm-mysql-generic/33cc4d7/ : FAILURE in 1h 18m 27s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/90/629190/1/check/manila-tempest-dsvm-postgres-container/65ab934/ : SUCCESS in 46m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/90/629190/1/check/manila-tempest-dsvm-postgres-zfsonlinux/efc88d6/ : SUCCESS in 1h 14m 07s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/90/629190/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/e52ee85/ : SUCCESS in 57m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/90/629190/1/check/manila-tempest-dsvm-generic-no-share-servers/98bda77/ : SUCCESS in 49m 35s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/90/629190/1/check/manila-tempest-dsvm-scenario/a6ec04a/ : FAILURE in 1h 12m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/90/629190/1/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/dd76089/ : SUCCESS in 38m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/90/629190/1/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/e1069e1/ : SUCCESS in 34m 01s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/90/629190/1/check/manila-tempest-minimal-dsvm-dummy/e6ae877/ : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2a201b3bf040e9aede9d761b96c9dc4a222543d0","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-01-08 19:40:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc6a2eb903c32e3664bacbb94ceffafca8291c94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 21:05:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/629190/2/check/openstack-tox-pep8/746bbd7/ : SUCCESS in 4m 29s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/90/629190/2/check/manila-tempest-dsvm-mysql-generic/891506b/ : FAILURE in 1h 22m 07s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/90/629190/2/check/manila-tempest-dsvm-postgres-container/c0eaf40/ : SUCCESS in 51m 37s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/90/629190/2/check/manila-tempest-dsvm-postgres-zfsonlinux/0023f35/ : SUCCESS in 1h 17m 09s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/90/629190/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/7e02afa/ : SUCCESS in 56m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/90/629190/2/check/manila-tempest-dsvm-generic-no-share-servers/981dc3c/ : SUCCESS in 1h 00m 33s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/90/629190/2/check/manila-tempest-dsvm-scenario/1ed1532/ : FAILURE in 1h 21m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/90/629190/2/check/manila-tempest-minimal-dsvm-cephfs-native/6834c54/ : SUCCESS in 26m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/90/629190/2/check/manila-tempest-minimal-dsvm-cephfs-nfs/9ce9a8d/ : SUCCESS in 27m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/90/629190/2/check/manila-tempest-minimal-dsvm-dummy/9a745bb/ : SUCCESS in 38m 19s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/90/629190/2/check/manila-tempest-minimal-dsvm-lvm/a37dfee/ : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4df5ff8c94ef833857d0bd03f1e70ef89b539cc9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-01-08 21:25:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/90/629190/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5da675c/ : FAILURE in 50s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/90/629190/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/be18eb0/ : FAILURE in 1m 51s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"77060296e8ea3b06cab83b03e9c88acd5f72af94","author":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"date":"2019-01-09 08:41:46.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHey Tom, looks good, please see my inline comment.","accounts_in_message":[],"_revision_number":2},{"id":"24067bf0458f005b176eab3c3abac2911fa8f9a3","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-01-09 10:35:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d396135c6f621661a8c2a47563e10598c80c2ff","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-01-09 13:05:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d9c99eb2061d10e3bde86381b29a926a00b37c97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 14:29:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/90/629190/3/check/manila-tempest-dsvm-mysql-generic/9272d5e/ : FAILURE in 1h 12m 53s\n- manila-tempest-dsvm-scenario http://logs.openstack.org/90/629190/3/check/manila-tempest-dsvm-scenario/a3d3511/ : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":3},{"id":"cea4a23e9973e89117ac16aee250bdea3a2935d0","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-01-29 16:20:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4de0d0d8447f90814755d6329590204b23ab3bee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 17:48:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/90/629190/4/check/manila-tempest-dsvm-mysql-generic/de7caee/ : FAILURE in 1h 13m 06s\n- manila-tempest-dsvm-scenario http://logs.openstack.org/90/629190/4/check/manila-tempest-dsvm-scenario/c884ca5/ : FAILURE in 1h 23m 01s","accounts_in_message":[],"_revision_number":4},{"id":"68e983b22958d97066b5259f50b97a874cae02bb","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-02-08 11:12:29.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"a80bb55afad6ad22b4bc976b6467d4247356822d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-03-07 14:08:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a58fe0e511eb937ebe3d9d4b66e4e5b64f289c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 15:46:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/90/629190/5/check/manila-tempest-dsvm-mysql-generic/77ac97b/ : FAILURE in 1h 20m 15s\n- manila-tempest-dsvm-scenario http://logs.openstack.org/90/629190/5/check/manila-tempest-dsvm-scenario/32bee4f/ : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":5},{"id":"74f4b7c837618dea277a2ece90afe56cf31ae4f5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-09-22 20:50:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"f4ea70d42eb2753b57f45d97d7686acd5aa19762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-22 22:26:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/15fda852cf11420097f690187df67191 : SUCCESS in 1h 30m 12s\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5e83c56b9f2943bba1e1d5609f60f873 : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":6},{"id":"1f07c5f1e29530840cae4b15bbf8ee3d6997e4c0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-23 03:25:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.159.105/logs/90/629190/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/576079a/ : POST_FAILURE in 1h 32m 12s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.159.105/logs/90/629190/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/22165e8/ : POST_FAILURE in 1h 37m 50s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3cff2fc38e32c9b8f4701e0443f78cf83f470eb7","revisions":{"f5751c7665df9ead639474369747c9da1b04e4b6":{"kind":"REWORK","_number":1,"created":"2019-01-08 13:51:11.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/90/629190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/90/629190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/1"}}},"commit":{"parents":[{"commit":"f9d49aba022d8312f3bfae42a9e7404f1aa04a98","subject":"Merge \"Extend testing for reset state operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9d49aba022d8312f3bfae42a9e7404f1aa04a98"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"subject":"DNM - test setting channel_timeout for ssh client","message":"DNM - test setting channel_timeout for ssh client\n\nChange-Id: Ie61fa513d8cbf08a997054ccfb7b09b918c78e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5751c7665df9ead639474369747c9da1b04e4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5751c7665df9ead639474369747c9da1b04e4b6"}]},"branch":"refs/heads/master"},"b24cbaeb867a7d9835199025e4034542d4535ca0":{"kind":"REWORK","_number":2,"created":"2019-01-08 19:40:34.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/90/629190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/90/629190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/2"}}},"commit":{"parents":[{"commit":"df99c2db0c9a1bf36e9c614555b20aa2cb7680d9","subject":"Migrate CentOS jobs to ubuntu bionic beaver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df99c2db0c9a1bf36e9c614555b20aa2cb7680d9"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 19:40:14.000000000","tz":-300},"subject":"[tempest] set channel_timeout for tempest ssh client","message":"[tempest] set channel_timeout for tempest ssh client\n\nto match the ssh timeouts we set for paramiko in the\nmanila share service.\n\nCloses-bug: 1810988\nChange-Id: Ie61fa513d8cbf08a997054ccfb7b09b918c78e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b24cbaeb867a7d9835199025e4034542d4535ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b24cbaeb867a7d9835199025e4034542d4535ca0"}]},"branch":"refs/heads/master"},"63603dcdcbfdac914bccdd62543c0157754e593d":{"kind":"REWORK","_number":3,"created":"2019-01-09 13:05:03.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/90/629190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/90/629190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/3"}}},"commit":{"parents":[{"commit":"df99c2db0c9a1bf36e9c614555b20aa2cb7680d9","subject":"Migrate CentOS jobs to ubuntu bionic beaver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df99c2db0c9a1bf36e9c614555b20aa2cb7680d9"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-09 13:04:45.000000000","tz":-300},"subject":"DNM - investigate ssh timeouts in tempest","message":"DNM - investigate ssh timeouts in tempest\n\nChange-Id: Ie61fa513d8cbf08a997054ccfb7b09b918c78e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/63603dcdcbfdac914bccdd62543c0157754e593d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/63603dcdcbfdac914bccdd62543c0157754e593d"}]},"branch":"refs/heads/master"},"8e018ae525d1f980546f086934739a7e6b58e913":{"kind":"REWORK","_number":4,"created":"2019-01-29 16:20:12.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/90/629190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/90/629190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/4"}}},"commit":{"parents":[{"commit":"3f92a6a82772d8b49abaf9e30cd74f267b9d1e81","subject":"Merge \"Fix availability zones tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f92a6a82772d8b49abaf9e30cd74f267b9d1e81"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-29 16:01:20.000000000","tz":-300},"subject":"DNM - investigate ssh timeouts in tempest","message":"DNM - investigate ssh timeouts in tempest\n\nChange-Id: Ie61fa513d8cbf08a997054ccfb7b09b918c78e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e018ae525d1f980546f086934739a7e6b58e913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8e018ae525d1f980546f086934739a7e6b58e913"}]},"branch":"refs/heads/master"},"0834ddb2ff267ca0d32057ea1ae8669eb29b661c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-07 14:08:46.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/90/629190/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/90/629190/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/5"}}},"commit":{"parents":[{"commit":"370569423eea5fad86efd646e7c5ea83e1337420","subject":"Add functional tests for manage/unmanage of shares in DHSS\u003dTrue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/370569423eea5fad86efd646e7c5ea83e1337420"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-03-07 14:08:46.000000000","tz":0},"subject":"DNM - investigate ssh timeouts in tempest","message":"DNM - investigate ssh timeouts in tempest\n\nChange-Id: Ie61fa513d8cbf08a997054ccfb7b09b918c78e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0834ddb2ff267ca0d32057ea1ae8669eb29b661c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0834ddb2ff267ca0d32057ea1ae8669eb29b661c"}]},"branch":"refs/heads/master"},"3cff2fc38e32c9b8f4701e0443f78cf83f470eb7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-09-22 20:50:24.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/90/629190/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/90/629190/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/90/629190/6"}}},"commit":{"parents":[{"commit":"171241eb94787293d8711a2284ccd7684b2f45bf","subject":"Merge \"Enable share-replica tests in multitenancy enviroments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/171241eb94787293d8711a2284ccd7684b2f45bf"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-01-08 13:50:58.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-09-22 20:50:24.000000000","tz":0},"subject":"DNM - investigate ssh timeouts in tempest","message":"DNM - investigate ssh timeouts in tempest\n\nChange-Id: Ie61fa513d8cbf08a997054ccfb7b09b918c78e22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cff2fc38e32c9b8f4701e0443f78cf83f470eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cff2fc38e32c9b8f4701e0443f78cf83f470eb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
