)]}'
{"id":"openstack%2Fmanila-tempest-plugin~638960","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1b772fb22ccd69e133e59635a7fa7afbe20edb2a","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I1b772fb22ccd69e133e59635a7fa7afbe20edb2a","subject":"Update json module to jsonutils","status":"ABANDONED","created":"2019-02-24 14:35:52.000000000","updated":"2019-12-19 14:53:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"93f1bba3bc27345e7713695ccdd4330ed07a49d7","_number":638960,"virtual_id_number":638960,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-05 10:02:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-24 17:03:28.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-02-24 17:29:32.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-27 13:06:12.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-27 13:33:18.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-07-05 05:42:29.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-05 10:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50284c303be34736ea5eb3e663f746e4c61bb550","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-02-24 14:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11386f8d5bc070978f888523e7285c402c1b1ba9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-24 16:14:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/60/638960/1/check/test-release-openstack/a2a62f4/ : SUCCESS in 3m 22s\n- requirements-check http://logs.openstack.org/60/638960/1/check/requirements-check/2bc3402/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/60/638960/1/check/openstack-tox-pep8/5f97b7a/ : SUCCESS in 6m 17s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/60/638960/1/check/manila-tempest-dsvm-mysql-generic/45d6118/ : FAILURE in 1h 35m 32s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/60/638960/1/check/manila-tempest-dsvm-postgres-container/79a07dd/ : SUCCESS in 38m 19s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/60/638960/1/check/manila-tempest-dsvm-postgres-zfsonlinux/a2800d8/ : FAILURE in 1h 11m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/60/638960/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/a07c363/ : SUCCESS in 49m 18s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/60/638960/1/check/manila-tempest-dsvm-generic-no-share-servers/3375dad/ : SUCCESS in 1h 13m 03s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/60/638960/1/check/manila-tempest-dsvm-scenario/74cc21a/ : SUCCESS in 1h 19m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/60/638960/1/check/manila-tempest-minimal-dsvm-cephfs-native/d1e6d3d/ : SUCCESS in 29m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/60/638960/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/7858f77/ : SUCCESS in 20m 38s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/60/638960/1/check/manila-tempest-minimal-dsvm-dummy/7f338b2/ : SUCCESS in 34m 28s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/60/638960/1/check/manila-tempest-minimal-dsvm-lvm/803a1ad/ : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"48e15eb5b9561ee0eea1bb2a3b069021a109305c","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-02-24 17:03:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d5f42f9100ae25cb3d6430d300e825262fcc2cb2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-02-24 17:29:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/60/638960/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/3db4698/ : FAILURE in 49m 18s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/60/638960/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/fcdd4e1/ : SUCCESS in 1h 14m 30s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1ee74ada59c5eaf35decd8fe4349cc3a56a1e44e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-03-05 15:04:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"154eefb0fc40476074f5a92e0234110bef83bf8c","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-03-22 08:03:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3bc7ec0e6711f724d05c57fcca12721d3ffbfd81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 09:41:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/60/638960/2/check/test-release-openstack/64a6326/ : SUCCESS in 3m 06s\n- requirements-check http://logs.openstack.org/60/638960/2/check/requirements-check/1634819/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/60/638960/2/check/openstack-tox-pep8/7da04cb/ : SUCCESS in 4m 35s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/60/638960/2/check/manila-tempest-dsvm-mysql-generic/70f9b66/ : SUCCESS in 1h 32m 27s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/60/638960/2/check/manila-tempest-dsvm-postgres-container/011e083/ : SUCCESS in 51m 30s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/60/638960/2/check/manila-tempest-dsvm-postgres-zfsonlinux/f64d2c7/ : SUCCESS in 1h 15m 57s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/60/638960/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/bf02332/ : SUCCESS in 1h 02m 34s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/60/638960/2/check/manila-tempest-dsvm-generic-no-share-servers/b39d9e0/ : SUCCESS in 1h 14m 15s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/60/638960/2/check/manila-tempest-dsvm-scenario/6ae0a10/ : SUCCESS in 1h 18m 55s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/60/638960/2/check/manila-tempest-minimal-dsvm-cephfs-native/88e059a/ : SUCCESS in 30m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/60/638960/2/check/manila-tempest-minimal-dsvm-cephfs-nfs/caa9223/ : SUCCESS in 23m 01s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/60/638960/2/check/manila-tempest-minimal-dsvm-dummy/94f5edc/ : FAILURE in 44m 34s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/60/638960/2/check/manila-tempest-minimal-dsvm-lvm/3920a74/ : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"86425840c2cb9e1a5162daac019ec8be00bf1774","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-27 13:06:12.000000000","message":"Patch Set 2: Code-Review-1\n\nplease fix the conflict","accounts_in_message":[],"_revision_number":2},{"id":"1e7617571ec254a2b778cac66c3d09b4c2b9b1bf","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-27 13:33:18.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"661e7c600a2bbae6a511f3b37d18c2d8455158c9","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-05 05:42:29.000000000","message":"Patch Set 2: Code-Review-1\n\nplease fix the conflict","accounts_in_message":[],"_revision_number":2},{"id":"d2813a602e296ab9e97151cc8b96fa5810d546a0","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 05:46:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32eabf2178ab94f962f1f1ab6f980c6b2c41eb68","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 05:47:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"96cb108e58635c5841d31d935ed46704f01dbb1f","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 06:04:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"108e13bc6ada3943ae2fd4e93adf0b33d25d1d5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 07:54:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/60/638960/5/check/requirements-check/5773d7a/ : FAILURE in 2m 20s\n- openstack-tox-pep8 http://logs.openstack.org/60/638960/5/check/openstack-tox-pep8/71561e2/ : FAILURE in 4m 59s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/60/638960/5/check/manila-tempest-dsvm-mysql-generic/8cf1a2a/ : FAILURE in 1h 09m 04s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/60/638960/5/check/manila-tempest-dsvm-postgres-container/0d2027b/ : FAILURE in 47m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/60/638960/5/check/manila-tempest-dsvm-postgres-zfsonlinux/23c087f/ : FAILURE in 57m 53s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/60/638960/5/check/manila-tempest-dsvm-postgres-generic-singlebackend/07d0fc0/ : SUCCESS in 1h 01m 03s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/60/638960/5/check/manila-tempest-dsvm-generic-no-share-servers/9695a6f/ : FAILURE in 48m 16s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/60/638960/5/check/manila-tempest-dsvm-scenario/9ee29dc/ : FAILURE in 1h 18m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/60/638960/5/check/manila-tempest-minimal-dsvm-cephfs-native/b6b8131/ : FAILURE in 31m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/60/638960/5/check/manila-tempest-minimal-dsvm-cephfs-nfs/5c57ad9/ : FAILURE in 1h 39m 13s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/60/638960/5/check/manila-tempest-minimal-dsvm-dummy/b3c8d8d/ : FAILURE in 40m 09s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/60/638960/5/check/manila-tempest-minimal-dsvm-lvm/ea49b42/ : FAILURE in 1h 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"d5ce152981f266b239eb0eb491102a81375d79b0","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-05 08:08:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"db2ac53c4ab0118016675791c974be6baab7f022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:02:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/60/638960/6/check/requirements-check/52d4c2a/ : FAILURE in 2m 01s\n- openstack-tox-pep8 http://logs.openstack.org/60/638960/6/check/openstack-tox-pep8/e3251c0/ : SUCCESS in 4m 00s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/60/638960/6/check/manila-tempest-dsvm-mysql-generic/25b2eef/ : SUCCESS in 1h 36m 41s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/60/638960/6/check/manila-tempest-dsvm-postgres-container/51b52b0/ : SUCCESS in 58m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/60/638960/6/check/manila-tempest-dsvm-postgres-zfsonlinux/f586557/ : SUCCESS in 1h 08m 59s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/60/638960/6/check/manila-tempest-dsvm-postgres-generic-singlebackend/33f56b1/ : FAILURE in 1h 00m 36s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/60/638960/6/check/manila-tempest-dsvm-generic-no-share-servers/d8a16ba/ : FAILURE in 1h 50m 38s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/60/638960/6/check/manila-tempest-dsvm-scenario/8d84575/ : FAILURE in 1h 25m 40s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/60/638960/6/check/manila-tempest-minimal-dsvm-cephfs-native/fe9acd4/ : FAILURE in 28m 27s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/60/638960/6/check/manila-tempest-minimal-dsvm-cephfs-nfs/b532682/ : SUCCESS in 1h 24m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/60/638960/6/check/manila-tempest-minimal-dsvm-dummy/68867a2/ : SUCCESS in 37m 01s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/60/638960/6/check/manila-tempest-minimal-dsvm-lvm/2787d41/ : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":6},{"id":"93f1bba3bc27345e7713695ccdd4330ed07a49d7","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-12-19 14:53:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ae171512d9d3a9f2a92fe90e8f07910bcaceca41","revisions":{"165a7fbb5c0a11c56bec71ebd583385849fc50f3":{"kind":"REWORK","_number":1,"created":"2019-02-24 14:35:52.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/638960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/638960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/1"}}},"commit":{"parents":[{"commit":"50a6760a64bc67ad6875db1496c501ed77b55b4e","subject":"Fix microversion request URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50a6760a64bc67ad6875db1496c501ed77b55b4e"}]}],"author":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:33:25.000000000","tz":480},"committer":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:35:48.000000000","tz":480},"subject":"Update json module to jsonutils","message":"Update json module to jsonutils\n\noslo project provide jsonutils, and the others use it\nin many place, this PS to update the json moudule to\noslo jsonutils for consistency.\n\nChange-Id: I1b772fb22ccd69e133e59635a7fa7afbe20edb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/165a7fbb5c0a11c56bec71ebd583385849fc50f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/165a7fbb5c0a11c56bec71ebd583385849fc50f3"}]},"branch":"refs/heads/master"},"29aa1473b0462db2ecb93a002203780f4e6476b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-03-22 08:03:00.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/638960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/638960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/2"}}},"commit":{"parents":[{"commit":"50a6760a64bc67ad6875db1496c501ed77b55b4e","subject":"Fix microversion request URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/50a6760a64bc67ad6875db1496c501ed77b55b4e"}]}],"author":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:33:25.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-03-22 08:02:57.000000000","tz":0},"subject":"Update json module to jsonutils","message":"Update json module to jsonutils\n\noslo project provide jsonutils, and the others use it\nin many place, this PS to update the json module to\noslo jsonutils for consistency.\n\nChange-Id: I1b772fb22ccd69e133e59635a7fa7afbe20edb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/29aa1473b0462db2ecb93a002203780f4e6476b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/29aa1473b0462db2ecb93a002203780f4e6476b0"}]},"branch":"refs/heads/master"},"36c1553d0d539b3e383aedb079df9b2be61d0f8b":{"kind":"REWORK","_number":3,"created":"2019-07-05 05:46:26.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/638960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/638960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/3"}}},"commit":{"parents":[{"commit":"0ac95412b1b0555775b4d45f51c27af2fa84724a","subject":"Look for api microversion header in \"Vary\" header data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ac95412b1b0555775b4d45f51c27af2fa84724a"}]}],"author":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:33:25.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 05:46:05.000000000","tz":480},"subject":"Update json module to jsonutils","message":"Update json module to jsonutils\n\noslo project provide jsonutils, and the others use it\nin many place, this PS to update the json module to\noslo jsonutils for consistency.\n\nChange-Id: I1b772fb22ccd69e133e59635a7fa7afbe20edb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36c1553d0d539b3e383aedb079df9b2be61d0f8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36c1553d0d539b3e383aedb079df9b2be61d0f8b"}]},"branch":"refs/heads/master"},"1120976a6deda15d622b810b163be5e635593f15":{"kind":"REWORK","_number":4,"created":"2019-07-05 05:47:29.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/638960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/638960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/4"}}},"commit":{"parents":[{"commit":"0ac95412b1b0555775b4d45f51c27af2fa84724a","subject":"Look for api microversion header in \"Vary\" header data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ac95412b1b0555775b4d45f51c27af2fa84724a"}]}],"author":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:33:25.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 05:47:22.000000000","tz":480},"subject":"Update json module to jsonutils","message":"Update json module to jsonutils\n\noslo project provide jsonutils, and the others use it\nin many place, this PS to update the json module to\noslo jsonutils for consistency.\n\nChange-Id: I1b772fb22ccd69e133e59635a7fa7afbe20edb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1120976a6deda15d622b810b163be5e635593f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1120976a6deda15d622b810b163be5e635593f15"}]},"branch":"refs/heads/master"},"939bd2c39f3eb077a0d53beff0305c9386d0dda8":{"kind":"REWORK","_number":5,"created":"2019-07-05 06:04:29.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/638960/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/638960/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/5"}}},"commit":{"parents":[{"commit":"0ac95412b1b0555775b4d45f51c27af2fa84724a","subject":"Look for api microversion header in \"Vary\" header data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ac95412b1b0555775b4d45f51c27af2fa84724a"}]}],"author":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:33:25.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 06:04:22.000000000","tz":480},"subject":"Update json module to jsonutils","message":"Update json module to jsonutils\n\noslo project provide jsonutils, and the others use it\nin many place, this PS to update the json module to\noslo jsonutils for consistency.\n\nChange-Id: I1b772fb22ccd69e133e59635a7fa7afbe20edb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/939bd2c39f3eb077a0d53beff0305c9386d0dda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/939bd2c39f3eb077a0d53beff0305c9386d0dda8"}]},"branch":"refs/heads/master"},"ae171512d9d3a9f2a92fe90e8f07910bcaceca41":{"kind":"REWORK","_number":6,"created":"2019-07-05 08:08:37.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/60/638960/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/60/638960/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/60/638960/6"}}},"commit":{"parents":[{"commit":"0ac95412b1b0555775b4d45f51c27af2fa84724a","subject":"Look for api microversion header in \"Vary\" header data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ac95412b1b0555775b4d45f51c27af2fa84724a"}]}],"author":{"name":"cao.yuan","email":"cao.yuan@99cloud.net","date":"2019-02-24 14:33:25.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-07-05 08:08:25.000000000","tz":480},"subject":"Update json module to jsonutils","message":"Update json module to jsonutils\n\noslo project provide jsonutils, and the others use it\nin many place, this PS to update the json module to\noslo jsonutils for consistency.\n\nChange-Id: I1b772fb22ccd69e133e59635a7fa7afbe20edb2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ae171512d9d3a9f2a92fe90e8f07910bcaceca41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ae171512d9d3a9f2a92fe90e8f07910bcaceca41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
