)]}'
{"id":"openstack%2Fmanila-tempest-plugin~664434","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-cifs","hashtags":[],"change_id":"Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b","subject":"Skip read-only test for CIFS","status":"MERGED","created":"2019-06-10 20:06:02.000000000","updated":"2019-06-20 21:11:59.000000000","submitted":"2019-06-20 21:11:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664434-1561065119747-a6cfe23d","meta_rev_id":"4ee0ce40040e1c030686636ce79c45234376f2f4","_number":664434,"virtual_id_number":664434,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2019-06-20 18:04:43.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"date":"2019-06-13 02:05:42.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-06-20 21:11:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-06-12 23:32:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-13 09:57:33.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 01:09:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 09:57:33.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 18:59:30.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2019-06-12 19:07:52.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2019-06-12 23:32:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-06-13 01:09:14.000000000","updated_by":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"reviewer":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-06-13 02:05:42.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-13 09:57:33.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-06-20 21:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"477dc34760e3973f563a165e94cdee460ce87982","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-10 20:06:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a50f20309676f55e6af2e18ceccb3e6bc50e2c8","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2019-06-10 21:07:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bfa47fd78467f120f13e0169fbbaf32710913c8","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-10 21:28:33.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f74d0252ffc1653d1ea05600628bd1ec3355818d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-10 21:28:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a170868a30354b2a8e91bc30e3a14a3081061bd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 23:04:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/2/check/openstack-tox-pep8/44df89b/ : FAILURE in 6m 04s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/2/check/manila-tempest-dsvm-mysql-generic/de87829/ : FAILURE in 1h 07m 15s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/2/check/manila-tempest-dsvm-postgres-container/336e75f/ : SUCCESS in 54m 16s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/2/check/manila-tempest-dsvm-postgres-zfsonlinux/4a9c52c/ : FAILURE in 1h 13m 43s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/adc4104/ : FAILURE in 59m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/2/check/manila-tempest-dsvm-generic-no-share-servers/8704083/ : FAILURE in 1h 06m 33s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/2/check/manila-tempest-dsvm-scenario/78a91e9/ : SUCCESS in 1h 29m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/2/check/manila-tempest-minimal-dsvm-cephfs-native/13074da/ : SUCCESS in 20m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/2/check/manila-tempest-minimal-dsvm-cephfs-nfs/51f4fc6/ : SUCCESS in 29m 56s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/2/check/manila-tempest-minimal-dsvm-dummy/26c6829/ : SUCCESS in 38m 20s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/2/check/manila-tempest-minimal-dsvm-lvm/c1cd56c/ : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e0b3a7d39100a182331dcc0e5d661b6deea15a88","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-11 06:39:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"400eb7ef256048e312c1516fe596abdce019eb89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:43:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/3/check/openstack-tox-pep8/fab3e2d/ : SUCCESS in 4m 35s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-mysql-generic/ae5e1bc/ : FAILURE in 1h 28m 02s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-container/6c766e0/ : SUCCESS in 51m 02s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-zfsonlinux/e6d1ffa/ : SUCCESS in 1h 06m 38s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-generic-singlebackend/c4bf0f1/ : SUCCESS in 43m 56s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-generic-no-share-servers/b6137d1/ : SUCCESS in 1h 19m 22s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-scenario/926e7fe/ : SUCCESS in 1h 30m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-cephfs-native/c2e3f45/ : SUCCESS in 25m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-cephfs-nfs/8b6bd12/ : SUCCESS in 24m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-dummy/725039a/ : SUCCESS in 36m 23s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-lvm/d0d113d/ : FAILURE in 27m 11s","accounts_in_message":[],"_revision_number":3},{"id":"3df6a2e9952abd8f7735b811b60a2ee6fd68fc33","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-11 10:30:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f97e3047d35bd508398c1cfed571245fc9f92ede","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:13:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/3/check/openstack-tox-pep8/c17e161/ : SUCCESS in 4m 23s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-mysql-generic/9149f53/ : SUCCESS in 1h 32m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-container/8d6c1bb/ : SUCCESS in 48m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-zfsonlinux/f1072b2/ : SUCCESS in 1h 16m 04s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-generic-singlebackend/9c4bc68/ : FAILURE in 1h 04m 17s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-generic-no-share-servers/44dfba4/ : SUCCESS in 1h 01m 46s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-scenario/dac4820/ : SUCCESS in 1h 28m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-cephfs-native/9961cdb/ : SUCCESS in 24m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-cephfs-nfs/ba96e16/ : SUCCESS in 24m 12s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-dummy/30ff4e6/ : FAILURE in 45m 58s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-lvm/7a3de25/ : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e01f63d1a2758677679385cb060fa3b701fd5b3f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-11 12:22:10.000000000","message":"Patch Set 3:\n\nrecheck\n\nsigh, CI seems flaky again","accounts_in_message":[],"_revision_number":3},{"id":"2f0d8727fa475065f232af9db4cb0334ed60ac4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 14:01:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/3/check/openstack-tox-pep8/2607ee9/ : SUCCESS in 5m 51s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-mysql-generic/13ad164/ : FAILURE in 1h 25m 25s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-container/e6d591b/ : SUCCESS in 50m 24s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-zfsonlinux/00ae535/ : SUCCESS in 54m 25s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-postgres-generic-singlebackend/6b41cb5/ : FAILURE in 54m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-generic-no-share-servers/6469aed/ : FAILURE in 1h 03m 00s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/3/check/manila-tempest-dsvm-scenario/254135e/ : SUCCESS in 1h 26m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-cephfs-native/1e12642/ : SUCCESS in 25m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-cephfs-nfs/db1b324/ : SUCCESS in 25m 32s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-dummy/bb4bf44/ : SUCCESS in 41m 21s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/3/check/manila-tempest-minimal-dsvm-lvm/c7ee561/ : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b33b90e8e2d6b3eeb9fb1d1ae7824aee6d3d9370","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-11 16:18:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/34/664434/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/336dd48/ : FAILURE in 2h 15m 46s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/34/664434/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1b4bbba/ : SUCCESS in 1h 31m 10s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b7b2ca4a896a11e434fc44003aa6f270e0000927","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-06-12 18:06:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nI\u0027ve added a comment in an added message that you may consider if you really need to upload a new PS :)... Otherwise, LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"060bf43e61af783a5de89164bfd3b9164dbfdd58","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-12 18:57:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"49c5ce60ec4294972c9c9448c28a7b3ea10ea136","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-12 18:57:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"879405e493ffd006e66c941b6517f421151e840f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-06-12 18:59:30.000000000","message":"Patch Set 4: Code-Review+1\n\nGreat... LGTM :)","accounts_in_message":[],"_revision_number":4},{"id":"58cb4cc472fffd8adf8ff3bf90b0d459d84b8aa5","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2019-06-12 19:07:52.000000000","message":"Patch Set 4:\n\nDo we have any CI skipping this test as expected?","accounts_in_message":[],"_revision_number":4},{"id":"d10cfdd98c67e2c969f8389228a75a93ec4330f9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-12 20:00:26.000000000","message":"Patch Set 4:\n\nHi Colleen, \n\nThanks for this patch!\n\nWhat Manila Share backend are you testing CIFS with? CIFS with IP access is a sketchy combination we have as a community - We certainly don\u0027t test it. \n\nThe default way of authenticating CIFS is with users, perhaps tied with an external security service such as Active Directory.","accounts_in_message":[],"_revision_number":4},{"id":"c80614507a35619c7821250a12c8bb521b545916","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-12 20:31:39.000000000","message":"Patch Set 4:\n\n\u003e Hi Colleen,\n \u003e \n \u003e Thanks for this patch!\n \u003e \n \u003e What Manila Share backend are you testing CIFS with? CIFS with IP\n \u003e access is a sketchy combination we have as a community - We\n \u003e certainly don\u0027t test it.\n \u003e \n \u003e The default way of authenticating CIFS is with users, perhaps tied\n \u003e with an external security service such as Active Directory.\n\nHi Goutham, this is just with the generic backend: share_driver\u003dmanila.share.drivers.generic.GenericShareDriver\n\nIf I understand you correctly, it sounds like you agree that this test should be skipped?","accounts_in_message":[],"_revision_number":4},{"id":"0b24d4003bccfb7fd610cee401e824d5ec201d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:49:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/4/check/openstack-tox-pep8/5ea9acd/ : SUCCESS in 4m 36s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/4/check/manila-tempest-dsvm-mysql-generic/bd989f9/ : FAILURE in 1h 16m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/4/check/manila-tempest-dsvm-postgres-container/3a74d0a/ : SUCCESS in 52m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/4/check/manila-tempest-dsvm-postgres-zfsonlinux/ccb4d6a/ : SUCCESS in 1h 15m 26s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/4/check/manila-tempest-dsvm-postgres-generic-singlebackend/fb670e0/ : SUCCESS in 1h 04m 00s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/4/check/manila-tempest-dsvm-generic-no-share-servers/930671f/ : FAILURE in 1h 05m 29s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/4/check/manila-tempest-dsvm-scenario/271677b/ : FAILURE in 1h 48m 55s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/4/check/manila-tempest-minimal-dsvm-cephfs-native/26651a3/ : SUCCESS in 29m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/4/check/manila-tempest-minimal-dsvm-cephfs-nfs/8bf38d6/ : SUCCESS in 32m 07s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/4/check/manila-tempest-minimal-dsvm-dummy/f0b0f2f/ : SUCCESS in 35m 53s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/4/check/manila-tempest-minimal-dsvm-lvm/52149b5/ : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":4},{"id":"e783e88a8bb5d5dbe69732b682392910ade726b5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-12 22:10:12.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e Hi Goutham, this is just with the generic backend: share_driver\u003dmanila.share.drivers.generic.GenericShareDriver\n\u003e If I understand you correctly, it sounds like you agree that this test should be skipped?\n\n\nThanks, I\u0027m okay with skipping given the current capability of these scenario tests. These tests only test IP based access control (See [a]). \"user\" based access control scenario testing has been on a wishlist for CIFS protocol shares. \n\nPlease see some comments inline regarding the link/explanation in the commit message. \n\n[a] https://opendev.org/openstack/manila-tempest-plugin/src/commit/a4904978b4c28e1acbfb12d26db63085325f94c3/manila_tempest_tests/tests/scenario/manager_share.py#L72-L75","accounts_in_message":[],"_revision_number":4},{"id":"02361f5c8bc38dc503801d6b68e5a21caae9cb8c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-12 22:12:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/34/664434/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/69722b9/ : SUCCESS in 1h 10m 29s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/34/664434/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8c8ad09/ : SUCCESS in 1h 22m 45s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"991f3eaedeed60c00a9218cc8e7a86d15e9af4b1","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-12 22:45:26.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"2c3e523cd037fef8b18f54794da22bb3b341f0c6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-12 23:32:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4ee350e293df69cc261c77b8292fc123ebfa5433","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 00:21:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/5/check/openstack-tox-pep8/4aaeef4/ : SUCCESS in 4m 07s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-mysql-generic/89ad2f4/ : FAILURE in 1h 24m 34s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-postgres-container/e4fdaf1/ : SUCCESS in 48m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-postgres-zfsonlinux/9e32ec2/ : SUCCESS in 1h 02m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-postgres-generic-singlebackend/52f2b49/ : SUCCESS in 44m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-generic-no-share-servers/23d0d74/ : FAILURE in 1h 11m 14s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-scenario/9b33dc8/ : SUCCESS in 1h 30m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-cephfs-native/9f62d29/ : SUCCESS in 27m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-cephfs-nfs/f34f7cc/ : SUCCESS in 29m 55s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-dummy/9446a83/ : SUCCESS in 42m 23s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-lvm/fcb877f/ : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":5},{"id":"94327559a1049c9c8e7b2251ff00a5442845e11b","author":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"date":"2019-06-13 01:09:14.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d4df6d9ddad3e8ae0c45dc530d172b2b714b5fc4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-13 02:05:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss http://13.56.179.158/logs/34/664434/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/d758772/ : SUCCESS in 1h 10m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss http://13.56.179.158/logs/34/664434/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6b6514a/ : SUCCESS in 1h 23m 22s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"85bd08ca870ba82f14ee36d4203fab965602c871","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-13 09:57:21.000000000","message":"Patch Set 5:\n\nThanks for this fix, Colleen!","accounts_in_message":[],"_revision_number":5},{"id":"f4015a6e62d8c15819af51402a20d1e19cd0bfa8","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-13 09:57:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8002599a8899f797e85d656bcf12a12094c20fc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:00:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3c305a4dac54164008e90a4a1fbcb26e5a7f4d16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 17:50:29.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"9817ae9b659b5f0b37b68bbc37fa5aa2b91431a9","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-20 18:04:43.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"07bf23b6b870402f3a8c3d8a4f199f9fd8d0bb40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 20:09:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/5/check/openstack-tox-pep8/1ac0406/ : SUCCESS in 4m 30s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-mysql-generic/3c1b369/ : FAILURE in 1h 15m 05s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-postgres-container/c02fa1e/ : SUCCESS in 42m 55s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-postgres-zfsonlinux/bf0d908/ : FAILURE in 18m 50s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-postgres-generic-singlebackend/fc8224b/ : SUCCESS in 59m 13s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-generic-no-share-servers/3a5d00d/ : SUCCESS in 1h 03m 46s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/34/664434/5/check/manila-tempest-dsvm-scenario/3d8744f/ : FAILURE in 1h 42m 11s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-cephfs-native/05f15f7/ : SUCCESS in 26m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-cephfs-nfs/b09327c/ : FAILURE in 1h 16m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-dummy/f537375/ : SUCCESS in 39m 18s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/5/check/manila-tempest-minimal-dsvm-lvm/7a4e126/ : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":5},{"id":"03378f4859b9db4b1bebbd497e4a8eab7c0372e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 20:09:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"88ff9409ea3d5db05d3669a31cee9b3c9e876644","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:11:59.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/34/664434/5/gate/openstack-tox-pep8/72ba2cf/ : SUCCESS in 5m 06s\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/34/664434/5/gate/manila-tempest-minimal-dsvm-dummy/d47a44a/ : SUCCESS in 38m 27s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/34/664434/5/gate/manila-tempest-minimal-dsvm-lvm/60e56c6/ : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":5},{"id":"e395751864fb8a8a6b5c228ea11a654b3e29d3f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:11:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f38c4c365590f4246fee0e5fffde6db55fa51ae6","revisions":{"24e53e08cae19af5c5e776173ba6db771feaa529":{"kind":"REWORK","_number":1,"created":"2019-06-10 20:06:02.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/34/664434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/664434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/1"}}},"commit":{"parents":[{"commit":"2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98","subject":"Fix CIFS access for non-multitenant setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-07 22:00:48.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-10 20:05:55.000000000","tz":-420},"subject":"Skip read-only test for CIFS","message":"Skip read-only test for CIFS\n\nWhen using the CIFS protocol and trying to share a share by IP, the\nshare fails to instantiate and the manila service produces this error:\n\n  InvalidShareAccessLevel: Invalid or unsupported share access level: ro.\n\nThis is because the RO access level is explicitly denied for a CIFS\nshare shared by IP[1]. This patch causes the test to be skipped in such\na case.\n\n[1] https://opendev.org/openstack/manila/src/commit/35d327e4638aa4bfb2ba68fbc602a827816e782b/manila/share/drivers/helpers.py\n\nChange-Id: Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24e53e08cae19af5c5e776173ba6db771feaa529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/24e53e08cae19af5c5e776173ba6db771feaa529"}]},"branch":"refs/heads/master"},"3704f7a10d768956cf3f4139f20e8601ced8aa3c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 21:28:33.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/34/664434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/664434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/2"}}},"commit":{"parents":[{"commit":"2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98","subject":"Fix CIFS access for non-multitenant setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-07 22:00:48.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-10 21:28:09.000000000","tz":-420},"subject":"Skip read-only test for CIFS","message":"Skip read-only test for CIFS\n\nWhen using the CIFS protocol and trying to share a share by IP, the\nshare fails to instantiate and the manila service produces this error:\n\n  InvalidShareAccessLevel: Invalid or unsupported share access level: ro.\n\nThis is because the RO access level is explicitly denied for a CIFS\nshare shared by IP[1]. This patch causes the test to be skipped in such\na case.\n\n[1] https://opendev.org/openstack/manila/src/commit/35d327e4638aa4bfb2ba68fbc602a827816e782b/manila/share/drivers/helpers.py#L493-L494\n\nChange-Id: Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3704f7a10d768956cf3f4139f20e8601ced8aa3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3704f7a10d768956cf3f4139f20e8601ced8aa3c"}]},"branch":"refs/heads/master"},"7f38e1146ace3f455ed832ad31c56ee4b91f1d8f":{"kind":"REWORK","_number":3,"created":"2019-06-11 06:39:19.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/34/664434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/664434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/3"}}},"commit":{"parents":[{"commit":"2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98","subject":"Fix CIFS access for non-multitenant setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-07 22:00:48.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-11 06:39:11.000000000","tz":-420},"subject":"Skip read-only test for CIFS","message":"Skip read-only test for CIFS\n\nWhen using the CIFS protocol and trying to share a share by IP, the\nshare fails to instantiate and the manila service produces this error:\n\n  InvalidShareAccessLevel: Invalid or unsupported share access level: ro.\n\nThis is because the RO access level is explicitly denied for a CIFS\nshare shared by IP[1]. This patch causes the test to be skipped in such\na case.\n\n[1] https://opendev.org/openstack/manila/src/commit/35d327e4638aa4bfb2ba68fbc602a827816e782b/manila/share/drivers/helpers.py#L493-L494\n\nChange-Id: Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f38e1146ace3f455ed832ad31c56ee4b91f1d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f38e1146ace3f455ed832ad31c56ee4b91f1d8f"}]},"branch":"refs/heads/master"},"08e72ea197210c67040f17b01170333cf48e133d":{"kind":"REWORK","_number":4,"created":"2019-06-12 18:57:38.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/34/664434/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/664434/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/4"}}},"commit":{"parents":[{"commit":"2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98","subject":"Fix CIFS access for non-multitenant setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-07 22:00:48.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-12 18:57:29.000000000","tz":-420},"subject":"Skip read-only test for CIFS","message":"Skip read-only test for CIFS\n\nWhen using the CIFS protocol and trying to share a share by IP, the\nshare fails to instantiate and the manila service produces this error:\n\n  InvalidShareAccessLevel: Invalid or unsupported share access level: ro.\n\nThis is because the RO access level is explicitly denied for a CIFS\nshare shared by IP[1]. This patch causes the test to be skipped in such\na case.\n\n[1] https://opendev.org/openstack/manila/src/commit/35d327e4638aa4bfb2ba68fbc602a827816e782b/manila/share/drivers/helpers.py#L493-L494\n\nChange-Id: Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08e72ea197210c67040f17b01170333cf48e133d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/08e72ea197210c67040f17b01170333cf48e133d"}]},"branch":"refs/heads/master"},"f38c4c365590f4246fee0e5fffde6db55fa51ae6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-06-12 22:45:26.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/34/664434/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/664434/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/664434/5"}}},"commit":{"parents":[{"commit":"2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98","subject":"Fix CIFS access for non-multitenant setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e4f3489f6f4c0f8853b3fdc4c6efb89e2789d98"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-06-07 22:00:48.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-06-12 22:42:26.000000000","tz":-420},"subject":"Skip read-only test for CIFS","message":"Skip read-only test for CIFS\n\nWhen using the CIFS protocol and trying to share a share by IP, the\nshare fails to instantiate and the manila service produces this error:\n\n  InvalidShareAccessLevel: Invalid or unsupported share access level: ro.\n\nThere are no Manila Share backends that support CIFS shares with read\nonly access when authenticating with IP addresses[1]. This patch causes\nthe test to be skipped in such a case.\n\n[1] https://docs.openstack.org/manila/latest/admin/share_back_ends_feature_support_mapping.html#mapping-of-share-drivers-and-share-access-rules-support\n\nChange-Id: Ic30eaf3c2b7f75eb4bc62159faceb6bd30d7e03b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f38c4c365590f4246fee0e5fffde6db55fa51ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f38c4c365590f4246fee0e5fffde6db55fa51ae6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
